)]}'
{"id":"openstack%2Fnetworking-ovn~705253","triplet_id":"openstack%2Fnetworking-ovn~stable%2Fqueens~I3e9e734b757f67d3e4ab1c5a02f8e7f9f0c9aa2e","project":"openstack/networking-ovn","branch":"stable/queens","hashtags":[],"change_id":"I3e9e734b757f67d3e4ab1c5a02f8e7f9f0c9aa2e","subject":"[OVN] Remove VLAN check when setting external_mac","status":"MERGED","created":"2020-01-31 16:11:57.000000000","updated":"2020-07-16 18:14:28.000000000","submitted":"2020-07-16 18:14:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705253-1594923268506-e29519fa","meta_rev_id":"5fb3793fd7fa71a5f4f02b741e2722281fdf1b8d","_number":705253,"virtual_id_number":705253,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-15 13:36:36.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-16 18:14:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2020-07-16 10:43:19.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-07-16 13:11:01.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-07-16 13:11:01.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-16 10:43:19.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-07-16 13:11:01.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-07-16 18:14:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"15041f926c0acce47abb9db0d47be975c94f2655","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-01-31 16:11:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"382c3a67704d9bc5b4e5f2f9e35766745921693e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 19:28:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0decd17d9ef446ba53c8ad9ecb2c048 : SUCCESS in 6m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82d6f88a820c4d799f12091f75b207c7 : SUCCESS in 12m 12s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b9f2421772fe43b19fadc2ed33d7173c : SUCCESS in 7m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bdc5c396a28547b4937ce71491a8579f : SUCCESS in 10m 56s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c4fb6e4602b54fd580c73b132e25b6db : SUCCESS in 1h 03m 43s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e7044f2758554a9fb3f73adccf6eff8a : SUCCESS in 1h 51m 19s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/4f5fc0dc66eb402face740f932adfe86 : SUCCESS in 21m 29s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/f4e23da1728a440888e177a80c6cade3 : FAILURE in 25m 17s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/f46aeae71d534402a375445b993af95e : FAILURE in 41m 21s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/48a75f1aa49542dc8cde9396e6f12acf : SUCCESS in 3h 06m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e974e3b74de43a29fce8b69cb327686 : SUCCESS in 11m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9f13c5888d52f51b0ffb7685bce79a341324300","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-05 09:07:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"37864536b4fb2a69a6922a42e329a537cfed1985","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 11:47:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5909dc420e914c23bdf131e0dd06682d : SUCCESS in 7m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9cec69c1250f4d9b97c9757503ec6e9c : SUCCESS in 11m 44s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/fd3a5b30c1a54f90a46c855cff44d09a : SUCCESS in 7m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e1abd3ca7e1f41f189186bbd78fb69b1 : SUCCESS in 10m 42s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/4a3d5adc418a45f0a78cdc92eb339197 : POST_FAILURE in 18m 13s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0555b466165d4063a2f45741e452f022 : SUCCESS in 1h 33m 40s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/43df3039324f4223b4a350773382264b : SUCCESS in 18m 02s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/a62fa2b8385d461b848c0a66a81460ba : SUCCESS in 17m 45s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/dabb18171df8459abf1253dd58acf243 : POST_FAILURE in 18m 42s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/c580dbf823924a149d824bc708a790f6 : SUCCESS in 2h 32m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ac2e957fa214d2cb757c192d3927e8d : SUCCESS in 12m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a14d6343f90e2039c8240e5e8ff024cf45c9090","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-05 11:54:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"57d8499c6bcd6b4259745ae80c5baf0946f8b49c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 14:27:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/894f54cf2d01413392ecd21efedcb84d : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b7187d73260f460bb4cb0b34f0b87443 : SUCCESS in 11m 24s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4bc715339ebb4376841aeaabde7ce50b : SUCCESS in 10m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c3c882834bb74bdf86f63392e3848ce3 : SUCCESS in 10m 33s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/625f117ff7364cda982a68c94cd9dc36 : POST_FAILURE in 16m 29s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4ca70fc22c02443ba79ada7c959f8044 : SUCCESS in 2h 02m 10s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/885e177cddb1403082388315ac0ab2c9 : SUCCESS in 18m 31s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/c22cae30fba140a4b276014e98bb0268 : SUCCESS in 17m 56s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/8cdd7f5b00ef4049a3c14ff7f185b28f : POST_FAILURE in 17m 01s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/24993529bf7e4a47b9ed5e293fd2973a : SUCCESS in 2h 24m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad7a1ce41d5a43678452c0b4b9966619 : SUCCESS in 13m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1fdd042986466fcfbfe3aca818f76720669248c","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-20 13:57:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"141ee9e233434ec8cf9c6d8debd0cc3cb971ebca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 17:17:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c70ff03c2fb64a46aef55e6239c5d869 : SUCCESS in 8m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/45eec143784c46dcbe8c7ce8bb1fe8ad : SUCCESS in 14m 08s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0a22610afbb840b08c93f5c3ded0bdbd : SUCCESS in 8m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2cce96f5c01f48bbbba3af0882e82790 : SUCCESS in 13m 08s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/40117695a1d9441ea5255e505b475009 : SUCCESS in 1h 05m 51s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d37b2aaac21c4595be36d23ed5a96b93 : TIMED_OUT in 2h 16m 52s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/b16caae2f95d4a3787b1fbcb12120298 : SUCCESS in 33m 29s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/4a8a3f6b4ba94a3a9475b9cce080bd8e : SUCCESS in 26m 21s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/ce46cd9fba404d1892c9d1360eeaa26e : FAILURE in 35m 30s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/e48f7653e1fc4aadaf2986a6d893feed : TIMED_OUT in 3h 13m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57fadb18250446228e563a232f5e5a38 : SUCCESS in 15m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e54743a9b525833d3ec48cd2281b7f98e4aadf9e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-05-11 11:08:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cde1eee84b9353c35c227f414678360006c3f95e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 14:21:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ace9295ec87147d8835d2fd1c4cca265 : SUCCESS in 6m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7075194e08ad477b9baee369376fdcc9 : SUCCESS in 10m 19s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/68599c6237c94460a5de1dda2b5f00e6 : SUCCESS in 6m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9ecdae382b5844b68da2b68e468056b4 : SUCCESS in 9m 37s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/b3ac674536564216a77d581768763f5f : FAILURE in 53m 57s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6485d2ca0cfa42c8939e0aa6f1b50be7 : SUCCESS in 1h 33m 16s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/ff32a71e8a6d471ba01f8744206afb9b : SUCCESS in 17m 20s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/91b5a7308cd34f5786c3910481e95fe8 : SUCCESS in 17m 15s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/0633cbb99ab9475ea84694828779b70d : FAILURE in 35m 45s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/c45056f94f34428cb0193430b9c6b831 : FAILURE in 3h 01m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f0ee042ca9e4a27a89c330e11cc5c9c : SUCCESS in 14m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85590e140e09a3287a38a16e6e5e7cb17b51c044","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-07-15 13:36:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bf577c32a4012d0677c0718a8c87b8f13f1539ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 16:27:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f63c782a4ad540e680be9c43a4df4564 : SUCCESS in 6m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c3bca0acd4546608a73b100f23e6464 : SUCCESS in 11m 20s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f324e37eca82456fbb57e995b79d8609 : SUCCESS in 6m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f29d94bd55af4b1d9c7b073392af7ce5 : SUCCESS in 13m 29s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/6e45b5b3874c4cdb8edd8c1a703b9bf5 : SUCCESS in 1h 08m 16s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b6411d9ba1b3481aaadf65ce14a1c245 : SUCCESS in 1h 32m 07s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/745e8020fa104448acc466356f87c101 : FAILURE in 46m 52s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/d981346da97146a18122beb6902d8ff2 : SUCCESS in 15m 36s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/98a00d0c2968480090190345053d23b2 : FAILURE in 28m 06s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/556eda83392247f08705e47ce51443db : FAILURE in 59m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/338f362dda2c486284d5bf00c764183d : SUCCESS in 15m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc608c74d9b88bb934e32276606c408e4c53fdde","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-07-16 08:00:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9dc23031f2ad4e20c7f23a22be94720d369f702b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 10:33:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/906c04869dfd4816a5e17c28b57021e6 : SUCCESS in 6m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0a60a9873b2840c7ab78aef8bb0e6a33 : SUCCESS in 10m 20s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c1a098911961479883e7085465a24f2e : SUCCESS in 6m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5f259b7172ed446bb83b72f66eded542 : SUCCESS in 10m 49s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/ee71900972a84a528d64b6701c29d39d : SUCCESS in 54m 49s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/765102ecea01432fb96eae8b1b5dc462 : SUCCESS in 1h 40m 52s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/69fd5a7a4130459daf0be1f9172a4ec1 : SUCCESS in 17m 36s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/d31c3123a9be4916b92fd9076600edcc : SUCCESS in 21m 41s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/4bb2c798186e47eba86e09debb635307 : FAILURE in 34m 15s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/dbdab8c5de234b1ab9072631b2c546a8 : FAILURE in 2h 12m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0983809af44498ba41ff1f9d62d5e81 : SUCCESS in 12m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"648628cd0070997da69f66a15276a1797bbcaad3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-07-16 10:43:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0d7042d91274c18c5114e2b152cdfb7505414675","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-16 13:11:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d96fb19672579b8d72dda9ed04d56991c5bf5124","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 13:11:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"399e6a4c96713c8254a0ee78ec8a36e2c561c438","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 18:14:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5fb3793fd7fa71a5f4f02b741e2722281fdf1b8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 18:14:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6eec9c7ea8c4e29bf700bdb79bc631f : SUCCESS in 7m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/353203ec3c0f4c8286ca6d8954a85cad : SUCCESS in 13m 29s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5899f835289b45388ac5777354dbd856 : SUCCESS in 8m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/29e91ddffbdc4485980dc271bda2f738 : SUCCESS in 11m 02s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/3e58c0263fb64dc9a8681d0422681aa6 : SUCCESS in 41m 09s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f105f06c82934718b7631a1a6a9b0177 : SUCCESS in 1h 37m 13s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/8764a16fa1a5489d8e105ee3d09b909f : SUCCESS in 14m 18s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/ed2757bf56954be5a47b25ddf4be5243 : SUCCESS in 20m 51s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/d2a0b8bf67764e59ab8440539fd05034 : SUCCESS in 2h 29m 04s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b915cc15ee569308cd217dc2716fe3a787d00c8b","revisions":{"b915cc15ee569308cd217dc2716fe3a787d00c8b":{"kind":"REWORK","_number":1,"created":"2020-01-31 16:11:57.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/53/705253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/53/705253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/53/705253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/53/705253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/53/705253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/53/705253/1"}}},"commit":{"parents":[{"commit":"3418f71acda7945a79fb609aa6cedf03940b4d98","subject":"[OVN] Delete NAT entry first on any FIP update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/3418f71acda7945a79fb609aa6cedf03940b4d98"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-01-31 16:08:26.000000000","tz":0},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-01-31 16:11:46.000000000","tz":0},"subject":"[OVN] Remove VLAN check when setting external_mac","message":"[OVN] Remove VLAN check when setting external_mac\n\nThis patch reverts [0].\n\nThe code wasn\u0027t accounting for VLAN provider networks, as stated in the\nbug #1842988, DVR won\u0027t work if the provider network (where the FIP is\ncreated) is VLAN.\n\nThere was also an incosistency in how the external_mac was set for the\nVLAN networks. Upon creating the FIP the code was checking for the\nnetwork type and not setting the external_mac attribute in case the\nnetwork was VLAN type. But, if the port went down and up again (e.g if\nyou reboot the VM) the event handler that set/unset the external_mac [1]\nwasn\u0027t check for the type. This is how people worked around the DVR\nproblem (as stated in bug #1842988).\n\nFor more information see bug #1842988.\n\n[0]\nhttps://github.com/openstack/networking-ovn/commit/c5aef51edc9843db605303ec8bd8610b6c55e9c2\n[1]\nhttps://github.com/openstack/networking-ovn/blob/eda5d7f80d877601170631c5f5485370ea701f42/networking_ovn/ml2/mech_driver.py#L794-L800\n\nCloses-Bug: #1842988\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n\n(Cherry-picked from Neutron Ifb795626dc9c2ac4f0104f491dd38c9b4cc902c9)\n\nChange-Id: I3e9e734b757f67d3e4ab1c5a02f8e7f9f0c9aa2e\n(cherry picked from commit f9b1dffada3ad6c1328843b30128e72b92cfcdf5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b915cc15ee569308cd217dc2716fe3a787d00c8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b915cc15ee569308cd217dc2716fe3a787d00c8b"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
