)]}'
{"id":"openstack%2Fnetworking-ovn~705296","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~Ibce080d3af7d1c1c9b146ce5c4d9e96719e675df","project":"openstack/networking-ovn","branch":"stable/train","hashtags":[],"change_id":"Ibce080d3af7d1c1c9b146ce5c4d9e96719e675df","subject":"OVN Metadata agent gets OVSDB updates for only its Chassis","status":"MERGED","created":"2020-01-31 20:57:48.000000000","updated":"2020-03-13 22:22:20.000000000","submitted":"2020-03-13 22:20:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705296-1584138046865-c6fdaf72","meta_rev_id":"7cc6fea5657e59108a2876f295520b0cd95d3acc","_number":705296,"virtual_id_number":705296,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-13 22:20:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"date":"2020-02-12 18:27:39.000000000","_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2020-03-11 11:50:58.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-03-13 19:29:35.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"date":"2020-03-12 09:12:51.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-03-13 19:29:35.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-03 14:39:26.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-03-11 11:50:58.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-03-12 09:12:51.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-03-13 19:29:35.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-03-13 22:20:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c2c3c67f645e966395fc4c541ee32bbb519e6798","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-01-31 20:57:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea63e2dc5a7a3c555483a5fad3f3b2c4bc00d335","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 23:55:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e2935bf85e54de28f79d272038e3c2b : SUCCESS in 7m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bcb2980f3fb74bc2b51178ca0b33da88 : SUCCESS in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa9dc22f9d12433693ab37ba73ea2254 : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32730bffcc504fce9c37d8f8b9c06aad : SUCCESS in 10m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b4d6b4801d240588246e15349bf0db9 : SUCCESS in 11m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f20e8c824f14375bc30875162444086 : SUCCESS in 10m 58s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/f9d0e0c2068a4cae84008b06b2b52002 : FAILURE in 2h 20m 54s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/01873351d9584f2a85a16f5462bce877 : SUCCESS in 2h 14m 21s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/da51265c771846b2a6780f18f13b7c55 : SUCCESS in 1h 23m 52s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/052036074c3e45c490e493f62909efd2 : SUCCESS in 1h 51m 47s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/e2e92d67329b42ce888e07c042256796 : SUCCESS in 21m 50s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/3176fe351d8b4501b4e9f037be976c1e : SUCCESS in 24m 07s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/27411dc512dc44be9fc98ceffb6347cd : SUCCESS in 2h 45m 44s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/de661afbaed54becbe0b97606f1df0b0 : SUCCESS in 1h 40m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c0dad31a1d842fc947402f49d211d62 : SUCCESS in 11m 12s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d2f53e0d51b44e088832fa21505ba250 : FAILURE in 2h 50m 42s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/8bc35e55645148b1a7e58f7f1be72860 : FAILURE in 1h 52m 22s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c1ef6d1491b4bf1a3265f141a4cd1c1 : SUCCESS in 40m 20s","accounts_in_message":[],"_revision_number":1},{"id":"13a279ee1ea7d7d61f616ad172c03c4e327be42c","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-02-01 00:17:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"40eb12442310629af3b1ea1834159352b214f855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 02:58:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ece709ac0fc488499b58b54e9da2164 : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d4bdb53874894880a97314e54e26f447 : SUCCESS in 8m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80aa6ad546424a4094d6fcdd5709a63e : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5bb4dd04c634f9e921315d64781567d : SUCCESS in 8m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d01c35e8c22f4a76a6fef67cb3689559 : SUCCESS in 9m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9656a9f127642aea3f11a8a9b640785 : SUCCESS in 9m 04s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/7c77f9b86bcd4f2ba9673557c26cf679 : FAILURE in 2h 13m 28s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/9a03d265c1034620961f4373a1a73678 : SUCCESS in 2h 04m 28s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/605c54265662456ea1aede76256c0367 : SUCCESS in 1h 35m 54s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0dc467a6359b4042b8e7d92102f2438b : SUCCESS in 1h 45m 45s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/77295d990a5c42eabd2c41cc094b089f : SUCCESS in 18m 56s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/2d36c52ba1e04b71873aa163d02c8c15 : SUCCESS in 18m 52s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/07571decf5924629890e82cca82941e2 : SUCCESS in 2h 36m 11s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/9a1303e568c149fc9cbad27e94fe306c : SUCCESS in 1h 50m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a18218a0129042239372cead257af7ff : SUCCESS in 9m 09s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/87d240cb8d224a1bb18b1a28045e5682 : FAILURE in 2h 17m 10s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/49460a9fe2bb4c618ea621d33fdbbf99 : FAILURE in 2h 08m 19s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b091d1c475254d94884c67a32b606cb8 : SUCCESS in 37m 32s","accounts_in_message":[],"_revision_number":2},{"id":"e8e70b433409e77d3dd18871656802242b36ef3e","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-02-03 14:39:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fafabeea0ea93bfd0e34a19a04bec64454b84787","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-03 15:34:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b6ff0859b25193a963b4ae9079320d4e0083b48","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-02-06 17:21:18.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"0686194e4b0f09d9769eaff094e55a8fd5d7d895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 20:04:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1c1c6a7c1e546859eed11316ed24fb5 : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/27b8821c371845b3af0eb766a178d5b4 : SUCCESS in 10m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61891b2751b24809bdeda9e8521c0396 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c26f8afcdec24804b8a18d72e57d3666 : SUCCESS in 9m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7a76d63e501441c85980779f73ece7b : SUCCESS in 12m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0cacf0efc3f439e821e01f52927993e : SUCCESS in 12m 43s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/e4edc5455e274072aa84bcba6601e571 : FAILURE in 2h 27m 21s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/ecfa77535d7c4fd38da031b5b6f27a2f : SUCCESS in 2h 08m 28s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/1f45f8e7372a4568b14b845de7d6d34c : SUCCESS in 1h 40m 43s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fb33ce8b228649d79b70abaf2dace82a : SUCCESS in 2h 12m 28s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/0b251cb3c5be4c0caf96200974cf8170 : SUCCESS in 28m 07s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/b2d66426e0314113911be1e98611bf8f : SUCCESS in 20m 55s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/157a90408b6a4373b306f3115fa6812a : FAILURE in 2h 28m 44s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/efd0ce3f24ca44d7908a1d4532b9488f : SUCCESS in 1h 28m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a8e91a561c041c8bf7a15b8cf3d9616 : SUCCESS in 13m 44s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9e9dc655f82a4a78adb6921a81998a02 : FAILURE in 2h 37m 00s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/72038856faf44c6986db69844fea07ce : FAILURE in 2h 09m 26s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c058e823eeec4e63828443aab96a35ed : SUCCESS in 36m 35s","accounts_in_message":[],"_revision_number":3},{"id":"a2e70906253864f1bf1dd15bd967319e6cdc85ab","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-02-11 19:43:45.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f416a0d1b43d9c5ecf75692661e925438e9c8330","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 22:23:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/399f6f3a686a4b9e91660576e410130f : SUCCESS in 7m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/be32e3dedfb2449a9828d6c983f30c66 : SUCCESS in 12m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abd53f654dc047b9b07b4f500be2ae17 : SUCCESS in 8m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94cb9f99a771434d8b8426736e6cf01c : SUCCESS in 13m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32772655a40449c0ade041e4a10709d6 : POST_FAILURE in 21m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02516dea16334007ba906c64ceab7be6 : SUCCESS in 17m 41s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/a5257644fd784fe0b67b2797be53b06e : FAILURE in 2h 31m 48s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/b7444b0aca084fd89971b3f4e65c0b98 : SUCCESS in 2h 33m 13s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/96d15dd5982640cdaa2e19c02cd9b12e : SUCCESS in 1h 23m 46s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5cf4924bd9894457b64f42b55b2e46ea : SUCCESS in 1h 46m 59s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/9ab825eeebad455aa127926e926138e1 : SUCCESS in 21m 55s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/624eb77e31cb4a94b0b2086956d03a17 : SUCCESS in 25m 47s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3d91d838b90f453fb72938460ee01682 : FAILURE in 16m 11s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/becb359558874f16a3faeaedded22663 : SUCCESS in 1h 21m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5470c0f2ab8a47fcb34ce3c2628ac80e : SUCCESS in 15m 39s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b2db6d99438644d9b0b255cdd3118fe4 : FAILURE in 2h 24m 46s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/811c174860d24285b3866893383af9f5 : FAILURE in 2h 12m 28s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/851c9c1b0db64d27badabf460595bf59 : SUCCESS in 45m 52s","accounts_in_message":[],"_revision_number":4},{"id":"7e2536ff2755fc1b4d400cb7e75a9ff6c5e48e21","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-12 09:28:04.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f746bd83d25186a8061fd23292b7dfa97268f6de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 12:28:33.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0369468a7e874e1e885e8e0b9da18cd0 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a48b1891c5bd4eec963d7b3f42057b8c : SUCCESS in 10m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecb22574422e49f19c4fcf32490399a9 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8dc83686a4943319994b464aa6ec263 : SUCCESS in 11m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e1620d2aa90345d680134ff573291a10 : SUCCESS in 11m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db1bc332649442f3bb23289d115f1cf2 : SUCCESS in 9m 59s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/0824ae10e5674bad9cfa786d94b28eba : FAILURE in 2h 49m 34s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/44470010c11945d68b4647ecbce6dd76 : SUCCESS in 2h 45m 41s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/1c92df43d0004996bde26e8b92e92e5e : SUCCESS in 1h 32m 00s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e6ea3c17b8b2436f98f6df8383a678e5 : SUCCESS in 2h 05m 46s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/f9e2a915d3ff48c7be4df8c9a3300140 : SUCCESS in 24m 06s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/3f035aa720c74c6da836e6b20424af6f : SUCCESS in 32m 31s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/be500eb47de749c4aabb8c9a6a704af7 : FAILURE in 1h 39m 04s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/1b76b61e14254cbcb9a903f011e7530d : SUCCESS in 1h 42m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/533a2e57b01147d8b9b8bdf7230f19fc : SUCCESS in 12m 14s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/10752c5ff5cc46bf8c345401989975a7 : FAILURE in 2h 46m 19s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/325b7a1a32f94113b49aee684e340d65 : FAILURE in 2h 56m 02s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1712a54b6cf24fe99b070fed1ef2fc46 : SUCCESS in 46m 38s","accounts_in_message":[],"_revision_number":4},{"id":"fca67dd55f1569511ce269c56b9d0e1f49f0a51e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-02-12 18:26:39.000000000","message":"Patch Set 4:\n\nthe OOO job is failing in some stupid way where it is trying to start the neutron ovs agent...\nhttps://4c6abe1a935072d00ef5-21b0bd21f45f83c83b77a87ffa43cecc.ssl.cf2.rackcdn.com/705296/4/check/networking-ovn-tripleo-ci-centos-7-containers-multinode/be500eb/logs/undercloud/home/zuul/install-undercloud.log\n\nSystemctlException(str(err))\\npaunch.utils.systemctl.SystemctlException: Command \u0027[\u0027systemctl\u0027, \u0027enable\u0027, \u0027--now\u0027, \u0027tripleo_neutron_ovs_agent\u0027]\u0027 returned non-zero exit status 1\\","accounts_in_message":[],"_revision_number":4},{"id":"027b4f7e55edfcb9821e836816e4afd3f90cd597","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-02-12 18:27:39.000000000","message":"Patch Set 4:\n\noh yeah, undercloud. :p","accounts_in_message":[],"_revision_number":4},{"id":"24617246d859f0dede099c518403b4ef778c7da4","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-03-11 11:50:42.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9ba5097849754c21e61781e520afc678e4185b6d","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-03-11 11:50:58.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"07a74cd9599eb72f107be88ac02dd2077d866f40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 15:13:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/066e43870a304015b3e810a68bcdf4dc : SUCCESS in 8m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74214a59e1b74c7ba824f373f513727a : SUCCESS in 12m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1f629e128134d4db8fa5cc944d5437a : SUCCESS in 10m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2b2637537eb43e7a76523bf95177eeb : SUCCESS in 9m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/407e05309cae4457b22ae6a721522e9b : SUCCESS in 10m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/848fa3cba85a43659f5b6b76d097be9d : SUCCESS in 11m 23s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/aa2c5ac7c51e4028a185d43140bbd1fb : SUCCESS in 2h 46m 48s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/a3f53c401a9e43ddbc2b200a794ce5fa : SUCCESS in 2h 13m 01s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/6e5ab473a23d4ebd8c869da218cff1d9 : SUCCESS in 1h 20m 26s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3826a332f06d44339d943c7229062a34 : SUCCESS in 2h 07m 23s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/d6ce02d3dd7b477cb11647d22c2dee44 : SUCCESS in 23m 03s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/45815fa1cba24b3984ab0c3fa73725c0 : SUCCESS in 21m 39s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/452e86eb31f04521857fdb4459f0b7aa : SUCCESS in 3h 13m 20s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/ad5f9c7b335447d288ab750196c8eec9 : SUCCESS in 1h 44m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56f7373229a448478b5960836d7ea95b : SUCCESS in 11m 14s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b4a88cfb0d4d40858b25e272541ecf10 : FAILURE in 2h 22m 04s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/bc642b6a0c824ccdafca5737063b1e3a : FAILURE in 1h 55m 13s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a298bc079aaf4cfdbba96545195c04bc : FAILURE in 56m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4e55f4e52b57990245b882b6ceabd256f0854e69","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-13 19:29:35.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"af7b1d900d453ce615f228e8b75c9a3d121a421d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 19:29:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cd273f400e700252549976215c240b03baa3ef2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 22:20:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d75488b40f2c49dd9e6137567a208478 : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6962bc2ed053462585287167edd3572a : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4ca574906504e2d87fdd5e58a2556f0 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7955f51ef49432b9b44583f76849ab0 : SUCCESS in 9m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77b80eeacb0a43368d40862f268596a1 : SUCCESS in 9m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56bb9b042b9b4680a4807bcac2c75ead : SUCCESS in 11m 56s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/cccdac59613c403dadde3055264d74c2 : SUCCESS in 2h 22m 00s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/66d2eb19fdf54742bfd20a590eae781c : SUCCESS in 1h 24m 24s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cc9da61fd03147b5b0497a931b806540 : SUCCESS in 1h 58m 26s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/d6611cf199ad44b3a17ebb06f1f1d0e5 : SUCCESS in 22m 35s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/eef02e877b4c4cc9af4cdf1e31714127 : SUCCESS in 19m 42s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b618223ac4d34266851c158357592a0d : SUCCESS in 2h 46m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2751f6b9f1f485b871b16b86a3b26bb : SUCCESS in 11m 09s","accounts_in_message":[],"_revision_number":4},{"id":"6ff9f4dc9893f76a4b4e72e1d0f4f3c846f1a890","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 22:20:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"7cc6fea5657e59108a2876f295520b0cd95d3acc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 22:22:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce9c66e068084c2aa3a232441f0c2478 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5ef7f92884ff77362f21ea9a2f00ad1823868eda","revisions":{"ffaa7c4b79267d7b61c8f147357f2f07a111cee3":{"kind":"REWORK","_number":1,"created":"2020-01-31 20:57:48.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/96/705296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/96/705296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/96/705296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/96/705296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/96/705296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/96/705296/1"}}},"commit":{"parents":[{"commit":"844540e2d00b85bd31b8e0795b2acb8c81f1da75","subject":"Merge \"Agent liveness - allow time to propagate checks\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/844540e2d00b85bd31b8e0795b2acb8c81f1da75"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-01-31 20:30:29.000000000","tz":-360},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-01-31 20:57:42.000000000","tz":-360},"subject":"OVN Metadata gets OVSDB updates for only its Chassis","message":"OVN Metadata gets OVSDB updates for only its Chassis\n\nThe metadata agent registers the Chassis table with ovsdb-server\nand therefor gets database updates every time *any* Chassis is\nupdated--even if the update is just a liveness check that updates\nnb_cfg.\n\nThis patch adds a condition so that metadata agent only gets updates\nfor the Chassis that it is running on.\n\nChange-Id: Ibce080d3af7d1c1c9b146ce5c4d9e96719e675df\nRelated-bug: #1861092\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ffaa7c4b79267d7b61c8f147357f2f07a111cee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ffaa7c4b79267d7b61c8f147357f2f07a111cee3"}]},"branch":"refs/heads/stable/train"},"f4f6b8c6a981c8e6b36257f8a90a1d35949d4841":{"kind":"REWORK","_number":2,"created":"2020-02-01 00:17:55.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/96/705296/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/96/705296/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/96/705296/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/96/705296/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/96/705296/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/96/705296/2"}}},"commit":{"parents":[{"commit":"844540e2d00b85bd31b8e0795b2acb8c81f1da75","subject":"Merge \"Agent liveness - allow time to propagate checks\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/844540e2d00b85bd31b8e0795b2acb8c81f1da75"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-01-31 20:30:29.000000000","tz":-360},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-02-01 00:17:53.000000000","tz":-360},"subject":"OVN Metadata gets OVSDB updates for only its Chassis","message":"OVN Metadata gets OVSDB updates for only its Chassis\n\nThe metadata agent registers the Chassis table with ovsdb-server\nand therefor gets database updates every time *any* Chassis is\nupdated--even if the update is just a liveness check that updates\nnb_cfg.\n\nThis patch adds a condition so that metadata agent only gets updates\nfor the Chassis that it is running on.\n\nChange-Id: Ibce080d3af7d1c1c9b146ce5c4d9e96719e675df\nRelated-bug: #1861092\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f4f6b8c6a981c8e6b36257f8a90a1d35949d4841"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f4f6b8c6a981c8e6b36257f8a90a1d35949d4841"}]},"branch":"refs/heads/stable/train"},"ee7866f4975e21c250bc25cad0388458a872b523":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-02-06 17:21:18.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/96/705296/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/96/705296/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/96/705296/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/96/705296/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/96/705296/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/96/705296/3"}}},"commit":{"parents":[{"commit":"844540e2d00b85bd31b8e0795b2acb8c81f1da75","subject":"Merge \"Agent liveness - allow time to propagate checks\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/844540e2d00b85bd31b8e0795b2acb8c81f1da75"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-01-31 20:30:29.000000000","tz":-360},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-02-06 17:20:08.000000000","tz":0},"subject":"OVN Metadata agent gets OVSDB updates for only its Chassis","message":"OVN Metadata agent gets OVSDB updates for only its Chassis\n\nThe metadata agent registers the Chassis table with ovsdb-server\nand therefor gets database updates every time *any* Chassis is\nupdated--even if the update is just a liveness check that updates\nnb_cfg.\n\nThis patch adds a condition so that metadata agent only gets updates\nfor the Chassis that it is running on.\n\nChange-Id: Ibce080d3af7d1c1c9b146ce5c4d9e96719e675df\nRelated-bug: #1861092\n(cherry-picked from neutron I452b7de09312ecea621c4b448cc63f037cad9675)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ee7866f4975e21c250bc25cad0388458a872b523"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ee7866f4975e21c250bc25cad0388458a872b523"}]},"branch":"refs/heads/stable/train"},"5ef7f92884ff77362f21ea9a2f00ad1823868eda":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-02-11 19:43:45.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/96/705296/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/96/705296/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/96/705296/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/96/705296/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/96/705296/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/96/705296/4"}}},"commit":{"parents":[{"commit":"59e705be402cd10f332b460d898cc478ded7160d","subject":"Merge \"Fix for getting port groups while using OVS up to 2.9\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/59e705be402cd10f332b460d898cc478ded7160d"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-01-31 20:30:29.000000000","tz":-360},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-02-11 19:43:45.000000000","tz":0},"subject":"OVN Metadata agent gets OVSDB updates for only its Chassis","message":"OVN Metadata agent gets OVSDB updates for only its Chassis\n\nThe metadata agent registers the Chassis table with ovsdb-server\nand therefor gets database updates every time *any* Chassis is\nupdated--even if the update is just a liveness check that updates\nnb_cfg.\n\nThis patch adds a condition so that metadata agent only gets updates\nfor the Chassis that it is running on.\n\nChange-Id: Ibce080d3af7d1c1c9b146ce5c4d9e96719e675df\nRelated-bug: #1861092\n(cherry-picked from neutron I452b7de09312ecea621c4b448cc63f037cad9675)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5ef7f92884ff77362f21ea9a2f00ad1823868eda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5ef7f92884ff77362f21ea9a2f00ad1823868eda"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
