)]}'
{"id":"openstack%2Fnetworking-ovn~708313","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~Ifc0f188122931bc89ae8f90096b4bcc7303ec080","project":"openstack/networking-ovn","branch":"stable/train","hashtags":[],"change_id":"Ifc0f188122931bc89ae8f90096b4bcc7303ec080","subject":"Bump-up os-log-merger to latest version","status":"MERGED","created":"2020-02-18 08:10:24.000000000","updated":"2020-02-19 13:11:30.000000000","submitted":"2020-02-19 13:09:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708313-1582117749187-7fccfbb9","meta_rev_id":"6ff8786350540db68e6879cd332c27ffd6ce2cd8","_number":708313,"virtual_id_number":708313,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-19 13:09:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2020-02-19 10:32:50.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2020-02-18 22:02:18.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":1,"date":"2020-02-19 10:32:50.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-18 22:02:18.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-02-19 10:32:50.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-02-19 13:09:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b9643835269720e6fe59742f3dd3e471211c3802","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-02-18 08:10:24.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"3fa6864e607305591830f35be60ad5b5db0be731","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 11:04:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a367a2b14f834466b589ba7eb19f5105 : SUCCESS in 8m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8babc2d85b9845e49d0ca0937b79ba43 : SUCCESS in 10m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91dce00ed4e5445c92a7a2dfc10ab355 : SUCCESS in 9m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/937adcb7d20246daa38ee01022970376 : SUCCESS in 10m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef83fe19139f43be93f8f29ce65de09b : SUCCESS in 13m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d1f79794aa94dfb9cd628f4864a3336 : SUCCESS in 15m 58s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/638e2981b1b74017948a85230f50493f : FAILURE in 2h 30m 11s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/9e76ad6a92a948aaad8ef65a1551eb09 : SUCCESS in 2h 00m 11s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/ea87f024fb5d4f1eaeaed9866e6cb526 : SUCCESS in 1h 20m 16s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cfe31da488ec4095917bcef3e6f076e0 : SUCCESS in 2h 09m 32s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/58463f59001f4af8af9dc57d4ccc9629 : SUCCESS in 26m 19s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/82f18d439861453caa87f06ca2efb13d : SUCCESS in 24m 40s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d398e511beb3470c990a0fa9e0b82195 : SUCCESS in 2h 40m 08s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/8b08d7c5981f44c691f52ba36b61e933 : SUCCESS in 1h 28m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2313c3c829794b7dae5bad7be21b7073 : SUCCESS in 12m 44s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/88779403ce8540258b70f2edbc207a5f : FAILURE in 2h 29m 12s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/8447f97343d64761a4e78b5f8a7572ec : FAILURE in 2h 26m 57s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a7f14127264d465f9b17c775a6ca9a7f : SUCCESS in 41m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0afb3b8b32872f875626d46b37649532e15d08bc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-18 22:02:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"45e1f381951e5e2e8ffd436bfa05e563d43808ac","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-02-19 10:32:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7d75fe1d52247ccfe286c850b264460745d5d0f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 10:33:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c8392e53c5ba6af2a60f4a39246aebb5f90ecc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 13:09:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61a68ac4854e43429d59ab9c6163ad52 : SUCCESS in 7m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/98921e550bca494ea417254e15d4d8ec : SUCCESS in 10m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cebeaf3e9afa4ae09b48e9f66eb0f5e2 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbb5f1c9ec3342938f5c9d8fe3d1a95c : SUCCESS in 13m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b492ba31ff234c1396543fbed0393483 : SUCCESS in 12m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50316deb015e403899b23f6c3458438e : SUCCESS in 12m 44s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/4a9c003f183d4d7e80d0bf605150e9a0 : SUCCESS in 2h 10m 53s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/f291f49475ee49bbbcadaa1fa8600e84 : SUCCESS in 1h 26m 57s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/90837d0c603f456da1c11f7dffcb9e82 : SUCCESS in 1h 53m 29s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/e0b1e0397b394176ae9456d32c5dbddc : SUCCESS in 22m 42s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/1a7a364d29644633b5c44913151dfb58 : SUCCESS in 22m 21s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/58197a168dc74d4b843fad9dd30b4745 : SUCCESS in 2h 30m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e133d734fcf4b03bfbb88188deb4b51 : SUCCESS in 13m 19s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a6f682101bb48afbc9c519dc50cda81 : SUCCESS in 49m 25s","accounts_in_message":[],"_revision_number":1},{"id":"4f98f19fc601d5af198ef6db126323669084cf50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 13:09:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6ff8786350540db68e6879cd332c27ffd6ce2cd8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 13:11:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b11ce3e40f6e4cdb8d252a87e011558e : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c2b413e5a720b15f05cc463f46d767140b545ddf","revisions":{"c2b413e5a720b15f05cc463f46d767140b545ddf":{"kind":"REWORK","_number":1,"created":"2020-02-18 08:10:24.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/13/708313/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/13/708313/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/13/708313/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/13/708313/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/13/708313/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/13/708313/1"}}},"commit":{"parents":[{"commit":"59e705be402cd10f332b460d898cc478ded7160d","subject":"Merge \"Fix for getting port groups while using OVS up to 2.9\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/59e705be402cd10f332b460d898cc478ded7160d"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-02-14 11:27:29.000000000","tz":60},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-02-18 08:10:24.000000000","tz":0},"subject":"Bump-up os-log-merger to latest version","message":"Bump-up os-log-merger to latest version\n\nThe virtualenv now is created with python3 by default.\nBecause of it we need to bump os-log-merger to the\nversion that supports py3.\n\nChange-Id: Ifc0f188122931bc89ae8f90096b4bcc7303ec080\nCloses-Bug: #1863220\n(cherry picked from commit bfcc2a39b6979cb5e929055993423c2622a38611)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c2b413e5a720b15f05cc463f46d767140b545ddf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c2b413e5a720b15f05cc463f46d767140b545ddf"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
