)]}'
{"id":"openstack%2Fnetworking-ovn~708331","triplet_id":"openstack%2Fnetworking-ovn~stable%2Fstein~I0b84f41865e3fdea49cf169df5431249c35f5ff8","project":"openstack/networking-ovn","branch":"stable/stein","hashtags":[],"change_id":"I0b84f41865e3fdea49cf169df5431249c35f5ff8","subject":"[OVN] Set \u0027unknown\u0027 address properly when port sec is disabled","status":"ABANDONED","created":"2020-02-18 10:31:12.000000000","updated":"2020-03-04 08:30:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"168fac53a6eec15b48978ca43b98b85fe01a800e","_number":708331,"virtual_id_number":708331,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-18 13:24:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"all":[{"value":-1,"date":"2020-02-20 09:13:46.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-02-18 10:31:12.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-18 10:31:12.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-02-18 13:24:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f70bfc519feefdd299bf31bc39241df3db46cca5","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-02-18 10:31:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c38c3c3f7d5780a3c42b23fe113d6a7a88896d26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 13:24:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78f251536c8e4e188f550012c642035e : SUCCESS in 8m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3159c16a61364ba5bf9ddbfe6baa455a : SUCCESS in 12m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d73167ebd9ab4ed785622589430782b1 : SUCCESS in 7m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/098546c90fc146409fce6832d0054fcd : SUCCESS in 13m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b58e6980c27466ab9f6f9f84c0a1cff : SUCCESS in 12m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e76276036b294825b388aed02992736a : SUCCESS in 12m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dc23f88365e42aea299d337ae8461fc : SUCCESS in 13m 27s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/7168f53ca4da4315bb6d574fc4026e66 : SUCCESS in 2h 10m 35s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/3610aa65d2314c3a98ba082ad9235dff : SUCCESS in 1h 22m 59s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ebbe0151edf14fb1803e266ce43e7221 : SUCCESS in 2h 00m 55s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/e883d3b927b34c11b71c220538397f33 : SUCCESS in 25m 06s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/f079c3a7b5b04a88ba930e93241e392d : SUCCESS in 25m 27s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a2890c3aee914674936fc5c9525b0fb2 : SUCCESS in 2h 51m 39s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/481ba68a5c27452296a4476237c4d18b : SUCCESS in 1h 36m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60add4bd066743878329d02d27a3f01c : SUCCESS in 12m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d8db49c97c9098c863d167438d82f30e6657365","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-02-20 09:13:46.000000000","message":"Patch Set 1: Workflow-1\n\nI stable/train OVN Octavia gate fails because of it. Please wait with merging it.","accounts_in_message":[],"_revision_number":1},{"id":"168fac53a6eec15b48978ca43b98b85fe01a800e","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-03-04 08:30:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf3b238d3282b6b7feaaee6c0ad4c388e53109a4","revisions":{"bf3b238d3282b6b7feaaee6c0ad4c388e53109a4":{"kind":"REWORK","_number":1,"created":"2020-02-18 10:31:12.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/31/708331/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/31/708331/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/31/708331/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/31/708331/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/31/708331/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/31/708331/1"}}},"commit":{"parents":[{"commit":"bfcc2a39b6979cb5e929055993423c2622a38611","subject":"Bump-up os-log-merger to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/bfcc2a39b6979cb5e929055993423c2622a38611"}]}],"author":{"name":"Daniel Alvarez","email":"dalvarez@redhat.com","date":"2020-01-13 15:56:04.000000000","tz":60},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-02-18 10:32:48.000000000","tz":60},"subject":"[OVN] Set \u0027unknown\u0027 address properly when port sec is disabled","message":"[OVN] Set \u0027unknown\u0027 address properly when port sec is disabled\n\nPrevious to this patch, the \u0027unknown\u0027 address was added to the\nOVN Logical_Switch_Port row. However, the \u0027\u003cip\u003e \u003cmac1\u003e \u003cmacN\u003e\u0027\nwas still there so the port security was not really disabled.\n\nThe correct behavior is to set the \u0027addresses\u0027 field to \u0027unknown\u0027\nand remove everything else.\n\nChange-Id: I0b84f41865e3fdea49cf169df5431249c35f5ff8\nCloses-Bug: #1728886\nSigned-off-by: Daniel Alvarez \u003cdalvarez@redhat.com\u003e\n(cherry picked from commit 03b87ad963d5d8165a92e5c7c284c1517333dd00)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/bf3b238d3282b6b7feaaee6c0ad4c388e53109a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/bf3b238d3282b6b7feaaee6c0ad4c388e53109a4"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
