)]}'
{"id":"openstack%2Fnetworking-ovn~708476","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~Iebc8278aca56f87fcf74cc2725a1ffc1197c1bf4","project":"openstack/networking-ovn","branch":"stable/train","topic":"fix_migration_cirros_image_download","hashtags":[],"change_id":"Iebc8278aca56f87fcf74cc2725a1ffc1197c1bf4","subject":"Fix cirros image download during resources creation for migration","status":"MERGED","created":"2020-02-18 18:23:02.000000000","updated":"2020-02-24 17:34:51.000000000","submitted":"2020-02-24 17:29:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708476-1582565385413-1d1e633c","meta_rev_id":"dd97dde0a5d8951f5455f8aa00b35c8592addc2e","_number":708476,"virtual_id_number":708476,"owner":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2020-02-19 12:16:30.000000000","_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-24 17:29:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2020-02-19 13:53:36.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},{"value":1,"date":"2020-02-19 13:12:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2020-02-24 16:31:39.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2020-02-24 16:31:39.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-19 11:22:53.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-02-19 11:33:34.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-02-19 13:12:42.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-02-19 13:53:36.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-02-24 16:31:39.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-02-24 17:29:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c0acf3d500bb22fb6097c11e1d26b5febf15dcf","author":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"date":"2020-02-18 18:23:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8691e39bb83f6bdefd0b76a82725abb43b87fc41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 19:17:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50d03796557d47c5ade1ddd09be4ece1 : SUCCESS in 8m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a4a3b4fd582431ea7a8873af85bb897 : SUCCESS in 13m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeaf5025637b4647a49bd4fb7f5200b4 : SUCCESS in 9m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b31de1122354362a22a0c5e138bcebd : SUCCESS in 15m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa561919d0ca4cc58bb35743ce4f761f : SUCCESS in 12m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be0a612f74c045e79ce44e795a05858f : SUCCESS in 14m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43bd1c7f710742c18e38eee08547f4aa : SUCCESS in 11m 40s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b80ed4a4d82d437a8c9f49295e0a3c2b : SUCCESS in 46m 47s","accounts_in_message":[],"_revision_number":1},{"id":"709bf44ce11e7bfd1e472f5fbe01845d3403bd36","author":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"date":"2020-02-18 20:11:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"25194b4e4cf03cce4f60add57066c21f5c5897f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 21:10:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b230b5fea354446bf5b4d36a1c7056d : SUCCESS in 7m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/467a0febbce146ba8994e6c55d78e645 : SUCCESS in 11m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c654a4e26dc8422bbabeba0725275c38 : SUCCESS in 9m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e63b773d4bc44389a07c2f328f43a827 : SUCCESS in 14m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cedb32d1d4c249489feda33b2daa9e52 : SUCCESS in 12m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c95730b16b234514b7f77f13627b1a79 : SUCCESS in 14m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d35c08bf3af4acbb3c0437c27f6bf4c : SUCCESS in 15m 33s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21d6a130f5624b9ea9ba064a7de7e020 : SUCCESS in 43m 28s","accounts_in_message":[],"_revision_number":2},{"id":"1a3dfda713555f8fa9a226d254f1d5c4e1ea8dc2","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-02-19 08:05:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"03328f672734c4a847b6ee29ce9604ef8b3f5db5","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-02-19 08:11:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b2f37a7b6b00af223f43c76e928e3de658a052e0","author":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"date":"2020-02-19 09:01:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"599b15de22c0436a5dd6c317ddf3e2981ef575e2","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-02-19 09:20:27.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks Roman!","accounts_in_message":[],"_revision_number":3},{"id":"fbbaaa5be8ee3096ed11a80be6f1083e0c891e74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 09:46:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9761b0a326a4fbe908fb0d30e84d260 : SUCCESS in 6m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/253d64478649405f93b73f63298f29a0 : SUCCESS in 10m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a14cf88afb02458386671e409f301344 : SUCCESS in 8m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/561fcf76666f4bc9953a2b1da9b68593 : SUCCESS in 10m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2764ca1058b3411c898cf3591300fa66 : SUCCESS in 10m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1ed1a0dd58047fe882189ec05bcc77a : SUCCESS in 11m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f1cca2e56c04fa9855ee3c8cf5744fa : SUCCESS in 12m 58s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9547184b476c4310812ec1483acfbf84 : SUCCESS in 38m 56s","accounts_in_message":[],"_revision_number":3},{"id":"3e827410d69a4285571a2540e422570a6da8817a","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-02-19 11:22:53.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"05a2f1f05076bc5ec74ae0a8b3fa0ffc0aaeca84","author":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"date":"2020-02-19 12:14:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3d0709ff91b0f26d49beaf212bd40e0da64a567c","author":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"date":"2020-02-19 12:16:30.000000000","message":"Patch Set 4:\n\nFolks, I have to return back to usage curl -L because new link also does not work directly.\n\n[stack@undercloud-0 tmp]$ curl -o cirros-0.4.0-x86_64-disk.img https://github.com/cirros-dev/cirros/releases/download/0.4.0/cirros-0.4.0-x86_64-disk.img\n  % Total    % Received % Xferd  Average Speed   Time    Time     Time  Current\n                                 Dload  Upload   Total   Spent    Left  Speed\n100   619  100   619    0     0   1144      0 --:--:-- --:--:-- --:--:--  1146\n[stack@undercloud-0 tmp]$ cat cirros-0.4.0-x86_64-disk.img \n\u003chtml\u003e\u003cbody\u003eYou are being \u003ca href\u003d\"https://github-production-release-asset-2e65be.s3.amazonaws.com/219785102/b2074f00-411a-11ea-9620-afb551cf9af3?X-Amz-Algorithm\u003dAWS4-HMAC-SHA256\u0026amp;X-Amz-Credential\u003dAKIAIWNJYAX4CSVEH53A%2F20200219%2Fus-east-1%2Fs3%2Faws4_request\u0026amp;X-Amz-Date\u003d20200219T121437Z\u0026amp;X-Amz-Expires\u003d300\u0026amp;X-Amz-Signature\u003dd4fcb0128c40711b457ccf4918bc2fa8df17f5bf31616a8148421be5d7adc5b9\u0026amp;X-Amz-SignedHeaders\u003dhost\u0026amp;actor_id\u003d0\u0026amp;response-content-disposition\u003dattachment%3B%20filename%3Dcirros-0.4.0-x86_64-disk.img\u0026amp;response-content-type\u003dapplication%2Foctet-stream\"\u003eredirected\u003c/a\u003e.\u003c/body\u003e\u003c/html\u003e[stack@undercloud-0 tmp]$ \n[stack@undercloud-0 tmp]$ curl -Lo cirros-0.4.0-x86_64-disk.img https://github.com/cirros-dev/cirros/releases/download/0.4.0/cirros-0.4.0-x86_64-disk.img\n  % Total    % Received % Xferd  Average Speed   Time    Time     Time  Current\n                                 Dload  Upload   Total   Spent    Left  Speed\n100   619  100   619    0     0   1958      0 --:--:-- --:--:-- --:--:--  1952\n100 12.1M  100 12.1M    0     0  2132k      0  0:00:05  0:00:05 --:--:-- 2621k\n[stack@undercloud-0 tmp]$ ls -l cirros-0.4.0-x86_64-disk.img \n-rw-rw-r--. 1 stack stack 12716032 Feb 19 12:15 cirros-0.4.0-x86_64-disk.img","accounts_in_message":[],"_revision_number":4},{"id":"be331f1981bd2c2402ec21b494130807999102e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 12:57:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96f469c5a930473c83e6c753cf7b443c : SUCCESS in 7m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9d1a8cf9a733497aa9096d76b6cca9a2 : SUCCESS in 11m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/347bda7050c34e199ee5d9be2bde692d : SUCCESS in 11m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f21f8250ec6543cba8f02c3eed45c8a7 : SUCCESS in 11m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d43885dffaf4184840643a847d87e49 : SUCCESS in 12m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1657114865ef497cb8b714d08d03a7f0 : SUCCESS in 12m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bef9afb28f94a4a8ae073f0014070b9 : SUCCESS in 13m 01s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f7cd6234dc0449f6938ad3853507bf54 : SUCCESS in 39m 17s","accounts_in_message":[],"_revision_number":4},{"id":"78c4b40d8cb715909e87e44e61b6fae3ea2a772b","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-02-19 13:12:42.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"0e6d956af7b143904311b0d90be93054a6cfa0ec","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-02-19 13:53:36.000000000","message":"Patch Set 4: Code-Review+2\n\nChange looks good and as mentioned migration is not migrated to neutron repo yet\n(and indeed -L flag is useful for github downloads, personal experience)","accounts_in_message":[],"_revision_number":4},{"id":"bba01bddab7321e18f67b3908e25de687865db71","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-02-24 16:31:39.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b8cd132ad5a0689026a3903a6e226ea226bc1fd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 16:31:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9f9492b45e7b7d857577b15dd1aeb98307c82ae9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 17:29:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"092127e7b8b3e2d74d7f606629d9b88c3ff221f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 17:29:45.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e47bf8ab7784a3fa877169fafa533a0 : SUCCESS in 6m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3418dcb80bc1421688d03b6eb79aa434 : SUCCESS in 12m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccbe9370fc9540f2818df1ef3a903f3d : SUCCESS in 12m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4d66021f3ba4148b6bf9c10ef35f2a1 : SUCCESS in 14m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52677907165d452c94f42bb104e58e1d : SUCCESS in 13m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7926c8da33f540a3844294f3af564fd1 : SUCCESS in 14m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35cc2719b4ca48499379012f4cdc3103 : SUCCESS in 10m 31s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/50b55fd4d2d641129c49c9f8db49c9a8 : SUCCESS in 47m 39s","accounts_in_message":[],"_revision_number":4},{"id":"6bac495f3539004a8786b0df3d749bcf42e76f57","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 17:34:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d08b0c2a33fb433b82cbc2694f1182ec : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b583f4fdddf27a820679355668e0b1426853b7cd","revisions":{"7559622d3225101f0f650a989bc3923d33b823f4":{"kind":"REWORK","_number":1,"created":"2020-02-18 18:23:02.000000000","uploader":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"ref":"refs/changes/76/708476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/76/708476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/76/708476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/76/708476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/76/708476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/76/708476/1"}}},"commit":{"parents":[{"commit":"59e705be402cd10f332b460d898cc478ded7160d","subject":"Merge \"Fix for getting port groups while using OVS up to 2.9\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/59e705be402cd10f332b460d898cc478ded7160d"}]}],"author":{"name":"Roman Safronov","email":"rsafrono@redhat.com","date":"2020-02-18 18:22:13.000000000","tz":120},"committer":{"name":"Roman Safronov","email":"rsafrono@redhat.com","date":"2020-02-18 18:22:13.000000000","tz":120},"subject":"Fix cirros image download during resources creation for migration","message":"Fix cirros image download during resources creation for migration\n\nThe link used for downloading cirros image is now redirecting to\nsome other location. Due to this reason the image was not downloaded\nproperly and created vm instance was not usable causing migration\nto fail. This patch adds a -L flag to curl allowing it to follow\nredirects and download the image successfully.\n\nChange-Id: Iebc8278aca56f87fcf74cc2725a1ffc1197c1bf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7559622d3225101f0f650a989bc3923d33b823f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7559622d3225101f0f650a989bc3923d33b823f4"}]},"branch":"refs/heads/stable/train"},"9d3d815da4125fd50201fd398001541a4120a726":{"kind":"REWORK","_number":2,"created":"2020-02-18 20:11:45.000000000","uploader":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"ref":"refs/changes/76/708476/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/76/708476/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/76/708476/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/76/708476/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/76/708476/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/76/708476/2"}}},"commit":{"parents":[{"commit":"59e705be402cd10f332b460d898cc478ded7160d","subject":"Merge \"Fix for getting port groups while using OVS up to 2.9\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/59e705be402cd10f332b460d898cc478ded7160d"}]}],"author":{"name":"Roman Safronov","email":"rsafrono@redhat.com","date":"2020-02-18 18:22:13.000000000","tz":120},"committer":{"name":"Roman Safronov","email":"rsafrono@redhat.com","date":"2020-02-18 20:11:26.000000000","tz":120},"subject":"Fix cirros image download during resources creation for migration","message":"Fix cirros image download during resources creation for migration\n\nThe link used for downloading cirros image is now redirecting to\nsome other location. Due to this reason the image was not downloaded\nproperly and created vm instance was not usable causing migration\nto fail. This patch adds a -L flag to curl allowing it to follow\nredirects and download the image successfully.\n\nChange-Id: Iebc8278aca56f87fcf74cc2725a1ffc1197c1bf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/9d3d815da4125fd50201fd398001541a4120a726"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/9d3d815da4125fd50201fd398001541a4120a726"}]},"branch":"refs/heads/stable/train"},"b02932793237ac349466b3802939aa4ffec0e80d":{"kind":"REWORK","_number":3,"created":"2020-02-19 09:01:46.000000000","uploader":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"ref":"refs/changes/76/708476/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/76/708476/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/76/708476/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/76/708476/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/76/708476/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/76/708476/3"}}},"commit":{"parents":[{"commit":"59e705be402cd10f332b460d898cc478ded7160d","subject":"Merge \"Fix for getting port groups while using OVS up to 2.9\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/59e705be402cd10f332b460d898cc478ded7160d"}]}],"author":{"name":"Roman Safronov","email":"rsafrono@redhat.com","date":"2020-02-18 18:22:13.000000000","tz":120},"committer":{"name":"Roman Safronov","email":"rsafrono@redhat.com","date":"2020-02-19 08:59:08.000000000","tz":120},"subject":"Fix cirros image download during resources creation for migration","message":"Fix cirros image download during resources creation for migration\n\nThe link used for downloading cirros image is now redirecting to\nsome other location. Due to this reason the image was not downloaded\nproperly and created vm instance was not usable causing migration\nto fail. This patch sets a correct link for cirros image.\n\nChange-Id: Iebc8278aca56f87fcf74cc2725a1ffc1197c1bf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b02932793237ac349466b3802939aa4ffec0e80d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b02932793237ac349466b3802939aa4ffec0e80d"}]},"branch":"refs/heads/stable/train"},"b583f4fdddf27a820679355668e0b1426853b7cd":{"kind":"REWORK","_number":4,"created":"2020-02-19 12:14:24.000000000","uploader":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"ref":"refs/changes/76/708476/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/76/708476/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/76/708476/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/76/708476/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/76/708476/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/76/708476/4"}}},"commit":{"parents":[{"commit":"59e705be402cd10f332b460d898cc478ded7160d","subject":"Merge \"Fix for getting port groups while using OVS up to 2.9\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/59e705be402cd10f332b460d898cc478ded7160d"}]}],"author":{"name":"Roman Safronov","email":"rsafrono@redhat.com","date":"2020-02-18 18:22:13.000000000","tz":120},"committer":{"name":"Roman Safronov","email":"rsafrono@redhat.com","date":"2020-02-19 12:11:22.000000000","tz":120},"subject":"Fix cirros image download during resources creation for migration","message":"Fix cirros image download during resources creation for migration\n\nThe link used for downloading cirros image is now redirecting to\nsome other location. Due to this reason the image was not downloaded\nproperly and created vm instance was not usable causing migration\nto fail. This patch sets a correct link for cirros image. It also\nadds a -L flag to curl allowing it to follow redirects and download\nthe image successfully.\n\nChange-Id: Iebc8278aca56f87fcf74cc2725a1ffc1197c1bf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b583f4fdddf27a820679355668e0b1426853b7cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b583f4fdddf27a820679355668e0b1426853b7cd"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
