)]}'
{"id":"openstack%2Fnetworking-ovn~715949","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~I1d576c94815de3b8f07c64936006477d1938d88c","project":"openstack/networking-ovn","branch":"stable/train","hashtags":[],"change_id":"I1d576c94815de3b8f07c64936006477d1938d88c","subject":"Mark TestVirtualPorts tests as unstable","status":"MERGED","created":"2020-03-30 14:07:31.000000000","updated":"2020-04-08 19:53:36.000000000","submitted":"2020-04-08 19:37:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715949-1586374661427-a14ab837","meta_rev_id":"3d66db09fe2d7b9cdc0a604564b4f04d37c339a8","_number":715949,"virtual_id_number":715949,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 19:37:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2020-04-06 15:40:34.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":1,"date":"2020-04-07 10:21:27.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-30 14:07:48.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-04-01 12:25:55.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-04-02 09:40:04.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-04-06 15:40:34.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-04-08 19:37:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4132c2df52f14220f34767f309aab5289abd932f","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-03-30 14:07:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c7b35953f67ccd313de13266466c246bbd72345","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 17:16:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d08ea31c573f42a9a8ea411a706d7fe9 : SUCCESS in 7m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9346a0a16b4c42a49bcde7d15987b542 : SUCCESS in 10m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d6832d150fb422888869529ee5bd684 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b538108b0e394343aa8eec7e69e0b53e : SUCCESS in 12m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6e5bc40b0654d81882bcbd75135e4d8 : SUCCESS in 10m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba62b57f020b4a289612a0728640abf3 : SUCCESS in 11m 07s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/f65e2c4ddff14875bc57cd4f21bb1934 : SUCCESS in 2h 36m 21s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/38f796a1dd6d4e7cacdee4a6ae307b58 : SUCCESS in 2h 50m 15s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/4118d97cff0b43b6bdf14d6916377506 : SUCCESS in 1h 27m 47s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a68655717d7b4d63814f4f9c06f23b42 : SUCCESS in 1h 48m 44s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/61f884ded4f94a90969753c948a64d98 : SUCCESS in 19m 50s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/8c54969196b24e5badd9ae5e4f3d84a2 : SUCCESS in 24m 33s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9a86efd252194361999e6a74f92e7498 : SUCCESS in 2h 59m 25s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/54a4c45bd8e54883981cb1e2a03c7320 : SUCCESS in 1h 33m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0857a29c742041a0bb00bcb52e65c5a4 : SUCCESS in 12m 20s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/58a48cef03d04a31ae8b109a14fab9e1 : SUCCESS in 2h 12m 39s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/75e9abe1a5524ec2b310025137b86f80 : FAILURE in 2h 18m 56s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/65875d18093046f1ac153607c9004502 : FAILURE in 51m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0bb5e2e9eb67c21afdb773348f0010d28ad7f854","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-01 08:35:50.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fb1ef23c59b0d692fa087b3051acc49869628b77","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-01 08:36:20.000000000","message":"Patch Set 2: Code-Review+2\n\nJust fixed a copy/paste issue in commit message","accounts_in_message":[],"_revision_number":2},{"id":"98f48ba332caf8b8d5a2fc8c104c287522bac1e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 11:55:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2d2206cf8c049fba9bc7a91db53882b : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4269a5fdd419458589077fb864c322d0 : SUCCESS in 9m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de2636312a9a4ec7bbe47def3fa2f9ad : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42c3d5ae724e4b02a86f2116b67db009 : SUCCESS in 9m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/767ea59fcd7d46c3b805c47b56a70668 : SUCCESS in 10m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43e65ed6385a4040a55fcf5f6dd5d8d4 : SUCCESS in 10m 52s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/941d4ad1b91547ae8c3ebf74f5306de2 : FAILURE in 2h 13m 49s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/a00624da8bfa40839ee9ee4090846853 : SUCCESS in 2h 34m 37s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/0650ca011f084aeeb0b178e8029679ab : SUCCESS in 1h 34m 47s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e74984f7f3164e09a92b2c1ee1b0c185 : SUCCESS in 2h 05m 00s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/9d951d2486ef406e89f8ef1c91960c43 : SUCCESS in 22m 59s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/222b634448f24c9088cf6a0ad70d2304 : SUCCESS in 23m 52s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/979ddd30114548c19471288fd6136eb4 : SUCCESS in 2h 45m 19s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/703b05db01e5423c9c2f48d1867f2818 : SUCCESS in 1h 37m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ff36ba2230742bfa3c4cacad8135221 : SUCCESS in 11m 28s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/13e0c33605534fabb5bfda3696cbf84d : FAILURE in 2h 19m 45s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d824cceaacbe40fe9e842fc21a2a5308 : POST_FAILURE in 1h 47m 36s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9c950a0c6989473087b457a7e9f997f4 : FAILURE in 49m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3b813ada3680b830ba3895525faadb61dc9d923a","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-04-01 15:01:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"66d2ee2398ec40f7a87dafda153423305f5d829f","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-04-01 15:01:22.000000000","message":"Patch Set 3:\n\nRebased on top of external ports to avoid conflict","accounts_in_message":[],"_revision_number":3},{"id":"9f4b26ab72cdf03be59cd25f378e7d2f60854974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 18:40:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a5eaa0677a547b5ade341ada2561aa3 : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/33b7a9b2885e4c848a963cc316832548 : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3808d42d65d44b28f39465802480891 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5920dea3c395471eb51d24102505ab00 : SUCCESS in 9m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff29bdbf53e4425e9d10dd0ee1ea32c6 : SUCCESS in 10m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71b0b8c5c1544f8eb190b963023e1ee5 : SUCCESS in 10m 54s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/f74666eb20fb42d0ae5a68e4b944ea4f : SUCCESS in 2h 10m 46s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/e902600f513c4246a8ab4c377a6bed01 : SUCCESS in 2h 19m 40s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/3e97e3dab3f04d059ef2180204acbceb : SUCCESS in 2h 03m 09s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/936708913adc49e1ae2d8246b87b3a4b : TIMED_OUT in 2h 17m 14s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/b90f69c96f7b408d97a33cc4aeff7125 : SUCCESS in 22m 19s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/01643f9c43aa403ebb22e0f4741d232a : SUCCESS in 28m 35s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/db1c36d9e33c422f86eed30a020b94e3 : SUCCESS in 3h 17m 30s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/8668461535f24ec486ebffcecc291643 : SUCCESS in 1h 34m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d496b156733a46428944f8943bc86217 : SUCCESS in 12m 21s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c9ab6be4833e424eb0b89a7b01153d79 : FAILURE in 2h 12m 14s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/4035e197552a4254b65fa5d519f86073 : FAILURE in 2h 05m 58s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3681101b446f438c9fadcb5b0f7d83b1 : FAILURE in 47m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ce724ff4bdc6a43cd6be9b4f568d8e537a79b2e8","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-02 09:39:43.000000000","message":"Patch Set 3:\n\nrecheck TIMED_OUT rally","accounts_in_message":[],"_revision_number":3},{"id":"1936dd5666b68243be197f5565a2edef7133d6c6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-02 09:40:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"acfd48cc3c81bd69adc580c98c7cbb9e0b532151","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 12:24:18.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad5d84a5e9c84b0d963e33b6de0dbd2a : SUCCESS in 7m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5746e61583994e3aa6dcbb864a7dd8d6 : SUCCESS in 9m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8e8bb1034744bd58759c63a9a48169a : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad98fdc0afc74aa99dbd24c89da639c5 : SUCCESS in 10m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c3f4dec487e44feafab5bb6c19a6b1f : SUCCESS in 10m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8b6eb198f3b4be988cdb3c2a2db1430 : SUCCESS in 14m 36s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/5e19f78fbf1c49c08847133a3eb57872 : FAILURE in 2h 21m 24s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/61d72de9b3894e958d7ad866667880e8 : SUCCESS in 2h 27m 23s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/39f342e6c76e4602b62ecbb7cdca1078 : FAILURE in 29m 49s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eaebdd641cd845858f86b5d24f9bc931 : SUCCESS in 2h 06m 40s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/e82fff5dc9a748b9827f8b63e8f47ab0 : SUCCESS in 25m 06s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/b6d5c141d0af4ab58cc666c20b03ed01 : SUCCESS in 21m 32s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/052a83c1c1664c4289399e9af7e29e6a : SUCCESS in 2h 28m 19s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/123d3ac4f9fc4ffd85e777e8acb040e7 : SUCCESS in 1h 20m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb04005c619e4c08bce25a38b4f4adfe : SUCCESS in 11m 55s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6535beb5a4864921b30313c4c0ecc2ff : FAILURE in 2h 09m 27s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/5b137617c4344228b1f94690f9fbbef9 : FAILURE in 2h 42m 14s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b63f338a91d54419ab8edcc316670fd8 : FAILURE in 49m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"497da7ee028f4bc295445ea199a0cd9df469ef5a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-02 12:59:41.000000000","message":"Patch Set 3:\n\ntempest requires Python \u0027\u003e\u003d3.6\u0027 but the running Python is 2.7.17\n(in networking-ovn-tempest-dsvm-ovs-release-python2 job)","accounts_in_message":[],"_revision_number":3},{"id":"4641528f00da36d988b19c97f78991a33600398c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-03 11:24:34.000000000","message":"Patch Set 3:\n\nrecheck https://review.opendev.org/#/c/716970/ merged","accounts_in_message":[],"_revision_number":3},{"id":"8cc4a64d4f4293ab17e4dcadf65e49889d04b639","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 14:34:26.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf79a17303a94abeab0b4e5a7add9b00 : SUCCESS in 7m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/edd835fa44cd4558adc3960fff8ed1cf : SUCCESS in 11m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12482c1497d64f4b8f0ac43d130be9c3 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aeb86adc1978489388befb37c29b746a : SUCCESS in 11m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d8a2d7137fa04f568ded3479c2177fd4 : SUCCESS in 11m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a57e9023f5e4dca94b7baf98f812e1c : SUCCESS in 12m 39s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/b8c7d239535149539875fce06dfb6409 : SUCCESS in 2h 24m 04s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/faa75ffcf3bb41c787ae1b0e7364dee4 : SUCCESS in 2h 56m 54s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/c0cc0006e30f4d45be4140e0bbc29d07 : FAILURE in 2h 24m 05s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b77ffd61a7a2436e830bfc7e05cae30e : SUCCESS in 1h 59m 58s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/660414eb4e574d288ccd4deebb64a589 : SUCCESS in 22m 51s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/8fcc1fea3bb34679bde515c9c0cf0422 : SUCCESS in 21m 21s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0280fbd753064871afcd8a75d1df9f5c : SUCCESS in 2h 31m 21s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/a4ac4846cecb46ca87cb27d5de1e57c7 : SUCCESS in 1h 38m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57ee79d4259c47f4a8993c30e85d91a2 : SUCCESS in 12m 18s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c2f68adb30d342f8b82a0a85c4463792 : SUCCESS in 2h 22m 40s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/104b84bcaa7049599f0d4e33794c4d54 : FAILURE in 2h 09m 55s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53b041c80a3a4fa9b721b234ea8f5fc0 : FAILURE in 48m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cacca27f17693864637c5ea0208e7f7f2b323bf6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-03 17:11:07.000000000","message":"Patch Set 3:\n\nrecheck tempest.api.compute.servers.test_servers_negative","accounts_in_message":[],"_revision_number":3},{"id":"2462a2c3ffb0d9ba3b7d1d64c46997e7ed859981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 20:04:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/455e1a02be5d4581915cac228245511f : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa024234212847ae943b6485a1d9877c : SUCCESS in 11m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a887abb29f946c8bcf11cbb71c50822 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec45decc7c2b4ff4af7f49e80c311756 : SUCCESS in 10m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f7b072c9a5e4081871ff346d89abd78 : SUCCESS in 13m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf623e11467148cb9e70f2d4518d5c64 : SUCCESS in 13m 20s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/94588fddcfee42afa3e9fa9b223225d8 : SUCCESS in 2h 08m 15s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/7331ecad8ffd474cbb20dd1e1d4a9c6e : SUCCESS in 2h 49m 09s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/e0e265dc363c4f949b44d93bcbf888db : SUCCESS in 2h 11m 35s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c47564505e414e0dbba72bd13734f083 : SUCCESS in 1h 48m 21s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/ef87d7a83a004259bb084c2381cc057e : SUCCESS in 24m 31s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/d187c23b6a8d47e9886b86f8a1e49a1b : SUCCESS in 21m 58s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/5a7043e7c85d4de594bf5766880c7e20 : SUCCESS in 2h 31m 10s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/a925b98f708c46969b18dac8f52e46ac : SUCCESS in 1h 22m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/decab4e7c11a4ffea75fe6fc316909d1 : SUCCESS in 11m 54s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bb246d32bac84ec1a2ac88379915ece9 : FAILURE in 2h 17m 38s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/a7a1e61831d24e9d914eb2b387c1c768 : FAILURE in 2h 03m 49s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a55be6a3e1c45a69917d01a5bf7cd0b : FAILURE in 43m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"85f4c1a5475e336978e12298e87ca9c94ebef461","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-03 20:28:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dc82c55e14a7ec031952996bb5a08df0967cbe88","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-04-06 07:55:57.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"0d0f35fcb98cdaa42fe8243ebe3039c56c19becd","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-04-06 07:57:28.000000000","message":"Patch Set 4:\n\nI think I will rebase this against master, I\u0027ve seem enough functional tests failures due to this problem. I can fix the merge conflicts on the other patches","accounts_in_message":[],"_revision_number":4},{"id":"935bdff404f028fd91c75b96b7250388c0b11f99","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-04-06 07:58:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"36c5e913a7f6a7756037261fb0ffc1b8465460a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 11:23:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1ca6f3db7b14de5a329ad925a010c21 : SUCCESS in 6m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab09cf26064545c39856f7f69acfb9a5 : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1d8da92c45545ada07d20c14269071e : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b79c77693b874eb1b34de4df95f5b83e : SUCCESS in 9m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01d9b665796a43ffaaa8995db0b4010c : SUCCESS in 9m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6bfe9934a044bb1bd2b67ef15adb540 : SUCCESS in 13m 38s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/741d8efc42824ba3bcf2f0353b8cbd6a : SUCCESS in 2h 30m 22s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c6458e4eb81e471498ee0a68ede277c0 : SUCCESS in 2h 19m 28s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/70ab794d0d5741679263aadd9c27b35b : SUCCESS in 2h 45m 20s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0e4571cc747340e69762d9a5222d128e : TIMED_OUT in 2h 13m 57s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/dfa70f6b852a411cb92c2b8b6a169373 : SUCCESS in 21m 17s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/c78ea2fd427e45d6933a56a4f20cc763 : SUCCESS in 41m 24s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f044932759d447ccb409beb3ec382b88 : SUCCESS in 3h 21m 48s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/2bd5893ba26e4edebd0332e9c3e5223a : SUCCESS in 2h 00m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd2508761eb244bd8c552c35babdf8de : SUCCESS in 13m 22s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9ad7efffc43e4348a61020a03769501a : FAILURE in 2h 28m 24s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/aa5d1e6bc7ac4ed38e3bc1f8f23046ee : FAILURE in 2h 21m 15s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/57c853773f914e73932ca5e973698a3e : FAILURE in 48m 03s (non-voting)\n\nWarning:\n  Comments left for invalid file networking_ovn/tests/unit/ml2/test_mech_driver.py","accounts_in_message":[],"_revision_number":5},{"id":"ce92f639d9650d78d93a7c8f7fd55c51f64953de","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-04-06 12:10:23.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"b1731e009fa7f2c0f6f973349af6c63358d675a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 15:10:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab2224ce49a24a1a88822fe8cde0fb5e : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/86b8ac9a5cb046979086f9375d3c2ebc : SUCCESS in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6523ed2c6e784b989c47bc019db99182 : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91267f4649264ad98983263c4846eca2 : SUCCESS in 8m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ec388e73ab54ab0a21cd4e4582fb3b9 : SUCCESS in 11m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9aaf4b202add4eb490d4a1ab24f7ea92 : SUCCESS in 11m 53s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/b6868f90b7ee40fbac44fd2654c97431 : SUCCESS in 2h 31m 46s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/dc35abdb7544481ab6539e1224aa70e5 : SUCCESS in 2h 24m 21s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/b64c08d3220f49dc89353e2a5f981358 : SUCCESS in 2h 28m 38s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/df289f6bc80a4905946df471b4b2c579 : SUCCESS in 2h 09m 02s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/a675d3931d27429db7b8ac3a283f77c8 : SUCCESS in 22m 17s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/baa196688133417d9aceb531c78acc05 : SUCCESS in 22m 47s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ffb8f6d1ffa54beab0b1fb9071ca5925 : SUCCESS in 2h 55m 19s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/1ba993fb52dd462a9764de26273870ba : SUCCESS in 1h 58m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c444359ea629494eb074b5836458f4fd : SUCCESS in 10m 42s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b4c34d8f007e486981615a713da23958 : SUCCESS in 2h 27m 48s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b98cfe300e39493bbf049fe8283a89e3 : FAILURE in 2h 09m 36s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f5459c7f1dd64bcca31af217e7397f74 : FAILURE in 44m 45s (non-voting)\n\nWarning:\n  Comments left for invalid file networking_ovn/tests/unit/ml2/test_mech_driver.py","accounts_in_message":[],"_revision_number":5},{"id":"097f109240082604cd03dc5684218700ea77680e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-06 15:40:34.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b717de97cad6e77f7d71dbf134f0379178dde1e8","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-04-07 10:21:27.000000000","message":"Patch Set 5: Workflow+1\n\nI\u0027m gonna approve this even tho it conflicts with the external ports because the gate is failing in multiple places, at least this should alleviate the functional job runs","accounts_in_message":[],"_revision_number":5},{"id":"a54515b0545ea71ebab95b884ede876ccb4b3ebc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 10:21:44.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a05e9c26928535e7b349c605122f06a9dc2988fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 13:22:56.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/531b39c3568e48b2965fe7dffeac7cf2 : SUCCESS in 8m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/439fe2c4b85e4e8a8d28e06ff77a31ed : SUCCESS in 11m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b51e2ba7bdd14478a5401089878d5d23 : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7485062211f146e881bd22d4c8d85cb7 : SUCCESS in 11m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/82d9db775f63443a888c5441c9ed4338 : SUCCESS in 12m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/842ade77e654429f8c1f50e7f6246e30 : SUCCESS in 13m 31s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/ca0c7a4bf311441da84f0cf975b53cdf : SUCCESS in 2h 37m 30s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/fad3a1f3fea74a269ce068e5d0b82d5c : SUCCESS in 2h 17m 41s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f1618f63f93d432fad34d9d88ccc1078 : FAILURE in 39m 32s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/042917e6d88144f1b355cc6b2cd85975 : SUCCESS in 23m 16s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/71d5a64356ce45acb44ec81446070334 : SUCCESS in 21m 47s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e880fd1a23a545c9aff79a854c1b9d3d : SUCCESS in 2h 53m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10bf6befec0340f5b723fcfc6210e1ea : SUCCESS in 13m 59s\n\nWarning:\n  Comments left for invalid file networking_ovn/tests/unit/ml2/test_mech_driver.py","accounts_in_message":[],"_revision_number":5},{"id":"9cd65197bd6fc5026f0f3fbae7f5225770a0ad75","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-04-07 14:00:32.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"d158962ac018eb2c2643c5b73757a9e8c2d0adf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 17:06:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/997a52ac00884522b7820ef402297485 : SUCCESS in 6m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b7d2ba2ef7d42a9bdabe91ed203ecff : SUCCESS in 12m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01b82c4e56a4407193f1498ac9b8dda0 : SUCCESS in 8m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7e0a7e8f9454ec6987acc547327d2a9 : SUCCESS in 11m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4b51334ea8646a7aef197b2cbd84dbf : SUCCESS in 14m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35ee9da9442046ce9a9968d5549c4d9d : SUCCESS in 14m 06s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/077150327ae54e9482f5826f95644f06 : SUCCESS in 2h 35m 56s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/9bfa64ffc600411d9ae318c5b85c936c : SUCCESS in 3h 00m 21s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/2d0e53943e534347a99f372ea8702f9e : SUCCESS in 2h 50m 39s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/763a4191a13e4caaa14e494f1d4f6bd9 : FAILURE in 35m 04s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/73a4aa4eff194406ba3106453afb31c8 : SUCCESS in 18m 42s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/54b9242f1b27436e94667868454a6e6c : SUCCESS in 26m 56s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4cc662ac969d4ed799317d0c35a488ef : SUCCESS in 2h 36m 24s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/349f2b31b9ba49ca8b963a503c439a3c : SUCCESS in 1h 36m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d66ca6a77e64544beb8ed883ac93ac7 : SUCCESS in 12m 58s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7f88eab8a8ea44669fdb8e211e096714 : FAILURE in 2h 25m 14s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/1e277693f72c4190b1f067e6447fe1a2 : FAILURE in 2h 08m 12s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ad98cd3c2a744119d4293d82cb908e4 : FAILURE in 49m 42s (non-voting)\n\nWarning:\n  Comments left for invalid file networking_ovn/tests/unit/ml2/test_mech_driver.py","accounts_in_message":[],"_revision_number":5},{"id":"bdafc330299a85e8a734f87d1cf21f18c3b20ca2","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-04-08 13:10:09.000000000","message":"Patch Set 5:\n\nrecheck\n\nhttps://review.opendev.org/718383 is now merged","accounts_in_message":[],"_revision_number":5},{"id":"6893eaf9ebc5b4397f4abcf2a5b39e4f4be412d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 16:35:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca9800010f9f4a63811563999076186e : SUCCESS in 7m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/03bf768542e347dcace5f06dd699ddf4 : SUCCESS in 11m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6561e1791e184b7e8376bdd10be6ac76 : SUCCESS in 10m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afcf645afcdb48cfb0a5433b4da5c7d7 : SUCCESS in 13m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0d5521d06ae4854a1139fdf45c7ffe6 : SUCCESS in 14m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/073804b1a4344576beb16ccfa848d8ca : SUCCESS in 15m 15s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/9cb98f5556ca498499cd4d01d40c29cd : SUCCESS in 3h 01m 40s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/6f5f972e855f4a80baa626dc4483b32a : SUCCESS in 2h 22m 53s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/50016829d8cd4a049d63409d3c810bed : SUCCESS in 2h 32m 22s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8e2695fb2ac74ba983c997c5714c5164 : SUCCESS in 2h 06m 08s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/1afc7f609e614b839e52d444b79c1815 : SUCCESS in 25m 54s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/622ee89e11074401bdbe6ceb8bdfe9ae : SUCCESS in 24m 46s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b408eec49aff493d9e75b3582d031fcb : SUCCESS in 2h 47m 45s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/c62fdf070b1742d48d0b95e23ea5b32d : SUCCESS in 1h 25m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0af7e1c32ef43948fc8a15862479aaf : SUCCESS in 13m 03s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/03ab24fef07c4864ad6c22ed6aea3cdf : FAILURE in 2h 23m 19s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/169f9d5a899140aab549e6aeb0d916c0 : FAILURE in 2h 01m 52s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8b2aa3a17274462c802ddc1e707d61eb : FAILURE in 45m 06s (non-voting)\n\nWarning:\n  Comments left for invalid file networking_ovn/tests/unit/ml2/test_mech_driver.py","accounts_in_message":[],"_revision_number":5},{"id":"cc9848ce8058bf93dfb288a863d1fb1212cd2f71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 16:36:05.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e6ce561b09ba06c0b84247a90e66e3dcc1596ad2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 19:37:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"267871b21f04a8cef0978fc0ff3a142e4077a42d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 19:37:41.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/424a570a72654a128aa6c291e4231d10 : SUCCESS in 7m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4f8162e4c54e4532beacf26899c14998 : SUCCESS in 11m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92ce9c0ec4a644f791f2835e69b180e1 : SUCCESS in 8m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bb20be2c5f6483e9f1ca4ddada51dc6 : SUCCESS in 14m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d92f9aced93e4b0998f42136eed3f1cf : SUCCESS in 11m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a2619b5b9e8436d8df6c526a8c00f70 : SUCCESS in 10m 58s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/4aa1fff2b1504bf2aefaf4951c5dc7f8 : SUCCESS in 2h 21m 32s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/df80db5430d94c58bb4114dea177f5fb : SUCCESS in 2h 08m 55s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d750530c820a41a9841639af89792295 : SUCCESS in 1h 59m 11s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/59ba0e2d4ba94f9e84db659742ae06c8 : SUCCESS in 23m 26s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/eca063fd8da44ade8cb64d64048cbd04 : SUCCESS in 24m 05s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/5f00f516226a4d06ba5a11b818da6f37 : SUCCESS in 2h 55m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bda010bffcd04a2794d310c2722f02f9 : SUCCESS in 10m 17s\n\nWarning:\n  Comments left for invalid file networking_ovn/tests/unit/ml2/test_mech_driver.py","accounts_in_message":[],"_revision_number":5},{"id":"3d66db09fe2d7b9cdc0a604564b4f04d37c339a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 19:53:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7610b2ae69d424b8c3a0b72191a9bfa : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ba456f28c1c9709ef8477fec5823181d04c52e9b","revisions":{"d108255e4089170fc8c6d4c3319e60f5b3631da3":{"kind":"REWORK","_number":1,"created":"2020-03-30 14:07:31.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/49/715949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/49/715949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/1"}}},"commit":{"parents":[{"commit":"dfd5ee3aff62c0c09b980e289ef35eaeafcd7053","subject":"Merge \"Clear lsp.addresses always if port is OVN LB VIP port.\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/dfd5ee3aff62c0c09b980e289ef35eaeafcd7053"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-03-30 14:03:51.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-03-30 14:07:05.000000000","tz":60},"subject":"Mark TestVirtualPorts tests as unstable","message":"Mark TestVirtualPorts tests as unstable\n\nBased on debug performed in [1] the TestVirtualPorts are unstable. At\nlease one of tests per 20 retries of functional job fail in various\nplaces.  We don\u0027t think it is worth to spend more time on debugging\nthose tests, because there is a possible easier solution: set virtual\nport type based on DEVICE_TYPE, and do not rely on allowed address pairs.\n\n[1] https://review.opendev.org/#/c/712888/\n\nRelated-Bug: #1865453\n\nChange-Id: Iffef8ceb9ca458820a7b3e604a53281f7e93b810 time on debugging\nthose tests, because there is a possible easier solution: set virtual\nport type based on DEVICE_TYPE, and do not rely on allowed address pairs.\n\n[1] https://review.opendev.org/#/c/712888/\n\n(cherry-picked from neutorn Iffef8ceb9ca458820a7b3e604a53281f7e93b810)\n\nChange-Id: I1d576c94815de3b8f07c64936006477d1938d88c\nRelated-Bug: #1865453\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/d108255e4089170fc8c6d4c3319e60f5b3631da3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/d108255e4089170fc8c6d4c3319e60f5b3631da3"}]},"branch":"refs/heads/stable/train"},"0e9adeac8096bada0507c59cd5a446ed1f1ef298":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-01 08:35:50.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/49/715949/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/49/715949/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/2"}}},"commit":{"parents":[{"commit":"dfd5ee3aff62c0c09b980e289ef35eaeafcd7053","subject":"Merge \"Clear lsp.addresses always if port is OVN LB VIP port.\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/dfd5ee3aff62c0c09b980e289ef35eaeafcd7053"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-03-30 14:03:51.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-04-01 08:35:33.000000000","tz":0},"subject":"Mark TestVirtualPorts tests as unstable","message":"Mark TestVirtualPorts tests as unstable\n\nBased on debug performed in [1] the TestVirtualPorts\nare unstable. At lease one of tests per 20\nretries of functional job fail in various\nplaces.\nWe don\u0027t think it is worth to spend more\ntime on debugging those tests, because there is\na possible easier solution: set virtual port\ntype based on DEVICE_TYPE, and do not rely\non allowed address pairs.\n\n[1] https://review.opendev.org/#/c/712888/\n\n(cherry-picked from neutron Iffef8ceb9ca458820a7b3e604a53281f7e93b810)\n\nChange-Id: I1d576c94815de3b8f07c64936006477d1938d88c\nRelated-Bug: #1865453\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/0e9adeac8096bada0507c59cd5a446ed1f1ef298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/0e9adeac8096bada0507c59cd5a446ed1f1ef298"}]},"branch":"refs/heads/stable/train"},"3c7a9f942ef04b9c2fe3acc9e393818c54448690":{"kind":"REWORK","_number":3,"created":"2020-04-01 15:01:06.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/49/715949/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/49/715949/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/3"}}},"commit":{"parents":[{"commit":"3b8f606254d3ed7b34c4ded393009e31a711f321","subject":"[OVN] Add support for external ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/3b8f606254d3ed7b34c4ded393009e31a711f321"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-03-30 14:03:51.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-04-01 15:00:51.000000000","tz":60},"subject":"Mark TestVirtualPorts tests as unstable","message":"Mark TestVirtualPorts tests as unstable\n\nBased on debug performed in [1] the TestVirtualPorts\nare unstable. At lease one of tests per 20\nretries of functional job fail in various\nplaces.\nWe don\u0027t think it is worth to spend more\ntime on debugging those tests, because there is\na possible easier solution: set virtual port\ntype based on DEVICE_TYPE, and do not rely\non allowed address pairs.\n\n[1] https://review.opendev.org/#/c/712888/\n\n(cherry-picked from neutron Iffef8ceb9ca458820a7b3e604a53281f7e93b810)\n\nChange-Id: I1d576c94815de3b8f07c64936006477d1938d88c\nRelated-Bug: #1865453\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/3c7a9f942ef04b9c2fe3acc9e393818c54448690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/3c7a9f942ef04b9c2fe3acc9e393818c54448690"}]},"branch":"refs/heads/stable/train"},"aa4eab0ee51a9e4e0d4336ea3bca2a0c2043319f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-04-06 07:55:57.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/49/715949/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/49/715949/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/4"}}},"commit":{"parents":[{"commit":"7497ef5d44bb8fb4290fc39874238593d360ee07","subject":"[OVN] Add support for external ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7497ef5d44bb8fb4290fc39874238593d360ee07"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-03-30 14:03:51.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-04-06 07:55:57.000000000","tz":0},"subject":"Mark TestVirtualPorts tests as unstable","message":"Mark TestVirtualPorts tests as unstable\n\nBased on debug performed in [1] the TestVirtualPorts\nare unstable. At lease one of tests per 20\nretries of functional job fail in various\nplaces.\nWe don\u0027t think it is worth to spend more\ntime on debugging those tests, because there is\na possible easier solution: set virtual port\ntype based on DEVICE_TYPE, and do not rely\non allowed address pairs.\n\n[1] https://review.opendev.org/#/c/712888/\n\n(cherry-picked from neutron Iffef8ceb9ca458820a7b3e604a53281f7e93b810)\n\nChange-Id: I1d576c94815de3b8f07c64936006477d1938d88c\nRelated-Bug: #1865453\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/aa4eab0ee51a9e4e0d4336ea3bca2a0c2043319f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/aa4eab0ee51a9e4e0d4336ea3bca2a0c2043319f"}]},"branch":"refs/heads/stable/train"},"ba456f28c1c9709ef8477fec5823181d04c52e9b":{"kind":"REWORK","_number":5,"created":"2020-04-06 07:58:45.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/49/715949/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/49/715949/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/49/715949/5"}}},"commit":{"parents":[{"commit":"4114bc5b8a135f46ba8da2bbc755fceaa148b547","subject":"Merge \"[OVN] Reduce the number of watched tables in MetadataProxyHandler\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4114bc5b8a135f46ba8da2bbc755fceaa148b547"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-03-30 14:03:51.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-04-06 07:58:42.000000000","tz":60},"subject":"Mark TestVirtualPorts tests as unstable","message":"Mark TestVirtualPorts tests as unstable\n\nBased on debug performed in [1] the TestVirtualPorts\nare unstable. At lease one of tests per 20\nretries of functional job fail in various\nplaces.\nWe don\u0027t think it is worth to spend more\ntime on debugging those tests, because there is\na possible easier solution: set virtual port\ntype based on DEVICE_TYPE, and do not rely\non allowed address pairs.\n\n[1] https://review.opendev.org/#/c/712888/\n\n(cherry-picked from neutron Iffef8ceb9ca458820a7b3e604a53281f7e93b810)\n\nChange-Id: I1d576c94815de3b8f07c64936006477d1938d88c\nRelated-Bug: #1865453\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ba456f28c1c9709ef8477fec5823181d04c52e9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ba456f28c1c9709ef8477fec5823181d04c52e9b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
