)]}'
{"id":"openstack%2Fnetworking-ovn~718669","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~I62f53dbd862c0f38af4a1434d453e97c18777eb4","project":"openstack/networking-ovn","branch":"stable/train","topic":"bug/1861510","hashtags":[],"change_id":"I62f53dbd862c0f38af4a1434d453e97c18777eb4","subject":"Don\u0027t reschedule hosts unless we need to","status":"MERGED","created":"2020-04-09 11:22:33.000000000","updated":"2020-04-16 01:24:36.000000000","submitted":"2020-04-16 01:22:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718669-1587000179190-69063b05","meta_rev_id":"c846f82a2a59c9aa8fcfd30c135e268758dd3f40","_number":718669,"virtual_id_number":718669,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2020-04-15 19:03:05.000000000","_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-16 01:22:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2020-04-15 15:18:47.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2020-04-15 15:43:30.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-14 15:10:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":1,"date":"2020-04-14 16:52:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2020-04-15 15:43:30.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-14 15:10:26.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-04-14 16:52:49.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-04-15 15:18:47.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-04-15 15:43:30.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-04-16 01:22:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"553df95cef317f3674ec364123050d9ce1918d2d","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-09 11:22:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96bd8fa42b40a4b7326723cb34ad7c6939e194a2","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-09 11:46:01.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"92ea329ab73fcef0dbc50a09cd463b8b29c84548","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 14:53:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/291afc365a8a4ad1a54632c5826666e4 : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7297db2aa4d1474b94e7f8c7fd922287 : SUCCESS in 13m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4302536b16e542749e4e855b73b939b1 : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04b275d5a2454f3589c5ee23f8d8149f : SUCCESS in 10m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0d3bb6891de4aa4953f294147c3c82f : SUCCESS in 12m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5866155c03d54fc583d54e18388a8d8a : SUCCESS in 12m 29s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/cc3a150d905e455bb1a631cd534b4bbf : SUCCESS in 2h 08m 15s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/67c6d3ee7a5d48d88d09c9b941c063b7 : SUCCESS in 2h 07m 04s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/a4931f42a5474e7f90de596fc76414c7 : SUCCESS in 2h 38m 35s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1d0ae97deb874a6da54b941ee8728533 : FAILURE in 2h 08m 41s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/924e236947ac4632bd9b7b0efe428a5d : SUCCESS in 35m 33s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/a844819840764b008fafabafb4a20aea : SUCCESS in 23m 36s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2a8c2ebed3ac4c219c68631b96684ce0 : SUCCESS in 2h 59m 51s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/a34a61571da247499cb339a4ea2851cc : SUCCESS in 1h 26m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a24cd6ce52645b48c6534f6195e4df4 : SUCCESS in 11m 15s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5d38be2090ab48a9b416440c1bc1823f : FAILURE in 2h 15m 42s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/2deb10ffe6a749a7952533ff709996b3 : FAILURE in 1h 53m 39s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0580929fc18b4bdfbc7c978bcba0e8ca : FAILURE in 46m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"066db66eeb98926b840921625de3529e23aad140","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-09 15:14:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dab0bcd5dc6ae4425cf64d1eca0b658cad1f310c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-09 15:14:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8fd38cf4d1fc2ea8343be852c208d74ccc6ab201","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 18:48:30.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f038e1ffcc484716a5dda033be7e9b86 : SUCCESS in 9m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/96ff593949b648f38c2336956b8c10cd : FAILURE in 12m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bc36135bcfe4cb3add11a269bacca29 : SUCCESS in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3ab61a8ee6447639a59e9a70dc2cc65 : SUCCESS in 10m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c732cab9dc7e48048258259e922a1bb4 : SUCCESS in 13m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db1930143add4d96b3be932a834b58b4 : SUCCESS in 11m 34s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/31ad3a3e7bfa487f85feeaf395e1ba3f : SUCCESS in 2h 20m 40s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/5ef87212bb9344cfba38f3ca82fce096 : SUCCESS in 2h 47m 44s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/756aaa6eeb76450db48a9586d524da1e : TIMED_OUT in 3h 09m 57s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f4fbb4a421b14bb0bcf41d67d9f77ff0 : SUCCESS in 2h 09m 39s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/f1ad9ecc0dcc459d9fb0fbbc0241a25c : SUCCESS in 26m 04s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/535d82f9387d4df3b802344598c18c46 : SUCCESS in 25m 32s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4fdcf715a63e453dbc9c278e5ce9d9e2 : SUCCESS in 2h 33m 29s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/cf3a96cb5d5e4a828a8a926aa341fa7c : SUCCESS in 1h 23m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dde86bf608924709b81b045d1ff0a886 : SUCCESS in 11m 39s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6898b178d6b5434dbe7b5cc73cc4fb5f : SUCCESS in 2h 33m 55s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/9b72983a085b47d594ecccfd97e7affe : FAILURE in 1h 57m 10s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7365c72f5bf74e2285a7c6aa263cc3b6 : FAILURE in 59m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ef5f599813ba63577f84c0ae991c5fcb656a9e99","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-10 08:40:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"19f3b6b1acd7b39f7395fa6580fd27943edc140a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 11:36:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fd873f69cc34637b9333aef656eba43 : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f86cd84288f2442fb3db87fc1a98549c : SUCCESS in 9m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f6c6b4fcefd43c89979010885ab4f62 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/717aa065d6ae4a53938995cc750c07f3 : SUCCESS in 10m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a29412ef7e014d40942b89a8526b1741 : SUCCESS in 11m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e312f96eed7e42c8bfaa02953b76c889 : FAILURE in 10m 12s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/9c6ee93259014310912e56dd16e88bff : SUCCESS in 2h 49m 58s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/68446ed8f2ff433fad7a5d656aeffc3d : SUCCESS in 2h 09m 33s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/3d5e3c7ad585475999226860c6ccf192 : SUCCESS in 2h 19m 02s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d765efc15fe44571aca0b5fc9c06d946 : SUCCESS in 1h 46m 52s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/58bbf8ffa7044a3a8443a421cb09af68 : SUCCESS in 19m 29s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/b9256ef9427f4f89979071c40b2cceb1 : SUCCESS in 24m 56s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b068fe4cfcbb4009aee60193663cc22d : SUCCESS in 2h 34m 16s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/2a9cb0bf62d5452788c4af12308dbe52 : SUCCESS in 1h 18m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f27865f266b4877ad4a2cad3144d0bc : SUCCESS in 10m 44s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/58e17ed953b84c3f91791e3abba172aa : FAILURE in 2h 28m 47s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/6898f49297c1428680fe69b9674d60b5 : FAILURE in 2h 14m 59s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d1038413dbb1493a84cb7c348c9a4ff9 : FAILURE in 47m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eaadc64ed906187083422cbf31a9dac8d377153d","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-14 08:15:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"22a6caf74a0875c4c0a63da914e418f0b9d3f31f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 08:16:20.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-ovn on branch stable/train.  The error was:\n\n  expected a list for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-lower-constraints-jobs-neutron\n      check:\n        jobs:\n      gate:\n        jobs:\n          - networking-ovn-tempest-dsvm-ovs-release\n          - networking-ovn-tempest-dsvm-ovs-release-python2\n          - networking-ovn-rally-task\n          - networking-ovn-dsvm-functional-py27\n          - networking-ovn-dsvm-functional\n          - networking-ovn-tripleo-ci-centos-7-containers-multinode\n          - openstack-tox-cover:\n              required-projects:\n                - openstack/neutron\n      experimental:\n        jobs:\n          - networking-ovn-tempest-dsvm-multinode:\n              voting: false\n          - kuryr-kubernetes-tempest-ovn\n      periodic:\n        jobs:\n          - networking-ovn-tempest-dsvm-ovs-master-fedora\n\n  in \"openstack/networking-ovn/zuul.d/project.yaml@stable/train\", line 1, column 3","accounts_in_message":[],"_revision_number":4},{"id":"e18bfb02af224cc49ca2e9095b3e19eeb5b45f95","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-14 08:17:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"dd6db58f951586a87277cdb733ee3c6f249318b5","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-14 08:33:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"cb4a0e6d637963d35fe71d947c384f29405cc5f4","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-14 08:56:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f8c4d7364518f4ce8928b91e9b38c467250470b4","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-14 09:35:28.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"66b4f3de9332640e90ca84c23c0a5ecbabb0cfbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 10:15:03.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/faac47984b7a419c93d4ebdc6dc27f60 : FAILURE in 11m 00s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/72d648f3707a4236931aa482ce8c718f : SUCCESS in 29m 39s","accounts_in_message":[],"_revision_number":8},{"id":"bca96ddd8f745837fdc79de56fa354df3e0234ab","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-14 10:33:44.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"9fae01f291aeee2875d0c294c3f8cd955bcdd434","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 14:37:52.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0253a3f56640472b99d8492e009805cb : SUCCESS in 6m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/96c6e3afe1664fdfb5106f2e8252964d : SUCCESS in 11m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57ad514e6de340ef8c02125a2a082338 : SUCCESS in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc1648f9af9d475fa0f12e5f4861edd2 : SUCCESS in 13m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57359b053e824d088fc5239290302be9 : SUCCESS in 12m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de49ffc9208e45e49e647e8ce40a1e5c : SUCCESS in 12m 54s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/3ab09a48b3da474893766a26aded3d00 : SUCCESS in 2h 38m 15s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/bdf9e6ae86ea400587ddee5182733108 : SUCCESS in 2h 33m 23s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/fe233a2c7759495d999847768cf76305 : SUCCESS in 2h 36m 31s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5610dc4e21c347d7bd2387c5ad1be2fb : SUCCESS in 2h 07m 18s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/87efbfd2ccda4fad9e70815d4c4667d9 : SUCCESS in 22m 40s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/f0d38b6042974590b3ff4dac8dcb3380 : SUCCESS in 23m 00s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/74cab288feb34645ae5d4fca7c6d64ba : SUCCESS in 2h 50m 52s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/2826db34fa44485a926bec113151cb75 : SUCCESS in 1h 44m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a0248f91e544eb1b49d6e45b67af4ce : SUCCESS in 13m 08s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/82e89e748f6b4f18bfc8f83fad52b109 : FAILURE in 2h 38m 20s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/52eca80bacb440468f48d7e5f6a54512 : FAILURE in 2h 06m 40s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2142f49d7b6b48e886e3341092aa30bd : FAILURE in 51m 17s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"80dbd4f6db2adb70a5cbdbb76f9b31fecbfe27d7","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-04-14 15:10:26.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"7878725880ab752d91df0818f20539a26bfa90bd","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-14 16:52:49.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"4712c52641fae49235d722988f8dc98f3cab2857","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-15 15:18:47.000000000","message":"Patch Set 9: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"3ee2b8a63086adbd09830408ec38d8abca1a1b28","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-15 15:43:30.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"7f378158cc2fd1d7818dbb78f75a8b02ba2a3dca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 15:43:55.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"9c4255cfd75519643a5c0b902db6ae9320cc2cf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 18:29:32.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c14debad7a641698b099aad8ade72ac : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa5b2e49372844a18066a31fc8a8c801 : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc33987d3e7e4a8a8801d05818126d9b : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/608604a667f44ff5adc60c4b1fb22334 : SUCCESS in 10m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e7791e8bd25c43bea723472da7ff1e24 : SUCCESS in 9m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4b071dec86d4890b595cda5e271f6e2 : SUCCESS in 10m 52s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/0b8a36fe0ea3453eb8e1d0d684b576a5 : SUCCESS in 2h 01m 12s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/804d421efe994492898c146ca10643c4 : SUCCESS in 2h 06m 37s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4a7bbec105364d55a3cea87720241583 : FAILURE in 2h 02m 46s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/d15b54841cf745a1b6313fead40a85b0 : SUCCESS in 22m 42s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/c4ba616dd9974f6895221ca253138640 : SUCCESS in 26m 12s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9a1e2ab1aedc42ff8f6afc236cc1a155 : SUCCESS in 2h 43m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a4a97ec4f014b77a77236c3a9d43289 : SUCCESS in 10m 35s","accounts_in_message":[],"_revision_number":9},{"id":"65649ebca2b5e4d4da2c7088ccf45cfd246df22f","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-15 19:03:05.000000000","message":"Patch Set 9:\n\nrecheck rally","accounts_in_message":[],"_revision_number":9},{"id":"4a9ce3eba1ae9b8dcf5dc243813267d42cde6433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 22:12:19.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f395690fa5d4f36b13b610edfd0f61c : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9facfc6ec9e24d679ee28bafbae57457 : SUCCESS in 10m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d5cc224c3c74766ad5f50d5d8d3aa8f : SUCCESS in 9m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6dc1b1504a3849858087a6ed4c9929a8 : SUCCESS in 9m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64071952fc314e12bd07c9a6b41b0d05 : SUCCESS in 10m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1598ad65955c4b48b76643d5a2846d4e : SUCCESS in 11m 22s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/b195a1a9d6f54d6886fdde5bb4c4361c : TIMED_OUT in 3h 06m 41s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/0857ebbc57b7458485b3049af9942251 : SUCCESS in 2h 44m 12s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/998ec2a86e424b3b8657c9e96be38ed4 : SUCCESS in 2h 10m 32s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b10b8641a89e4d5bae42ae42864b18c1 : SUCCESS in 2h 09m 59s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/8628cce138fd4dd38a1d18bb04ea85e5 : SUCCESS in 21m 04s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/46b4ed10bca34a63ad083600df03f881 : SUCCESS in 19m 36s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b3bfa7b4319a4307addb4ab5ce780834 : SUCCESS in 2h 49m 07s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/a875c20684154cd79d67447b7a91da86 : SUCCESS in 1h 26m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10b623a7cb884c6b9e5a61d23c88780b : SUCCESS in 10m 34s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2befca25d7b74d3eaf916ac7b737b6f0 : FAILURE in 2h 23m 33s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/2b651f950a624161b62ee87c9b69c6c3 : FAILURE in 1h 46m 24s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1d11daa5ec684d9b807f04e193c5b14b : FAILURE in 44m 43s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"26d318e2abfba013f2f1e8161fb39a96bbb1047c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 22:12:31.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"342b6f6baa13721e37cd153e237ed80fe5911289","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 01:22:58.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98c4c8c3b42343d7831b871eedab35c4 : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d11516c87e8b43b0bf41b6ba925d69a0 : SUCCESS in 10m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24b28f2b606443539a810550d238000a : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3db1080aefe741b1805d5a3eb0617af8 : SUCCESS in 8m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e31a0a46ceed4452abc7835605c6760e : SUCCESS in 10m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32e53cc10faa47a1a39df49a63bc2044 : SUCCESS in 11m 04s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/86068851e50645adabdab9700e38868e : SUCCESS in 2h 29m 46s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/2dbc6bd9c5364c5bb997153eb0222812 : SUCCESS in 3h 02m 35s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/224e6eb17ab34cbd8c3d544ad010a2eb : SUCCESS in 1h 54m 34s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/b8ded0a2661147989bd5b76e1837baed : SUCCESS in 24m 52s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/f118c1f68a0e44cc942371687583114f : SUCCESS in 21m 09s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8aca80c6a14f461e8ce152ac64f9fa8e : SUCCESS in 3h 05m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0d01f6862b54f2bb6034860aec7d1e4 : SUCCESS in 10m 42s","accounts_in_message":[],"_revision_number":9},{"id":"fc927a38958d97166718bb01c8af4fa50d0228b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 01:22:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"eefb6cac7b3e890498012f0d1552f1187adbf40f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 01:24:36.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/368d85aba3534bf28037843406c64315 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"6429a8a26ecc76344bf4ba124be790dfdb3d65f4","revisions":{"45ec36ea0ab4b44ecb0d232bd0b2cd3736b87329":{"kind":"REWORK","_number":1,"created":"2020-04-09 11:22:33.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/69/718669/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/69/718669/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/1"}}},"commit":{"parents":[{"commit":"f2aa8afeae9af3ec9e9b028488e67e3dc393334f","subject":"[OVN] Add support for external ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f2aa8afeae9af3ec9e9b028488e67e3dc393334f"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-20 11:11:28.000000000","tz":0},"committer":{"name":"Maciej Jozefczyk","email":"mjozefcz@redhat.com","date":"2020-04-09 11:20:51.000000000","tz":0},"subject":"Don\u0027t reschedule hosts unless we need to","message":"Don\u0027t reschedule hosts unless we need to\n\nOnly reschedule gateways/update segments when things have changed\nthat would require those actions.\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\n\nChange-Id: I62f53dbd862c0f38af4a1434d453e97c18777eb4\nCloses-bug: #1861510\nCloses-bug: #1861509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/45ec36ea0ab4b44ecb0d232bd0b2cd3736b87329"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/45ec36ea0ab4b44ecb0d232bd0b2cd3736b87329"}]},"branch":"refs/heads/stable/train"},"caef1b86006da1264080af232ae0ca54fa58a3ba":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-09 11:46:01.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/69/718669/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/69/718669/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/2"}}},"commit":{"parents":[{"commit":"f2aa8afeae9af3ec9e9b028488e67e3dc393334f","subject":"[OVN] Add support for external ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f2aa8afeae9af3ec9e9b028488e67e3dc393334f"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-20 11:11:28.000000000","tz":0},"committer":{"name":"Maciej Jozefczyk","email":"mjozefcz@redhat.com","date":"2020-04-09 11:25:16.000000000","tz":0},"subject":"Don\u0027t reschedule hosts unless we need to","message":"Don\u0027t reschedule hosts unless we need to\n\nOnly reschedule gateways/update segments when things have changed\nthat would require those actions.\n\nThis code has been cherry-picked from neutron:\nhttps://review.opendev.org/#/c/705660\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\n\nConflicts:\n   networking_ovn/tests/unit/l3/test_l3_ovn.py\n   networking_ovn/tests/unit/ovsdb/test_ovsdb_monitor.py\n\nChange-Id: I62f53dbd862c0f38af4a1434d453e97c18777eb4\nCloses-bug: #1861510\nCloses-bug: #1861509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/caef1b86006da1264080af232ae0ca54fa58a3ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/caef1b86006da1264080af232ae0ca54fa58a3ba"}]},"branch":"refs/heads/stable/train"},"99c16b4ccd89ec0f17c175dd2b57e9ea9d440a38":{"kind":"REWORK","_number":3,"created":"2020-04-10 08:40:02.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/69/718669/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/69/718669/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/3"}}},"commit":{"parents":[{"commit":"f2aa8afeae9af3ec9e9b028488e67e3dc393334f","subject":"[OVN] Add support for external ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f2aa8afeae9af3ec9e9b028488e67e3dc393334f"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-20 11:11:28.000000000","tz":0},"committer":{"name":"Maciej Jozefczyk","email":"mjozefcz@redhat.com","date":"2020-04-10 08:37:02.000000000","tz":0},"subject":"Don\u0027t reschedule hosts unless we need to","message":"Don\u0027t reschedule hosts unless we need to\n\nOnly reschedule gateways/update segments when things have changed\nthat would require those actions.\n\nThis code has been cherry-picked from neutron:\nhttps://review.opendev.org/#/c/705660\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\n\nConflicts:\n   networking_ovn/tests/unit/l3/test_l3_ovn.py\n   networking_ovn/tests/unit/ovsdb/test_ovsdb_monitor.py\n\nChange-Id: I62f53dbd862c0f38af4a1434d453e97c18777eb4\nCloses-bug: #1861510\nCloses-bug: #1861509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/99c16b4ccd89ec0f17c175dd2b57e9ea9d440a38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/99c16b4ccd89ec0f17c175dd2b57e9ea9d440a38"}]},"branch":"refs/heads/stable/train"},"7b559cd38a347ee6dc15f27116efad2a8ed82c0d":{"kind":"REWORK","_number":4,"created":"2020-04-14 08:15:52.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/69/718669/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/69/718669/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/4"}}},"commit":{"parents":[{"commit":"f2aa8afeae9af3ec9e9b028488e67e3dc393334f","subject":"[OVN] Add support for external ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f2aa8afeae9af3ec9e9b028488e67e3dc393334f"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-20 11:11:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-04-14 08:15:49.000000000","tz":0},"subject":"Don\u0027t reschedule hosts unless we need to","message":"Don\u0027t reschedule hosts unless we need to\n\nOnly reschedule gateways/update segments when things have changed\nthat would require those actions.\n\nThis code has been cherry-picked from neutron:\nhttps://review.opendev.org/#/c/705660\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\n\nConflicts:\n   networking_ovn/tests/unit/l3/test_l3_ovn.py\n   networking_ovn/tests/unit/ovsdb/test_ovsdb_monitor.py\n\nChange-Id: I62f53dbd862c0f38af4a1434d453e97c18777eb4\nCloses-bug: #1861510\nCloses-bug: #1861509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7b559cd38a347ee6dc15f27116efad2a8ed82c0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7b559cd38a347ee6dc15f27116efad2a8ed82c0d"}]},"branch":"refs/heads/stable/train"},"f67aecca5c6babcd268c45ddf4e3671f87df6405":{"kind":"REWORK","_number":5,"created":"2020-04-14 08:17:23.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/69/718669/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/69/718669/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/5"}}},"commit":{"parents":[{"commit":"f2aa8afeae9af3ec9e9b028488e67e3dc393334f","subject":"[OVN] Add support for external ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f2aa8afeae9af3ec9e9b028488e67e3dc393334f"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-20 11:11:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-04-14 08:17:14.000000000","tz":0},"subject":"Don\u0027t reschedule hosts unless we need to","message":"Don\u0027t reschedule hosts unless we need to\n\nOnly reschedule gateways/update segments when things have changed\nthat would require those actions.\n\nThis code has been cherry-picked from neutron:\nhttps://review.opendev.org/#/c/705660\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\n\nConflicts:\n   networking_ovn/tests/unit/l3/test_l3_ovn.py\n   networking_ovn/tests/unit/ovsdb/test_ovsdb_monitor.py\n\nChange-Id: I62f53dbd862c0f38af4a1434d453e97c18777eb4\nCloses-bug: #1861510\nCloses-bug: #1861509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f67aecca5c6babcd268c45ddf4e3671f87df6405"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f67aecca5c6babcd268c45ddf4e3671f87df6405"}]},"branch":"refs/heads/stable/train"},"9e25de63d0d87eb766645330099c3eb3d2d6b33f":{"kind":"REWORK","_number":6,"created":"2020-04-14 08:33:24.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/69/718669/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/69/718669/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/6"}}},"commit":{"parents":[{"commit":"f2aa8afeae9af3ec9e9b028488e67e3dc393334f","subject":"[OVN] Add support for external ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f2aa8afeae9af3ec9e9b028488e67e3dc393334f"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-20 11:11:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-04-14 08:33:02.000000000","tz":0},"subject":"Don\u0027t reschedule hosts unless we need to","message":"Don\u0027t reschedule hosts unless we need to\n\nOnly reschedule gateways/update segments when things have changed\nthat would require those actions.\n\nThis code has been cherry-picked from neutron:\nhttps://review.opendev.org/#/c/705660\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\n\nConflicts:\n   networking_ovn/tests/unit/l3/test_l3_ovn.py\n   networking_ovn/tests/unit/ovsdb/test_ovsdb_monitor.py\n\nChange-Id: I62f53dbd862c0f38af4a1434d453e97c18777eb4\nCloses-bug: #1861510\nCloses-bug: #1861509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/9e25de63d0d87eb766645330099c3eb3d2d6b33f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/9e25de63d0d87eb766645330099c3eb3d2d6b33f"}]},"branch":"refs/heads/stable/train"},"45c4843f15a363089d0f1685f7cf8e22891b9f52":{"kind":"REWORK","_number":7,"created":"2020-04-14 08:56:29.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/69/718669/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/69/718669/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/7"}}},"commit":{"parents":[{"commit":"f2aa8afeae9af3ec9e9b028488e67e3dc393334f","subject":"[OVN] Add support for external ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f2aa8afeae9af3ec9e9b028488e67e3dc393334f"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-20 11:11:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-04-14 08:56:10.000000000","tz":0},"subject":"Don\u0027t reschedule hosts unless we need to","message":"Don\u0027t reschedule hosts unless we need to\n\nOnly reschedule gateways/update segments when things have changed\nthat would require those actions.\n\nThis code has been cherry-picked from neutron:\nhttps://review.opendev.org/#/c/705660\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\n\nConflicts:\n   networking_ovn/tests/unit/l3/test_l3_ovn.py\n   networking_ovn/tests/unit/ovsdb/test_ovsdb_monitor.py\n\nChange-Id: I62f53dbd862c0f38af4a1434d453e97c18777eb4\nCloses-bug: #1861510\nCloses-bug: #1861509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/45c4843f15a363089d0f1685f7cf8e22891b9f52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/45c4843f15a363089d0f1685f7cf8e22891b9f52"}]},"branch":"refs/heads/stable/train"},"9cb322f80a0607ab71007cb103c64cda084f6088":{"kind":"REWORK","_number":8,"created":"2020-04-14 09:35:28.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/69/718669/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/69/718669/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/8"}}},"commit":{"parents":[{"commit":"f2aa8afeae9af3ec9e9b028488e67e3dc393334f","subject":"[OVN] Add support for external ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f2aa8afeae9af3ec9e9b028488e67e3dc393334f"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-20 11:11:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-04-14 09:35:12.000000000","tz":0},"subject":"[WIP] Don\u0027t reschedule hosts unless we need to","message":"[WIP] Don\u0027t reschedule hosts unless we need to\n\nOnly reschedule gateways/update segments when things have changed\nthat would require those actions.\n\nThis code has been cherry-picked from neutron:\nhttps://review.opendev.org/#/c/705660\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\n\nConflicts:\n   networking_ovn/tests/unit/l3/test_l3_ovn.py\n   networking_ovn/tests/unit/ovsdb/test_ovsdb_monitor.py\n\nChange-Id: I62f53dbd862c0f38af4a1434d453e97c18777eb4\nCloses-bug: #1861510\nCloses-bug: #1861509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/9cb322f80a0607ab71007cb103c64cda084f6088"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/9cb322f80a0607ab71007cb103c64cda084f6088"}]},"branch":"refs/heads/stable/train"},"6429a8a26ecc76344bf4ba124be790dfdb3d65f4":{"kind":"REWORK","_number":9,"created":"2020-04-14 10:33:44.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/69/718669/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/69/718669/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/69/718669/9"}}},"commit":{"parents":[{"commit":"f2aa8afeae9af3ec9e9b028488e67e3dc393334f","subject":"[OVN] Add support for external ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f2aa8afeae9af3ec9e9b028488e67e3dc393334f"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-20 11:11:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-04-14 10:31:44.000000000","tz":0},"subject":"Don\u0027t reschedule hosts unless we need to","message":"Don\u0027t reschedule hosts unless we need to\n\nOnly reschedule gateways/update segments when things have changed\nthat would require those actions.\n\nThis code has been cherry-picked from neutron:\nhttps://review.opendev.org/#/c/705660\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\n\nConflicts:\n   networking_ovn/tests/unit/l3/test_l3_ovn.py\n   networking_ovn/tests/unit/ovsdb/test_ovsdb_monitor.py\n\nChange-Id: I62f53dbd862c0f38af4a1434d453e97c18777eb4\nCloses-bug: #1861510\nCloses-bug: #1861509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6429a8a26ecc76344bf4ba124be790dfdb3d65f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6429a8a26ecc76344bf4ba124be790dfdb3d65f4"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
