)]}'
{"id":"openstack%2Fnetworking-ovn~718693","triplet_id":"openstack%2Fnetworking-ovn~stable%2Fqueens~I62f53dbd862c0f38af4a1434d453e97c18777eb4","project":"openstack/networking-ovn","branch":"stable/queens","topic":"bug/1861510","hashtags":[],"change_id":"I62f53dbd862c0f38af4a1434d453e97c18777eb4","subject":"Don\u0027t reschedule hosts unless we need to","status":"MERGED","created":"2020-04-09 12:51:44.000000000","updated":"2020-04-17 13:21:37.000000000","submitted":"2020-04-17 13:21:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718693-1587129697832-ab29ac72","meta_rev_id":"ef9990b1f651b84d6ae4d458f3d01af1c2e596e8","_number":718693,"virtual_id_number":718693,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2020-04-15 12:12:18.000000000","_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-17 13:21:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2020-04-17 10:32:09.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2020-04-15 15:25:14.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":1,"date":"2020-04-17 10:32:09.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-15 15:25:14.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-04-17 10:32:09.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-04-17 13:21:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"41e591a0dbf24bf64e4f083e66454890209e641e","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-09 12:51:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03bcc84e9adc6bf0ef01f0a1c9d3f96f4decf0a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 15:57:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9f2703e33864eef8541125066d7d84d : SUCCESS in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2b9e09e4f424608b9102d42384bc897 : SUCCESS in 9m 33s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/fb678a09318f402ebcdeb9ffbfdd694c : SUCCESS in 8m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4eea8eb7855e4d56a7cc7cdae7814528 : FAILURE in 12m 34s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/2f400ade139141bb8b640d7c449d0696 : SUCCESS in 1h 05m 38s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5b8d72a86aaf4c618880c293b7fecdf2 : FAILURE in 35m 30s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/e239009fc6b141db942632d2671a3064 : SUCCESS in 23m 54s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/ac6c5074d3304981980b5e818a8ec7d1 : SUCCESS in 21m 22s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/7703e6a2f1ff496187898b8615ab6e2f : FAILURE in 33m 30s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/1c2f1b8dd69147ff995b98c99c231697 : FAILURE in 2h 55m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d539e3cc73704f1abb0b568c1d9d5788 : FAILURE in 13m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea2aa616571188fde549766c139530503007f5ae","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-10 08:50:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7527013eb69cbc87712093c1af79fac9a9b03e90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 11:37:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eede3612f3e47519ff47540f595628b : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4887359c1d0c42ff99685cbfc36bda1a : FAILURE in 9m 42s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0da7da5031bb41418c6b9709e123babb : SUCCESS in 5m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9f08b29ea1114970bf40e425b4ca804b : FAILURE in 10m 39s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c36129246254445fb9fd4019d61bc39f : SUCCESS in 1h 04m 55s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f4f8a14ad65a4a1ba742b7a39724439d : FAILURE in 34m 24s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/ff1319b231474d309f5d537e6cb8e832 : SUCCESS in 19m 26s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/b29950e3899b4c54a16b3ed75c4cccc8 : SUCCESS in 19m 43s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/d5609e432f6f4f6e86beb63530e40ec6 : FAILURE in 38m 13s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/8cb8bcd27ab041c58c925fbb45fb889b : FAILURE in 2h 44m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a117e41f5361434786887c8bd8678777 : FAILURE in 13m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ded9d9035672e63044a6ff620e1ccb29aadad611","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-14 10:57:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"74a6a13c3e83504207de81aac011a8d9796a54f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 16:58:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e2efbb95b194510ad3df0226f315b80 : SUCCESS in 9m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b891172e54541bfaec8f84cc19c199b : SUCCESS in 11m 46s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0f2369fd5e29418fbeef19b5ab234262 : SUCCESS in 7m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4d247757d5a842e9999356d11434d08e : SUCCESS in 13m 43s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/d1df39ec68be4414a9f25016af3d6e2b : SUCCESS in 1h 00m 10s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/221a1aa5198f47699b8881a79082642d : FAILURE in 32m 03s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/ff075a87b3854beea9ba62f973f2056b : SUCCESS in 23m 39s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/166fae02955f46bd89823c25a09cc94f : SUCCESS in 24m 00s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/f19e7bd3f9f441199cc3c78fb517f91a : FAILURE in 34m 06s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/258202f9a94c49ee87b7d5ba8f377a0a : FAILURE in 3h 06m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97241b6a2e04427aac781710dd69b464 : SUCCESS in 12m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d7ca944132d6c0c8da45e0e76dd5bfb803aad08d","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-14 17:19:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ef7d846b66a6fea420e0c021f96c37c3d10916c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 20:40:35.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c4e915387944ec7857361eabe8acea2 : SUCCESS in 6m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3bc5b6c1ecb14ce19414f82853cfc13e : SUCCESS in 10m 42s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/39979466cf374bb4a21adfb0e080b2de : SUCCESS in 5m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e168dd0529b846e3a91c0982239370a6 : SUCCESS in 11m 59s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/5a726ec8a6a14924abdf616355adb04b : SUCCESS in 1h 19m 49s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5a10844943cc45199214a9c9df795e7f : FAILURE in 30m 39s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/3f1b0a0e1bf34cc7990962d8eee3e526 : SUCCESS in 18m 50s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/ef2bfe9ddfee460d997a7f34d3c08429 : SUCCESS in 25m 04s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/e566b5c3e8f849a794b776a6c1585305 : FAILURE in 36m 15s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a0e85781a95840fcbb232d55f97790ae : TIMED_OUT in 3h 11m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34fd43ae478641b7a4f25def8f7bbc99 : SUCCESS in 13m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"53427d69edfb9629b90761626340180b8d77fd88","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-15 12:12:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7f913b116ef1a1edf0f5c0ae23898189531e6648","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-15 15:25:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a8b325a42d4e845a498700ff2f89a976d15d07d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 15:29:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d97a2dfef91469eb8b98a12c11e5d83 : SUCCESS in 8m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ee6e36aaf7aa443198ba151f568fb09d : SUCCESS in 11m 09s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/382c5afff1ef4e57902a97fb817ed6fb : SUCCESS in 7m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e8190452010f4f3b8d1729e751a3fd25 : SUCCESS in 12m 25s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/315d7854bff544c69c6dc3b41d6dce0c : SUCCESS in 1h 03m 06s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1c8f9e7f3ef84a0c94b84496723e7c3b : SUCCESS in 1h 34m 08s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/89827ee0b5f8477da2c1433a9738150e : SUCCESS in 18m 31s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/cd04939a48ff4a4398423b9a10f973a0 : SUCCESS in 26m 13s\n- networking-ovn-tempest-dsvm-neutron-api-scenario-ovs-release https://zuul.opendev.org/t/openstack/build/015ec3744cce405d8023088224e1ccb8 : FAILURE in 35m 12s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9b1f7b79e7734f1eab6136704439af3b : FAILURE in 3h 03m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/999b951b73cd484f9d3ddb40737919bf : SUCCESS in 13m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0f64cf87d73dad2d8f3e18e67905ad6f72d48681","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-04-17 10:32:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThanks Maciej!","accounts_in_message":[],"_revision_number":3},{"id":"bf645aec5481799d7560d56e550ea6b3ac747ca5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 10:32:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1a09c7594572e5831853bbc9902c3d7813f1a27a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 13:21:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"123c80ffd23e996d27cf00b5d9ef93ddd2433690","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 13:21:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52fa0976149843dd8af5fd971417bb69 : SUCCESS in 8m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/79aedd36352f41a99e74ba74395a5472 : SUCCESS in 10m 24s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c4b747e214264917828248cf5440c0ec : SUCCESS in 6m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fc6e3845a24f43189871c1bfebe20d01 : SUCCESS in 15m 38s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/966cb44ee1ac4585a84c234b62db0cf3 : SUCCESS in 1h 04m 49s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/328166d3beaf488781cd45764741d6af : SUCCESS in 1h 40m 29s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/26b54552c3924773815b9d69d0ff8d95 : SUCCESS in 22m 54s\n- networking-ovn-dsvm-functional-py35 https://zuul.opendev.org/t/openstack/build/e2cc1e8031c34d6f8f18e6ebfdd1fbc0 : SUCCESS in 19m 55s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ceaeb4f874224d4f897ec10b37dc440d : FAILURE in 2h 43m 16s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c71c95a7a3b1f8f2c8b9cf3b048d8dddde8bf891","revisions":{"ab17e196c2f22c07bcd0d97c42340bd85b44670e":{"kind":"REWORK","_number":1,"created":"2020-04-09 12:51:44.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/93/718693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/93/718693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/718693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/718693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/718693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/93/718693/1"}}},"commit":{"parents":[{"commit":"443d681c52a702bfa49ec539ab2714517e5fe800","subject":"Merge \"Propagate mtu to all subnets\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/443d681c52a702bfa49ec539ab2714517e5fe800"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-20 11:11:28.000000000","tz":0},"committer":{"name":"Maciej Jozefczyk","email":"mjozefcz@redhat.com","date":"2020-04-09 12:51:37.000000000","tz":0},"subject":"Don\u0027t reschedule hosts unless we need to","message":"Don\u0027t reschedule hosts unless we need to\n\nOnly reschedule gateways/update segments when things have changed\nthat would require those actions.\n\nThis code has been cherry-picked from neutron:\nhttps://review.opendev.org/#/c/705660\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\n\nConflicts:\n      networking_ovn/common/utils.py\n      networking_ovn/ovsdb/ovsdb_monitor.py\n      networking_ovn/tests/unit/l3/test_l3_ovn.py\n      networking_ovn/tests/unit/ovsdb/test_ovsdb_monitor.py\n\nChange-Id: I62f53dbd862c0f38af4a1434d453e97c18777eb4\nCloses-bug: #1861510\nCloses-bug: #1861509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ab17e196c2f22c07bcd0d97c42340bd85b44670e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ab17e196c2f22c07bcd0d97c42340bd85b44670e"}]},"branch":"refs/heads/stable/queens"},"ed2a5c94754973a0372b1b3010128a67b1c1a0f0":{"kind":"REWORK","_number":2,"created":"2020-04-10 08:50:44.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/93/718693/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/93/718693/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/718693/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/718693/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/718693/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/93/718693/2"}}},"commit":{"parents":[{"commit":"443d681c52a702bfa49ec539ab2714517e5fe800","subject":"Merge \"Propagate mtu to all subnets\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/443d681c52a702bfa49ec539ab2714517e5fe800"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-20 11:11:28.000000000","tz":0},"committer":{"name":"Maciej Jozefczyk","email":"mjozefcz@redhat.com","date":"2020-04-10 08:44:38.000000000","tz":0},"subject":"Don\u0027t reschedule hosts unless we need to","message":"Don\u0027t reschedule hosts unless we need to\n\nOnly reschedule gateways/update segments when things have changed\nthat would require those actions.\n\nThis code has been cherry-picked from neutron:\nhttps://review.opendev.org/#/c/705660\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\n\nConflicts:\n      networking_ovn/common/utils.py\n      networking_ovn/ovsdb/ovsdb_monitor.py\n      networking_ovn/tests/unit/l3/test_l3_ovn.py\n      networking_ovn/tests/unit/ovsdb/test_ovsdb_monitor.py\n\nChange-Id: I62f53dbd862c0f38af4a1434d453e97c18777eb4\nCloses-bug: #1861510\nCloses-bug: #1861509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ed2a5c94754973a0372b1b3010128a67b1c1a0f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ed2a5c94754973a0372b1b3010128a67b1c1a0f0"}]},"branch":"refs/heads/stable/queens"},"c71c95a7a3b1f8f2c8b9cf3b048d8dddde8bf891":{"kind":"REWORK","_number":3,"created":"2020-04-14 10:57:29.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/93/718693/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/93/718693/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/718693/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/718693/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/718693/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/93/718693/3"}}},"commit":{"parents":[{"commit":"443d681c52a702bfa49ec539ab2714517e5fe800","subject":"Merge \"Propagate mtu to all subnets\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/443d681c52a702bfa49ec539ab2714517e5fe800"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-20 11:11:28.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-04-14 10:57:20.000000000","tz":0},"subject":"Don\u0027t reschedule hosts unless we need to","message":"Don\u0027t reschedule hosts unless we need to\n\nOnly reschedule gateways/update segments when things have changed\nthat would require those actions.\n\nThis code has been cherry-picked from neutron:\nhttps://review.opendev.org/#/c/705660\n\nCo-Authored-By: Terry Wilson \u003ctwilson@redhat.com\u003e\n\nConflicts:\n      networking_ovn/common/utils.py\n      networking_ovn/ovsdb/ovsdb_monitor.py\n      networking_ovn/tests/unit/l3/test_l3_ovn.py\n      networking_ovn/tests/unit/ovsdb/test_ovsdb_monitor.py\n\nChange-Id: I62f53dbd862c0f38af4a1434d453e97c18777eb4\nCloses-bug: #1861510\nCloses-bug: #1861509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c71c95a7a3b1f8f2c8b9cf3b048d8dddde8bf891"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c71c95a7a3b1f8f2c8b9cf3b048d8dddde8bf891"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
