)]}'
{"id":"openstack%2Fnetworking-ovn~727985","triplet_id":"openstack%2Fnetworking-ovn~stable%2Frocky~Id83c2899a03af98e32b0be7ba6ad661e646f07bc","project":"openstack/networking-ovn","branch":"stable/rocky","hashtags":[],"change_id":"Id83c2899a03af98e32b0be7ba6ad661e646f07bc","subject":"[OVN][metadata] Adding ERROR trace upon unexpected data","status":"MERGED","created":"2020-05-14 09:05:39.000000000","updated":"2020-05-25 17:15:23.000000000","submitted":"2020-05-25 17:15:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727985-1590426923703-fdce670a","meta_rev_id":"e90871933f3933631c5ff0eed3423cb4bb509ff9","_number":727985,"virtual_id_number":727985,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-25 17:15:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2020-05-25 14:28:30.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-05-19 17:49:57.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":1,"date":"2020-05-25 14:28:30.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"date":"2020-05-20 10:48:17.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-14 09:05:39.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-05-19 17:49:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-05-20 10:48:17.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-05-25 14:28:30.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-05-25 17:15:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6f47efa3274604af6e1ad9f52758ae0546909b34","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-05-14 09:05:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"385bb89aa10d7a0923f3a843aa113d7b9653664a","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-05-14 09:27:59.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fc5662f7676cc2763e565396f056dbdc62364047","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 11:56:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/849e7493e0334a0ab484f1f7f0c93d8e : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d12a532b61344cb58802de90ea4a3014 : SUCCESS in 9m 57s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e5203f10a82546e5bd36759cfc252fb5 : SUCCESS in 5m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b0ccc327a47541fdb2113ca5089744f4 : SUCCESS in 12m 54s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/0917d3b9ca39475bb900d00df5ad7bcc : SUCCESS in 1h 38m 42s\n- networking-ovn-tempest-dsvm-ovs-release-python3 https://zuul.opendev.org/t/openstack/build/66f1813415914c819c9d201b8918bc6e : SUCCESS in 1h 06m 37s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/07d9b51442f049e18a0491ee210494f2 : SUCCESS in 1h 32m 04s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/6caca5e875d34baa973877eebab3edec : SUCCESS in 21m 42s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/d2ee81cebabc4a17b605c2d9c5a67acc : SUCCESS in 21m 40s\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/7d3d37a29e8b433f844dab54ecb5bb44 : SUCCESS in 2h 24m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16119e310b0d4f71ae0e6633197d2af6 : SUCCESS in 12m 11s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/e3edb54e46074129bfc85649002e64e3 : SUCCESS in 1h 13m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56a2264f03eb48c884a1c0b714e772d7 : SUCCESS in 11m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4a0207e1127481cf1c68103864c55d521de2affd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-19 17:49:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1d58a50336e514efae0a53c97449c870f2d915a8","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-05-25 14:28:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a535ac94fb8a67888f293979766499acaa46ed20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 14:28:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5ae509efe0dae70cc92b38045d8e34e5baa699fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 17:15:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fac2a275cdd24c47a145c84da78964e1 : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6363c49f31c14e819791c317dac278ea : SUCCESS in 9m 15s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e3a2d74e79984f3da27aea527048cff2 : SUCCESS in 4m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a77e944ce9e347bda81d0e2052bbd822 : SUCCESS in 9m 23s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/6cd5febc301f4719a798a0b3270c409a : SUCCESS in 1h 45m 31s\n- networking-ovn-tempest-dsvm-ovs-release-python3 https://zuul.opendev.org/t/openstack/build/b3795e345bea46d0b297389cb1b320b0 : SUCCESS in 1h 00m 43s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/71495b8f08fb42d590031d5921539e1c : SUCCESS in 1h 31m 10s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/2244edd48c684890be4a3969159051f4 : SUCCESS in 18m 42s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/19d2f208c7d9449e9bec9946667c1c23 : SUCCESS in 22m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8f4da2db4a043e49efb1ec31cd287bc : SUCCESS in 10m 10s","accounts_in_message":[],"_revision_number":2},{"id":"e90871933f3933631c5ff0eed3423cb4bb509ff9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 17:15:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"48e104a4cd176e480b899d3f43dc997b43ad6b48","revisions":{"3469afdfde160c15162dbf578f2f8627f66780be":{"kind":"REWORK","_number":1,"created":"2020-05-14 09:05:39.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/85/727985/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/85/727985/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/85/727985/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/85/727985/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/85/727985/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/85/727985/1"}}},"commit":{"parents":[{"commit":"80a186e2c2200da0f8c6ba3e7e84a417cae842b8","subject":"Merge \"mech_driver.OVNMechanismDriver \"_ovn_client\" is a read-only property\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/80a186e2c2200da0f8c6ba3e7e84a417cae842b8"}]}],"author":{"name":"Daniel Alvarez","email":"dalvarez@redhat.com","date":"2020-04-24 13:33:11.000000000","tz":120},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-05-14 09:07:54.000000000","tz":120},"subject":"[OVN][metadata] Adding ERROR trace upon unexpected data","message":"[OVN][metadata] Adding ERROR trace upon unexpected data\n\nWhen a metadata request arrives to the metadata agent, OVN\nwill need to figure out which port it corresponds to and it\u0027ll do\nbased on the network ID and IP address which should be a unique pair.\n\nIf it\u0027s not unique or it doesn\u0027t exist, then there is an error\nand an ERROR trace is logged.\n\nChange-Id: Id83c2899a03af98e32b0be7ba6ad661e646f07bc\nRelated-Bug: 1874733\nSigned-off-by: Daniel Alvarez \u003cdalvarez@redhat.com\u003e\n(cherry picked from commit 591adfee971906dc86e9357903f164d01657ed65)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/3469afdfde160c15162dbf578f2f8627f66780be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/3469afdfde160c15162dbf578f2f8627f66780be"}]},"branch":"refs/heads/stable/rocky"},"48e104a4cd176e480b899d3f43dc997b43ad6b48":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-14 09:27:59.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/85/727985/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/85/727985/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/85/727985/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/85/727985/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/85/727985/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/85/727985/2"}}},"commit":{"parents":[{"commit":"80a186e2c2200da0f8c6ba3e7e84a417cae842b8","subject":"Merge \"mech_driver.OVNMechanismDriver \"_ovn_client\" is a read-only property\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/80a186e2c2200da0f8c6ba3e7e84a417cae842b8"}]}],"author":{"name":"Daniel Alvarez","email":"dalvarez@redhat.com","date":"2020-04-24 13:33:11.000000000","tz":120},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-05-14 09:27:54.000000000","tz":0},"subject":"[OVN][metadata] Adding ERROR trace upon unexpected data","message":"[OVN][metadata] Adding ERROR trace upon unexpected data\n\nWhen a metadata request arrives to the metadata agent, OVN\nwill need to figure out which port it corresponds to and it\u0027ll do\nbased on the network ID and IP address which should be a unique pair.\n\nIf it\u0027s not unique or it doesn\u0027t exist, then there is an error\nand an ERROR trace is logged.\n\nChange-Id: Id83c2899a03af98e32b0be7ba6ad661e646f07bc\nRelated-Bug: 1874733\nSigned-off-by: Daniel Alvarez \u003cdalvarez@redhat.com\u003e\n(cherry picked from Neutron commit 591adfee971906dc86e9357903f164d01657ed65)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/48e104a4cd176e480b899d3f43dc997b43ad6b48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/48e104a4cd176e480b899d3f43dc997b43ad6b48"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
