)]}'
{"id":"openstack%2Fnetworking-ovn~728817","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~I2db34290c07c95cf2c5623886da5c2d9a0f9769f","project":"openstack/networking-ovn","branch":"stable/train","hashtags":[],"change_id":"I2db34290c07c95cf2c5623886da5c2d9a0f9769f","subject":"Revert \"Use tempest version 23.0.0 for stable/train python2 job\"","status":"MERGED","created":"2020-05-18 12:03:36.000000000","updated":"2020-05-19 15:46:50.000000000","submitted":"2020-05-19 15:41:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728817-1589902912414-df490097","meta_rev_id":"a7b589d68dbc5d5d44cc6619888421ad3d26717e","_number":728817,"virtual_id_number":728817,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-19 15:41:52.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-05-18 12:27:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-05-18 15:14:30.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-05-18 12:38:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2020-05-18 15:09:43.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-05-18 15:14:30.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"date":"2020-05-18 12:03:36.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"date":"2020-05-18 12:03:36.000000000","_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-18 12:03:36.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-05-18 12:03:36.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-05-18 12:27:10.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-05-18 12:38:48.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-05-18 15:09:43.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-05-18 15:14:30.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-05-19 15:41:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2be632fe0dcd0d66c6d8d2a1e64ad939eaa988f6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-18 12:03:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af29b014b4c784e9936151b1af3946dc83770cc9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-18 12:27:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"29ebc17d406f16eadf495bad5fb26ee6b4a80996","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-05-18 12:38:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3433f885fe6265fb67a9aabd4bab7e2c39a8101e","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-05-18 12:39:18.000000000","message":"Patch Set 1:\n\nThere shouldn\u0027t be any installation issue because of:\n\nhttps://review.opendev.org/#/c/703011/","accounts_in_message":[],"_revision_number":1},{"id":"ad4426444a84ab41c44993742735e818db603f1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 14:48:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e433af77f33647da995c9292bed18b7a : SUCCESS in 5m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6fc974116099404c831fe8bb9c77039d : SUCCESS in 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ec6318427504013978177135e16830a : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04383ff68a544b548e36fb892e231307 : SUCCESS in 12m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0da758a83213436f86c12ce0ed67d4f0 : SUCCESS in 9m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac334316e82e4296bff885d7a0b066ab : SUCCESS in 10m 21s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/74bd34180ad5451d83f51303249dc679 : SUCCESS in 1h 34m 19s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/a296e2e1020c4c96b849b591ee5ac90e : SUCCESS in 1h 19m 51s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/d30ca709097a44ffbb40760d8f7b3503 : SUCCESS in 1h 33m 45s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/549a7e7ee6b949b7996bf42fcee77135 : SUCCESS in 2h 12m 14s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/5d61affbf091465787632f148a5c8879 : SUCCESS in 20m 30s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/7c2491dfdb10484cbd6d48088e0c1bad : SUCCESS in 23m 42s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/412788e37cfc4225875259ddd11e7553 : SUCCESS in 2h 40m 04s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/43f3b6a7dd8e4d13ab59b20ced65f732 : SUCCESS in 1h 35m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75607c640f5443c98c06c84a2cc76fae : SUCCESS in 11m 50s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8b348c185e974e458d7e5873edd08947 : SUCCESS in 2h 14m 57s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/04cadfff90fc4f51904e00cf469559a4 : SUCCESS in 1h 17m 05s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0661dd4bb9884fe9bcabf40b3a682ccf : FAILURE in 44m 35s (non-voting)\n\nWarning:\n  Comments left for invalid file networking_ovn/tests/unit/ml2/test_mech_driver.py","accounts_in_message":[],"_revision_number":1},{"id":"968f20280d7c37efd86fa4acafab32a7f789c16d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-18 15:09:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"28057c513837fc5c4c6765d57c05d53fd8b43747","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-05-18 15:14:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nAwesome! Thanks","accounts_in_message":[],"_revision_number":1},{"id":"f9a4e006efb022c14d1613c10032549cf760699b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 15:15:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81661d7069621d644124729c547b5a40aeeecc4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 17:29:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fd3c15ed69248e7835eb8fa801e1565 : SUCCESS in 7m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/32e1d885fb034460aeaf6d0000b6e07b : SUCCESS in 14m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c5295dd03204504a2585553df80e092 : SUCCESS in 10m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e72f1ed9d9646d9b5ca68fcb9ad9666 : SUCCESS in 12m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5fda564eac544e289b6e5476d9aad16f : SUCCESS in 10m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0279054fcb74752a81ee0933273188e : SUCCESS in 12m 03s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c97c45a031b049fc8d5fed08aa04481d : SUCCESS in 1h 27m 23s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/71f11285a6dd431796bf0353c6d8ddbc : SUCCESS in 1h 42m 14s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/218900bb540c43b3a3825f536fe79a6c : SUCCESS in 2h 11m 13s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/c326275cc9e6470e850d5a48d662d756 : SUCCESS in 25m 02s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/87ddee721e8d40af8db76c94ebd2eec1 : SUCCESS in 21m 18s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/155d555b63c64e228482a4fadbb97fc1 : FAILURE in 2h 06m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4de50c3064b248e6a44000560378a17f : SUCCESS in 15m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c9d8f16f3c6d0d183ca71df7a7d63e886b1c6af5","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-05-18 19:23:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8c2080ac2f890846b3957876cf0a8af993ad8f67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 22:14:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46324f6c42724109bfb822a311a439a9 : SUCCESS in 5m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/23ff5b4c4e5b4b4698a5232b7e0edfaf : SUCCESS in 11m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31aae6f6917f460e953e11b56edf7c41 : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c47bd6e3cf147a7870aa8837e8e3c40 : SUCCESS in 11m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e739d6363534d7b9915fb94c19d581d : SUCCESS in 12m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13d01b1cd8c44cce99b345a226c3971c : SUCCESS in 11m 09s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/a7befc11f6fd449c8a5f37236a2f1c4e : SUCCESS in 1h 35m 15s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/7cf4c12ee5ee4b78a3d037b5007c7abd : SUCCESS in 1h 37m 17s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/a589afb996bc44faa04ce2858c74df68 : SUCCESS in 1h 28m 26s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c54e2adaafd54ecfa346e286828f61c6 : SUCCESS in 2h 11m 42s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/7c4508ba46744a07907defcceac7ad85 : FAILURE in 28m 12s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/0e9a64400dac425f8819bbce762ed014 : SUCCESS in 24m 02s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c986aab92875496294403a82f95e6636 : SUCCESS in 2h 49m 09s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/26d2a32bd896402397e283a0bed4ad8e : SUCCESS in 1h 23m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae9c45ddf40647298ffe9178e7103b4f : SUCCESS in 10m 33s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/82f1ff17e9d74155bdb897f8ccd748cd : SUCCESS in 2h 06m 12s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/25ee0edc5d774b0792bcd43dd5be7f97 : SUCCESS in 1h 32m 43s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9307edc6bc55498cb10fe5c41c3e119f : FAILURE in 52m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a32d29552b38c98af9ad9a78dd9f6b56aefd03d","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-05-19 07:16:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f57d694d81de51c819223f6e838e0951df336d91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 09:55:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14161458e2fa48c8ad85782dfb92e10d : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e523640fd67b4adfbfc8124542734d01 : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/233243c10a0147d1a22dacaebba7b8ed : SUCCESS in 10m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8636a3a1be2c460fa4b5da6defa8c306 : SUCCESS in 10m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59bc89b3e841406f98203deeed2033bd : SUCCESS in 12m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e01a562e2c7d44418ff7379578a65256 : SUCCESS in 17m 14s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/0dd3f56369ec42d5bdfd20baf7ee75b9 : SUCCESS in 1h 25m 49s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/5eda246263a945e988da0e9960bdbcaf : SUCCESS in 1h 33m 05s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/8956d6f733d04711bdbf1a81551dcce2 : SUCCESS in 1h 18m 40s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aaf47198063444fc8a41abee69cb89f4 : SUCCESS in 1h 49m 34s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/4b1bae8627f04138bca65003f4b115c3 : SUCCESS in 23m 31s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/2df75736b57c4561b567246feb99ebec : FAILURE in 33m 18s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/59135150b64e4d04b1f33761e50874a6 : SUCCESS in 2h 27m 48s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/7c944a1682f64186a5a1e4f1412531bd : SUCCESS in 1h 24m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3041642776e4f0a86a12b0180c6336b : SUCCESS in 11m 46s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/bcd60481884f4824b8fb1079ac7f7296 : SUCCESS in 2h 17m 03s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/815902363efc40a1bdd23c0422845b98 : SUCCESS in 1h 36m 56s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ad0462621974397a3ec83a3aadccaeb : FAILURE in 49m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d39ec2710366831f289da4677df9a05e930c576a","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-05-19 10:06:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"40bd061f95cd55e7237739f865b149c731a494ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 12:48:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/840ca67debca407082ae1a0b3e5b59b8 : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8013c0bc9402452ca67f1d56508b2fb5 : SUCCESS in 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39d88be9e8ec42b9b4c3186fc7ca1c3d : SUCCESS in 9m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd8c8968ea7d4fbbb0b06e837a5e9bbc : SUCCESS in 10m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/afd5193b398b4008b4272c8af27e7772 : SUCCESS in 13m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c3926c01de44701af74698e1eb48d6d : SUCCESS in 12m 42s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/4b62fd2a5eba4213bc33895bb06b8f86 : SUCCESS in 1h 37m 20s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/fea490e01a894df4826219cf47807fd2 : SUCCESS in 1h 25m 50s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/748d1a27d0c24b7983fcfe34d4af2878 : SUCCESS in 1h 20m 29s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8ddb6c77f1cc4ea2a74b5ca791e0b80c : SUCCESS in 2h 02m 00s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/5a3418434b85422cab9d7ecc36908eba : SUCCESS in 22m 09s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/12b938ce9ebd498c9606a79fe6287e29 : SUCCESS in 23m 10s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/92b354685c964a5fbad7d75df29fcd42 : SUCCESS in 2h 37m 25s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/b9f1de4313b14aac904336a8c260d17c : SUCCESS in 1h 28m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b18d384c6784597853cf5f5d998bfe0 : SUCCESS in 10m 34s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5324e1c7f8c647798493ece54187b0d0 : SUCCESS in 2h 29m 52s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/61bb53b2bd0a4370b7ee26c4febb95cc : SUCCESS in 1h 46m 17s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f995d0d14ac3407dad6e1d5c26278f07 : FAILURE in 43m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"156caf68249af2f1bc3e53f984d0ace65bfcec69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 12:49:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f5984fb5f2a096f0b2e98288587e180d77924018","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 15:41:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6d54e75340345f8bfd07c080948ab26 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de27539035454f9e943f6e1ee876289a : SUCCESS in 10m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89b981e24a004642b89f5d4aa07eaf40 : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a63ec15d60b4c18b8ec7800e9223a28 : SUCCESS in 9m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e24f0203eea49259066158cd2c8ca08 : SUCCESS in 10m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e000d87cf3904799bfa75fc27cc03dcd : SUCCESS in 10m 58s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/0d921a4951b14c6d98d610042daab40c : SUCCESS in 1h 37m 23s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/8aeba34c9b5845e188f889ddc2350412 : SUCCESS in 1h 42m 20s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e3c6ae6e84ea4de7af586d750d3bf86d : SUCCESS in 2h 01m 29s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/b26a059fa51b427f95d5aa534681628d : SUCCESS in 26m 13s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/9d9b849803ac4ec0a42f81829b990383 : SUCCESS in 19m 30s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/60d03087bb66427e8c971760ac6cf83e : SUCCESS in 2h 47m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e299843217e4988a74f4a984fb5c2de : SUCCESS in 10m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5eacce5ac9d265bc1fbd8866607f8da710cd9e81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 15:41:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a7b589d68dbc5d5d44cc6619888421ad3d26717e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 15:46:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c2f31a17f1448e097234c41fbc7399b : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af5a648f1b9dd377cfa49ec6230af64891e0d7e3","revisions":{"af5a648f1b9dd377cfa49ec6230af64891e0d7e3":{"kind":"REWORK","_number":1,"created":"2020-05-18 12:03:36.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/17/728817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/17/728817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/17/728817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/17/728817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/17/728817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/17/728817/1"}}},"commit":{"parents":[{"commit":"c0772d3ac693c64ec5bde421bdb0663f5ed787c1","subject":"Use tempest version 23.0.0 for stable/train python2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c0772d3ac693c64ec5bde421bdb0663f5ed787c1"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-05-18 12:03:36.000000000","tz":0},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-05-18 12:03:36.000000000","tz":0},"subject":"Revert \"Use tempest version 23.0.0 for stable/train python2 job\"","message":"Revert \"Use tempest version 23.0.0 for stable/train python2 job\"\n\ntempest runs in its own venv so it does not need to be pinned to last py2 version\n\nThis reverts commit c0772d3ac693c64ec5bde421bdb0663f5ed787c1.\n\nChange-Id: I2db34290c07c95cf2c5623886da5c2d9a0f9769f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/af5a648f1b9dd377cfa49ec6230af64891e0d7e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/af5a648f1b9dd377cfa49ec6230af64891e0d7e3"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
