)]}'
{"id":"openstack%2Fnetworking-ovn~734862","triplet_id":"openstack%2Fnetworking-ovn~stable%2Fstein~Ib95ecd29f3c4577ff8019df4dba2f4131bacd053","project":"openstack/networking-ovn","branch":"stable/stein","hashtags":[],"change_id":"Ib95ecd29f3c4577ff8019df4dba2f4131bacd053","subject":"Enforce Python 3 usage for rally job","status":"MERGED","created":"2020-06-10 14:39:16.000000000","updated":"2020-06-11 09:18:46.000000000","submitted":"2020-06-11 09:15:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734862-1591866937906-59896c92","meta_rev_id":"3da0fc36eddaa784293ca806d1f55748c4f7fe81","_number":734862,"virtual_id_number":734862,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-11 09:15:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2020-06-11 06:55:52.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-06-10 18:17:57.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2020-06-11 06:55:52.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 18:17:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-11 06:55:52.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-06-11 09:15:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"71e21c0d2dcf88cae0fba0162af70a58eca0526e","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-10 14:39:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"539ceb6a6f76ad3ea04f3aee387a07b55281a6c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 16:33:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b823774e23554f8eaf8d7755dbb8e13a : SUCCESS in 7m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9260d65dd8504b3884dc088fba81828e : SUCCESS in 12m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ce6f025b3404416ba04dde87caa2fa6 : SUCCESS in 6m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8f23106b6182420e8f858ecfd42a3f81 : SUCCESS in 8m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8e940b28e2f409db0a7543f2bbf27c6 : SUCCESS in 8m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa1afa17059d49de81b25e9e0e51f7ea : SUCCESS in 11m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc1a5304471c439fb292dc1104a92a9b : SUCCESS in 12m 37s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/852029f4284b4a29b85a28435b3e0268 : SUCCESS in 1h 24m 23s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/ad644650e8044218b8a6c5e03d401596 : SUCCESS in 1h 32m 40s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9d5901270ef14658aa2ed1125ca12f73 : SUCCESS in 1h 40m 23s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/47b449cdc12c4827874abbfacee53c5d : SUCCESS in 24m 11s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/3f558c19dd5344669ec87fd7c1bb2e4a : SUCCESS in 22m 33s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/36924590eb0241dba97beddc9baefed0 : SUCCESS in 1h 44m 42s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/bfcb11c0b0a94a0597da6b80c4faf04c : SUCCESS in 1h 31m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad2257c657a24e5383c0f7f4c4d4b3f9 : SUCCESS in 12m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"378aa1f00f22be3942d80d33818c9155e701bcb3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-10 18:17:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1f163652447f512108f31ef03856f9b361783199","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-06-11 06:55:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1b5a50d5539f512c796adc40c021f44d3abcda9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 06:56:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a11a825fec9e7a64dff9a203427dddb7af7108d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 09:15:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"de9ff04d5a05881a22acd5147ec6c3eb387d6885","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 09:15:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47de74e72cb54187b76a740270fbd9d4 : SUCCESS in 9m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/86f6d895c78b464bba9dc4754cc43b98 : SUCCESS in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0da23253d9b34a969ee3a0fa2b519c14 : SUCCESS in 7m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3b0fa78ab16f42fa9dfedec42f8754d1 : SUCCESS in 12m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75c3b6acd3254668af7ffc1e97055929 : SUCCESS in 11m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c886662adbd4dc1912debe3122d8558 : SUCCESS in 12m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/743ab3b39cb84fe4a16c2358bd2f294f : SUCCESS in 12m 19s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/bd64def5a27149c5911288611fbcd444 : SUCCESS in 1h 11m 21s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/c353da8499ae4adba5e1ee510e831c5a : SUCCESS in 1h 08m 24s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8d18d6da269240769e54d84e2d926fd9 : SUCCESS in 1h 04m 07s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/70ba7112c8134429baed36cc7da6aa77 : SUCCESS in 20m 22s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/3e3e907013df4489bacca88f4171a40b : SUCCESS in 19m 33s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9faaa6df66e6404ba462fd8a74cea88e : SUCCESS in 2h 17m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3da0fc36eddaa784293ca806d1f55748c4f7fe81","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 09:18:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2fcd7cc81054d218e087a29429fadc9 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f87fbc9c2497f63af3626caf6e94d6fdfc4762b4","revisions":{"f87fbc9c2497f63af3626caf6e94d6fdfc4762b4":{"kind":"REWORK","_number":1,"created":"2020-06-10 14:39:16.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/62/734862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/62/734862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/62/734862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/62/734862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/62/734862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/62/734862/1"}}},"commit":{"parents":[{"commit":"aa8a4002062277197904ae730a7074468c0c33df","subject":"Merge \"Fix ovn-db-sync-util after removing l3 ovsdb connection\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/aa8a4002062277197904ae730a7074468c0c33df"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-06-10 14:41:05.000000000","tz":120},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-06-10 14:41:05.000000000","tz":120},"subject":"Enforce Python 3 usage for rally job","message":"Enforce Python 3 usage for rally job\n\nrally-openstack has dependencies that require python3 libs.\nLets try to stick to python3 in that job.\n\nChange-Id: Ib95ecd29f3c4577ff8019df4dba2f4131bacd053\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f87fbc9c2497f63af3626caf6e94d6fdfc4762b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f87fbc9c2497f63af3626caf6e94d6fdfc4762b4"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
