)]}'
{"id":"openstack%2Fnetworking-ovn~737293","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~Ia0ba2e78d4f97294a93ff127d9098978af21ba53","project":"openstack/networking-ovn","branch":"stable/train","hashtags":[],"change_id":"Ia0ba2e78d4f97294a93ff127d9098978af21ba53","subject":"Fix functional jobs by installing missing dependencies","status":"MERGED","created":"2020-06-22 15:36:37.000000000","updated":"2020-06-24 14:25:29.000000000","submitted":"2020-06-24 13:47:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737293-1593006434486-4f349c0a","meta_rev_id":"a073448bca7d6e8a74fde285e3ded76a545fd349","_number":737293,"virtual_id_number":737293,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"date":"2020-06-24 08:16:54.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-24 13:47:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":1,"date":"2020-06-23 15:14:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2020-06-24 08:53:01.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-06-23 15:14:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2020-06-23 19:37:46.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2020-06-24 11:14:30.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-23 15:14:47.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2020-06-23 15:14:50.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-06-23 19:37:46.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-06-24 11:14:30.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-06-24 13:47:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3f9c4972198061c5c00cf300303922270fd9f246","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-06-22 15:36:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca9ec8617026b14db12e71b44f43042f20db709e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-06-22 16:02:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bb1912c54c15f6a6713b7001341e9cb6ecd1b245","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 19:16:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9f380a4286f47eea59584c26e11733d : SUCCESS in 8m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/72f65e665bb64fa4bed2744404567040 : SUCCESS in 12m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b00975635ac4e7d952d800185d4abfe : SUCCESS in 9m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/944ab58429ac4f6a9514569d94a1b3fb : SUCCESS in 16m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0647d994daeb4316ac50ff536937496a : SUCCESS in 11m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d2151ee2eb14591ad013813e3aadced : SUCCESS in 11m 49s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/adbd6c603234406fb76691e40555d1d8 : SUCCESS in 1h 37m 33s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/2b484a4330244185aea7f83c23290fd4 : SUCCESS in 1h 18m 45s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/8af02e78176543858edcead19c91c7c9 : SUCCESS in 1h 27m 01s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bcd2fea48d774939873b747c45e704c4 : SUCCESS in 2h 04m 31s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/091fffed96d34ad5a1b37b585b34fcf9 : FAILURE in 14m 18s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/84e8803878a847fca5c30cbbbdd2ecd7 : FAILURE in 19m 27s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ba0b045e140e468080095dc2ff490f66 : SUCCESS in 3h 07m 22s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/4b01e3badc40446394092ef902f5483a : SUCCESS in 1h 24m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/824a163929cd4261b700adf46457a449 : SUCCESS in 14m 55s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fa9f05500c7447bba23288fac387df8e : FAILURE in 2h 11m 57s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/08228d95152843fd8a6c00abbdc4244d : SUCCESS in 1h 20m 01s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/783ccd4f8796455b9f71a70eab2ef5c6 : FAILURE in 53m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a3ea2989723e304157a247cec9b5d5c3300b5feb","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-06-23 13:55:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"703075426e2cfe6ac2e925d72efce09f806c53fc","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-06-23 14:34:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7def0d432de9e5fb547b89cf866699fc5976251a","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-23 14:40:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f899aeb07cd3cbe19eaa87c1ce6bd468d614fba7","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-23 15:12:15.000000000","message":"Patch Set 4: Code-Review+1\n\nnetworking-ovn-dsvm-functional-py27\nnetworking-ovn-dsvm-functional\n\nare green\n\nI +1 this change","accounts_in_message":[],"_revision_number":4},{"id":"3dab1d9109d0492f299969f24def4a63bfffe72e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-06-23 15:13:52.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"66f339649aff9fc997ac73e00760d71baddfa86a","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-23 15:14:47.000000000","message":"Patch Set 5: Code-Review+1\n\nnice!","accounts_in_message":[],"_revision_number":5},{"id":"508c2bc3d12b92ddc17007db34b1d50bd20f9d60","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-23 15:14:50.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"1f4f042ff6a671005a4bfeaef08ac1c448211a77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 18:09:29.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14570ec108a74381bc1c8a445e147dc4 : SUCCESS in 6m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f517b543626846d0ae6d46ad63ec9bd2 : SUCCESS in 10m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a33023e1e5574bf8a38b15f872eb3f9a : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/690c2da2e53e40b195b0fe00aa47b968 : SUCCESS in 8m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7b5413c89e1427c92e571a72c343960 : SUCCESS in 9m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/716b0e81d6e54df58ac3ade4971161a4 : SUCCESS in 9m 03s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/01ca50f9cc7f498ba17b88af5db8ec6e : SUCCESS in 1h 29m 51s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/723acfb384f04fc790e79e34f003ac50 : SUCCESS in 1h 38m 16s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/3d3d06c366ee4c10bc1824953e888e48 : SUCCESS in 1h 52m 14s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/809a8c572df44d13b67a79c00f3c5b2e : TIMED_OUT in 2h 15m 14s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/ed7b9d0d5061474986af408a9b2cac7c : SUCCESS in 30m 57s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/0b02a7f9573c4d71b16df47c6c700636 : SUCCESS in 20m 23s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6efdcd39e1ca4060908cceb13548093c : SUCCESS in 2h 47m 38s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/e96e064d96ed4bb684aaba3e09e4a00a : SUCCESS in 1h 28m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/629bdab3a0b541dbb7952e40a932ecc2 : SUCCESS in 10m 04s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8349171c6f0b4cf9991359e4bc9685aa : FAILURE in 1h 43m 52s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/7457b4dad142496f86f24ec82d785781 : SUCCESS in 1h 23m 36s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e945948f95d8416999126aaba7dcc3b4 : FAILURE in 57m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f5d3d3e1c76640920365e3803994d66b0f674895","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-23 19:37:46.000000000","message":"Patch Set 5: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"de9d976008b28fe17a367b0609bf5a07cc61e0e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 22:22:52.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/975038087ce4442a8b53711a080f75db : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2add02531d1f452abb805a7fa55c43e6 : SUCCESS in 9m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d39414805bc42da8145d9fda19ee38e : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b553b91874e4f768a7720ef704329d5 : SUCCESS in 9m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aad6fc11e34d41a482bcedebe9af5175 : SUCCESS in 15m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d074277198ff497daf51f3bf5f16fe06 : SUCCESS in 8m 00s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/5d8b74387b61499893510fc450de81ac : SUCCESS in 56m 13s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/504d5730773f43948f2785519edf44f0 : SUCCESS in 1h 26m 53s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/a826a1124d004cdb841debb64e2566a4 : SUCCESS in 1h 28m 55s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/797ac5bdf1b44a129b71f39075a60f66 : TIMED_OUT in 2h 14m 09s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/72ed2bca4a5b4bc79b629b359facb798 : SUCCESS in 24m 31s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/df3b0c14830049f098c7ea869e900151 : SUCCESS in 14m 30s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/335d49232a644fcda8dd3cc4ca9789aa : SUCCESS in 2h 42m 35s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/498055b33e2d40c9b628c1b10ae81499 : SUCCESS in 1h 28m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/990b46725cd649479846b2a58cb3d96e : SUCCESS in 10m 37s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/36b5c2ee95764535ae897e65b56f126d : FAILURE in 2h 11m 23s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/774d29062bb044dab48baa781eba74e9 : SUCCESS in 1h 27m 09s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f297b22d3824022af3ea50ec523457f : FAILURE in 51m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"54a6932ee409e8409b4351d5d86fc0e96f8d9ad7","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-06-24 08:16:54.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"b3e01f4b4a6e60a2da72aeb224ecfd7e37e45510","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-06-24 08:53:01.000000000","message":"Patch Set 5: Code-Review+2\n\nLooks clean fix, waiting for zuul","accounts_in_message":[],"_revision_number":5},{"id":"20e21d0d6a3aca4d3176ffac202af14fcdde2f9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 11:10:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fe3b81a90a1473a839da8a05a685dc0 : SUCCESS in 6m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5720e69818444f1eb9162e4de3adfd45 : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1874365031f49cd8b0deb33cac1c53e : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94664071495e4f98a0dd02df53dd8b45 : SUCCESS in 7m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26e2e0d4ddda4b0ca500538d1f61f799 : SUCCESS in 8m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f62fb2bd06b41adadf4da96d994e795 : SUCCESS in 11m 07s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/7cf5552c1e2142b7a2049eac36c75fd6 : SUCCESS in 1h 19m 38s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/f14fd7c85ca3423d8b5d1e978f6495bf : SUCCESS in 1h 14m 11s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/05b0ef682fbb47d48a126143c4e76eae : SUCCESS in 1h 15m 39s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c6271487adc740d589724255f65bb249 : SUCCESS in 1h 44m 13s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/b1b74984ff584ee0884eceb6ab2fb29d : SUCCESS in 35m 51s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/7060ccfdb7714a5e90113e75a481cb40 : SUCCESS in 26m 04s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4684ba0cc48149b4a9bf8f573195a002 : SUCCESS in 2h 47m 20s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/b24a38ba158b496db459bb62d0bf7888 : SUCCESS in 1h 01m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9dd610df4094354857bcc1f0c6b9996 : SUCCESS in 11m 11s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6875f8da422c4a59a70ef22a26f556db : FAILURE in 2h 03m 53s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e4d5b08660ca43de80f951b889a87b99 : SUCCESS in 1h 43m 58s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e63a2f50d3724266b185282ce60ab446 : FAILURE in 1h 02m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"102f318643c9575dfddab6d41596f0f7e51a76d0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-06-24 11:14:30.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"3b5a4293d3d7996faca9d439c3123afcc1287b1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 11:14:41.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8cdbb5ee02c16a437e7718ced4c9679b8b3dc6c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 13:47:14.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d14dc77a9d4c44d5b19d4856487e6cae : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e55823387f241d89537137472b4c09a : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18e53d5d96904f268b3fc54f33315954 : SUCCESS in 9m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d25acdb9581341e298bb668e8a7f0445 : SUCCESS in 10m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71ab5a30647b454487bef116a4de4792 : SUCCESS in 9m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1ce6ac1f0d24319b0c253bb258fe9c2 : SUCCESS in 8m 30s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/431b963469504e6f9e6d719ffc71a5e9 : SUCCESS in 51m 41s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/3ae83b8d1be64ff29dff7bb044deeb64 : SUCCESS in 1h 32m 10s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d1fdb798da5b417a85294b2a28f1f14d : SUCCESS in 1h 51m 56s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/18974be20a5e4c50885e5e547c46ed2f : SUCCESS in 22m 18s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/96d9e0d9074340608ab78a4456684f5a : SUCCESS in 21m 09s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/601e459a95c54d108eb67901f0521ec0 : SUCCESS in 2h 30m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4765ff5e60734cd294bc658c0add1ad2 : SUCCESS in 10m 44s","accounts_in_message":[],"_revision_number":5},{"id":"9ce80cb1a938379420052b5b2a8ec0efef909843","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 13:47:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"4d437512459515d35c193a206bc221fef8ad51c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 13:48:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bffd155d3e7d45dea09ca8ed7c80382d : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":5},{"id":"d224be4120f7338a8d4b43a67770b5952be570b0","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-24 14:25:13.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit cb105360c45657258992c1ac9c6f51f469acd414","accounts_in_message":[],"_revision_number":5},{"id":"35886052535c65629ba653f36b4357859c3d8577","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-24 14:25:21.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 7ca6fdf65f40feeacd0f113179ae8e14861fec04","accounts_in_message":[],"_revision_number":5},{"id":"a073448bca7d6e8a74fde285e3ded76a545fd349","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-24 14:25:29.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 8bdecc2314fd6967765a56c642568a9f5e440d58","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b568395182bd6c9d1c6046375adfa44289c757b0","revisions":{"cafed62cd445f60b1dcddb206f0bc679176b1b3e":{"kind":"REWORK","_number":1,"created":"2020-06-22 15:36:37.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/93/737293/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/93/737293/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/1"}}},"commit":{"parents":[{"commit":"4cb62b7cb1cc565f3e30cbdb1c304ce6a6153816","subject":"Merge \"Fix ssh to nodes during ML2/OVS to ML2/OVN migration\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4cb62b7cb1cc565f3e30cbdb1c304ce6a6153816"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-06-22 15:36:14.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-06-22 15:36:14.000000000","tz":60},"subject":"DO NOT REVIEW: Test fix for functional jobs","message":"DO NOT REVIEW: Test fix for functional jobs\n\nChange-Id: Ia0ba2e78d4f97294a93ff127d9098978af21ba53\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/cafed62cd445f60b1dcddb206f0bc679176b1b3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/cafed62cd445f60b1dcddb206f0bc679176b1b3e"}]},"branch":"refs/heads/stable/train"},"d32dd262bab72e05e783725b368d1739aab23eb0":{"kind":"REWORK","_number":2,"created":"2020-06-22 16:02:11.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/93/737293/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/93/737293/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/2"}}},"commit":{"parents":[{"commit":"4cb62b7cb1cc565f3e30cbdb1c304ce6a6153816","subject":"Merge \"Fix ssh to nodes during ML2/OVS to ML2/OVN migration\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4cb62b7cb1cc565f3e30cbdb1c304ce6a6153816"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-06-22 15:36:14.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-06-22 16:02:02.000000000","tz":60},"subject":"DO NOT REVIEW: Test fix for functional jobs","message":"DO NOT REVIEW: Test fix for functional jobs\n\nChange-Id: Ia0ba2e78d4f97294a93ff127d9098978af21ba53\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/d32dd262bab72e05e783725b368d1739aab23eb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/d32dd262bab72e05e783725b368d1739aab23eb0"}]},"branch":"refs/heads/stable/train"},"84a3b76327631164b751f94864f118750cad4b6e":{"kind":"REWORK","_number":3,"created":"2020-06-23 13:55:34.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/93/737293/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/93/737293/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/3"}}},"commit":{"parents":[{"commit":"4cb62b7cb1cc565f3e30cbdb1c304ce6a6153816","subject":"Merge \"Fix ssh to nodes during ML2/OVS to ML2/OVN migration\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4cb62b7cb1cc565f3e30cbdb1c304ce6a6153816"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-06-22 15:36:14.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-06-23 13:55:26.000000000","tz":60},"subject":"DO NOT REVIEW: Test fix for functional jobs","message":"DO NOT REVIEW: Test fix for functional jobs\n\nChange-Id: Ia0ba2e78d4f97294a93ff127d9098978af21ba53\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/84a3b76327631164b751f94864f118750cad4b6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/84a3b76327631164b751f94864f118750cad4b6e"}]},"branch":"refs/heads/stable/train"},"49da97f520c4acbb61a1f57463b7392092e58e81":{"kind":"REWORK","_number":4,"created":"2020-06-23 14:34:21.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/93/737293/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/93/737293/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/4"}}},"commit":{"parents":[{"commit":"4cb62b7cb1cc565f3e30cbdb1c304ce6a6153816","subject":"Merge \"Fix ssh to nodes during ML2/OVS to ML2/OVN migration\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4cb62b7cb1cc565f3e30cbdb1c304ce6a6153816"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-06-22 15:36:14.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-06-23 14:34:12.000000000","tz":60},"subject":"DO NOT REVIEW: Test fix for functional jobs","message":"DO NOT REVIEW: Test fix for functional jobs\n\nChange-Id: Ia0ba2e78d4f97294a93ff127d9098978af21ba53\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/49da97f520c4acbb61a1f57463b7392092e58e81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/49da97f520c4acbb61a1f57463b7392092e58e81"}]},"branch":"refs/heads/stable/train"},"b568395182bd6c9d1c6046375adfa44289c757b0":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-06-23 15:13:52.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/93/737293/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/93/737293/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/93/737293/5"}}},"commit":{"parents":[{"commit":"4cb62b7cb1cc565f3e30cbdb1c304ce6a6153816","subject":"Merge \"Fix ssh to nodes during ML2/OVS to ML2/OVN migration\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4cb62b7cb1cc565f3e30cbdb1c304ce6a6153816"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-06-22 15:36:14.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-06-23 15:11:43.000000000","tz":60},"subject":"Fix functional jobs by installing missing dependencies","message":"Fix functional jobs by installing missing dependencies\n\nSix is needed for the OVS/OVN compilation and tox to run the tests\nitself.\n\nChange-Id: Ia0ba2e78d4f97294a93ff127d9098978af21ba53\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b568395182bd6c9d1c6046375adfa44289c757b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b568395182bd6c9d1c6046375adfa44289c757b0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
