)]}'
{"id":"openstack%2Fnetworking-ovn~737372","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~I69bb705a33e6d25035987e3f777fd793810208a4","project":"openstack/networking-ovn","branch":"stable/train","topic":"bug/1884256-stable/train","hashtags":[],"change_id":"I69bb705a33e6d25035987e3f777fd793810208a4","subject":"Add ensure-tox in functional job","status":"ABANDONED","created":"2020-06-22 20:37:51.000000000","updated":"2020-06-23 15:12:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"378d22a18427c9d14aa1b45d50472a0c7c2ff28f","_number":737372,"virtual_id_number":737372,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-22 23:19:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-22 23:19:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c1e1e8f7088ee21b9be1ca450b2a79b53e0d655","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-22 20:37:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f209edc76e19911923596b1061c62db21736d9fb","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-22 20:38:14.000000000","message":"Topic set to bug/1884256-stable/train","accounts_in_message":[],"_revision_number":1},{"id":"a000323da8100273f8ef6273bd057b5badc4ebc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 23:19:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bab7ec496ffa463b961fbda51e9f6e83 : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/23223913135142deaa113d6a8cc3e53a : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8cd976660924ca6a351fe2f66721fea : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58cd18e1c1084942a969f454e6ea8b22 : SUCCESS in 9m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c91ce0003254053b03a2838d0b95f7c : SUCCESS in 11m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aab4b825600040be8d9d4c139ef8f7fb : SUCCESS in 12m 06s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/874768dce46a421ab77be048431dae73 : SUCCESS in 1h 29m 46s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/8ddc8969bbe84a65b340bf62a06549b8 : SUCCESS in 1h 21m 51s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/3c4378da576d4d079556c82161c751e8 : SUCCESS in 1h 23m 03s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1a465fc09dd84664a5c082c137bf1460 : SUCCESS in 1h 59m 55s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/ad06fde6e2cb4e39aa84f4f99f3050e2 : FAILURE in 13m 33s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/a9552fae98ff45ba88e3674d80b49982 : FAILURE in 13m 21s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6244fd393d1547b89a7ffd2fd1106247 : SUCCESS in 2h 36m 49s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/10be7431d335475b84888b1950205edb : SUCCESS in 1h 25m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e586aaecbc864a3398c1c387c37e7c13 : SUCCESS in 13m 33s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e018789c3daa408b9a134a522ef66c56 : FAILURE in 2h 19m 18s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/14d7794227544626813de42f9df3cc85 : SUCCESS in 1h 29m 33s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff40236d57e94d6d9fdc244cec32cec7 : FAILURE in 58m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"378d22a18427c9d14aa1b45d50472a0c7c2ff28f","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-06-23 15:12:37.000000000","message":"Abandoned\n\nNah... use: https://review.opendev.org/#/c/737293","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7bbb73fc072bcc7543fa0da80d31056c21938da0","revisions":{"7bbb73fc072bcc7543fa0da80d31056c21938da0":{"kind":"REWORK","_number":1,"created":"2020-06-22 20:37:51.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/72/737372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/72/737372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/737372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/737372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/72/737372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/72/737372/1"}}},"commit":{"parents":[{"commit":"4cb62b7cb1cc565f3e30cbdb1c304ce6a6153816","subject":"Merge \"Fix ssh to nodes during ML2/OVS to ML2/OVN migration\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/4cb62b7cb1cc565f3e30cbdb1c304ce6a6153816"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-22 20:32:26.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-06-22 20:36:48.000000000","tz":-240},"subject":"Add ensure-tox in functional job","message":"Add ensure-tox in functional job\n\nnetworking-ovn functional jobs run tests using the tox env.\nEven tempest jobs need tox to be present to run the tempest\ntests in venv. But functional job derived from devstack-minimal\njobs does not make sure tox is present or not. devstack jobs does not\nmak sure that as those are not only used for tox run but also for\nother integration testing.\n\nCurrent networking-ovn functional job failure-\nhttps://zuul.opendev.org/t/openstack/build/c374add3ea02455b8464642e3a2f49c6\n\nlet\u0027s call ensure-tox role in job to make sure tox is present.\n\nRelated-Bug: 1884256\nChange-Id: I69bb705a33e6d25035987e3f777fd793810208a4\n(cherry picked from commit 86e90bfce2169194524cabf4fe6eff5e232c7b91)\n(cherry picked manually from\n   Neutron commit 8e2cf239c8c1409a3d9a25452930f766894e3657)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7bbb73fc072bcc7543fa0da80d31056c21938da0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7bbb73fc072bcc7543fa0da80d31056c21938da0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
