)]}'
{"id":"openstack%2Fnetworking-ovn~765620","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~I97b27eaa293fb161724d34da88b4398f8b590c33","project":"openstack/networking-ovn","branch":"stable/train","topic":"bug/1894136","hashtags":[],"change_id":"I97b27eaa293fb161724d34da88b4398f8b590c33","subject":"Fix leaked file descriptors by cleaning up objects","status":"MERGED","created":"2020-12-04 20:35:23.000000000","updated":"2021-01-30 08:55:43.000000000","submitted":"2021-01-30 08:53:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765620-bug/1894136","meta_rev_id":"d1822db37540b5316746590b0bcded2d5d81a2d4","_number":765620,"virtual_id_number":765620,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-30 08:53:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2021-01-29 15:52:49.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2021-01-15 11:07:54.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":1,"date":"2021-01-29 15:52:49.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-04 23:33:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-15 11:07:54.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-01-15 19:39:29.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2021-01-15 19:40:25.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"}],"messages":[{"id":"df0412d0566213834c28e2262faae4408edcbc0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-12-04 20:35:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cca0d9562ee6366a308761fbf866da11459c7879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 23:33:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5034822c4c6c49f782441351936258a3 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ed75631330b4c54bde4980353e60a6b : FAILURE in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f458a50055e247699d92584e6faf18b6 : SUCCESS in 12m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abfa510d52834bc69a117e288ef02f35 : FAILURE in 9m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/091bc3de54e847ca9aea0d5239fc710f : FAILURE in 9m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d16dbde3e164f0d95f8b6c45a643557 : FAILURE in 11m 54s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/690f5935a0de4b00b367b7aeaa29a480 : FAILURE in 2h 04m 40s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/5fac652d83394ad7a3602af97b5c257f : FAILURE in 2h 17m 59s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/4cdae1af420249f4ac14b4ce87b9b536 : FAILURE in 2h 09m 43s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ade10afcd43e479e8669a979766843e5 : SUCCESS in 1h 57m 52s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/87b4d80d4b3f4044b7b9516ceb840af2 : FAILURE in 17m 59s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/9125a4f51ec049edb440778c8f194491 : FAILURE in 8m 59s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/68d1869b46a147b6bdce42bd03199b30 : SUCCESS in 2h 30m 35s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/9025dadafa5042ca88ec8b7cc55775b3 : SUCCESS in 1h 37m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50f978473556459f893109fc0c6969a4 : FAILURE in 10m 08s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e8ce8020c92d49278d21adb3c1ac571f : FAILURE in 2h 26m 11s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/934bb638cc6e4d66874ab59fb983f1ae : SUCCESS in 1h 27m 41s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3949210da04459389dff5fbfe0f325d : SUCCESS in 45m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05bd7b3e3329f1ba136083646919d61cde2908e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-12-07 18:53:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1a14e8296702f5bdefde81b77d46746fa7fd3cc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-07 22:10:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81dae8386695437cb692dc14c4b6b53d : FAILURE in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3dddea8ee7104d6989d23ec4a4cf589b : FAILURE in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fcc6dd3d722488490f80911246c2329 : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0e708d936ad4e6a9f552de4cda8cd89 : FAILURE in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65fe3904b11e49278ecfd030cd2fe72e : FAILURE in 3m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85152d3c8e9842b4ad3b4b7af89b5151 : FAILURE in 2m 17s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/4a7940e23d40419cbf5aea2f8dc093fe : FAILURE in 2h 21m 06s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/86b0fc41b4d344f5ab5848466b9894ed : FAILURE in 2h 01m 55s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/d2418c2f7f6a4e93bf6b245dfbc71605 : FAILURE in 10m 33s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cff847b4cb70421d872879fbb6c43393 : SUCCESS in 1h 54m 51s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/8e6c6a5e6eb34380bce63744d84d8234 : FAILURE in 22m 00s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/462a3e20020e40a6a1ec3136d7865375 : FAILURE in 9m 52s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ad2d37f2856744518b5b67528aff7216 : SUCCESS in 2h 27m 12s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/8ed62b631fb9406980ef3f135abd1a79 : FAILURE in 14m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa2cdc4bcb084b6d8d3edbb5a1655dbe : FAILURE in 3m 35s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4b390b4960634308a78221bedf73ca0f : FAILURE in 2h 39m 40s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/40e463c26cf84011bf8f1c5fd4213247 : SUCCESS in 1h 19m 33s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f617ba500cd748f6a62789efe7b97231 : SUCCESS in 45m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c35d367608a0afe88853793968d1979d70b55f4b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-04 22:45:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f444865528a4397ec5e19176e59ee6536a7c40fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 01:33:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9deebcd2bc4f40b8b504f049b44ffa00 : FAILURE in 2m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c9b2674f1b546f6b381b798fd911915 : FAILURE in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/509bafb117bc4e5fbedd3b29a3d64be9 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77c307653b6743df8f93ae8142220d0e : FAILURE in 3m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df26c78d68b64a71a851d2e89324eb23 : FAILURE in 3m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8ed413bd6604f729332ae53ea225c45 : FAILURE in 2m 34s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/e4322cc52eae4833bca121d66577239d : FAILURE in 2h 03m 17s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/e79a0d446d934a44a84e536eb0bcec12 : FAILURE in 2h 12m 06s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/6075fb0faef84c2c9d8b238c66c8545b : FAILURE in 1h 55m 52s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/37f2d8a350194971bf9ef62c0e49e110 : SUCCESS in 1h 51m 07s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/27fd6fa9ef27417a9e32ef522590ca29 : FAILURE in 17m 36s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/b5f45ec580ac4f40b66b90cf2528f483 : FAILURE in 16m 16s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/fe54688dc9e2488db782ff83fe1daf9a : SUCCESS in 2h 42m 56s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/d2bb140dd2c24f1bbdf1ee2e5d5fb545 : SUCCESS in 1h 28m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a5b0e9459f54c4db7106220e2558b2c : FAILURE in 3m 54s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/140257ad252b4268b0c8aa7a537a703f : SUCCESS in 1h 32m 36s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/2b6989ef3d164d9e8575970395665cde : SUCCESS in 1h 18m 18s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bcc8c9dfc87c4342af2cd471eed334c1 : SUCCESS in 40m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2a8b63f89d6b828139a4ab110c7f465ed055b626","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-14 19:02:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"673631d1cd14bf9446b60cd3d650f5f79aa09098","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 22:13:13.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3153d5b8347e45ab93a67b66e6beaeb3 : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/feb5621caffb4ff39825cc5d54d24e05 : FAILURE in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71091261034749b788392bdb6364c339 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ef6ff69ca4f47508fdc8741baed6a6f : FAILURE in 7m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c8a70de664f94f019ae91ff1a899fcbc : FAILURE in 12m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee24adf9f22c4965bbee038fe6859b44 : FAILURE in 11m 09s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/f2850559e9de48d3bd9362417d3766b3 : FAILURE in 2h 01m 08s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/923f6719a6bc42ed88febae945d4cb52 : SUCCESS in 1h 36m 58s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/734c4623df4f4a15b93b4d746f5b254d : SUCCESS in 2h 30m 00s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/466eba2692d54518bd118213ef7898c0 : SUCCESS in 1h 58m 49s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/d319fb881a6b45828262c27cbb1208c1 : SUCCESS in 22m 28s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/f999fea88f254dabb36d55db58f710f0 : SUCCESS in 23m 46s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2f6eb9f1bb624bfb9bc449df622dcf24 : SUCCESS in 2h 47m 15s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/24dd579d2b36494c8ca87f96f92f4259 : SUCCESS in 1h 21m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b490dc4c9974efd9877594e9e127ea7 : FAILURE in 10m 08s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/210e4ab16a864a85a37a8b824bbff915 : SUCCESS in 1h 51m 13s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/43c41499f06c43b6ad9d351e7dc3a993 : SUCCESS in 1h 15m 53s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6dbc474477214940b206c8ff1cce4838 : SUCCESS in 40m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b0132abc7a286e38a8d9ed7d86d5367e6517d033","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-14 22:30:31.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"859f91728fc51d62acb09360f095b9ffdd7c03a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 01:52:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b36d574ae7ce4f9ebe0706363ec17ec7 : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/531dc89c809a4b9bab416540cafee715 : FAILURE in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/185b3fa56476479887cad592d7dbf4f5 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4fa322a8be2437d8103ba69b7f5e725 : FAILURE in 11m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b943b8d678b4a238433736f7f9134f5 : FAILURE in 12m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d0cd2185f4d4b92b2a61e2bb5217973 : FAILURE in 11m 20s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/379686a398fa41bab618c149b7351b7d : FAILURE in 1h 55m 31s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/eadb61577868451da940b6d19fea3182 : SUCCESS in 1h 51m 30s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/3a65ea1fbdab4f0697edeaebfe38c707 : SUCCESS in 1h 50m 54s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e97fd7c2a00940a1b8eb107994624223 : SUCCESS in 1h 52m 02s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/e517e99dd4a141129885ceb473226edc : SUCCESS in 18m 01s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/49622776e1c047f8ac6988f0057711be : SUCCESS in 21m 55s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4d5bae5011eb411ea4fefecbcb9e39eb : TIMED_OUT in 3h 14m 54s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/39670b7185b648408cee726c2065a31a : SUCCESS in 2h 19m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f94a3432db244aa780018242c7ebe3c7 : FAILURE in 19m 35s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1632a84b284d453ab68926105238ce4e : SUCCESS in 2h 50m 08s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/693885ac3de949d299a44b34e24b94d5 : SUCCESS in 1h 21m 57s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/30c857b8dd4a4fadb126319c97efc834 : SUCCESS in 35m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"73cc1fc19226be091fc54e932bd8852da3128f15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-15 02:04:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"43408f6d133a43cbd5f7be57d30c3899e8670769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 04:42:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc38533b6f46430ab97baebb225aa80c : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/250b5d2e267c4adaa41013f5fd23a406 : SUCCESS in 10m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a0b58f5d5454fdf910c98077a58bb59 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96d1a694e853477b8ff92d5ba0c432f7 : SUCCESS in 10m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79c311fa67054f398193c3547f5f69b2 : SUCCESS in 10m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a23cfe64ad24d908ef66d0117153660 : SUCCESS in 9m 51s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/644bb9ba3b184bb8af1733dcd91d1d53 : FAILURE in 2h 02m 12s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/86a3a6d15f29411bb688a1311b3c3ca2 : SUCCESS in 2h 09m 28s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/db70ed4579ed4091a5c2b0f0530fb3e3 : SUCCESS in 1h 58m 45s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f10f1f42aa2140dc8ab3e01949f60c44 : SUCCESS in 1h 53m 30s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/bb31c04e21ba4e138c912e6d47e39eb1 : SUCCESS in 23m 16s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/b0981cdb35154300bf3c02224343c0d2 : SUCCESS in 19m 33s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/65ef8075432e47b19b78475a7fd5b2e4 : SUCCESS in 2h 31m 18s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/5f78eca79a5948b0be3d37ed4d4ab877 : SUCCESS in 1h 21m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdc78c58966542daa873e7235ef3d931 : SUCCESS in 11m 24s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/647a7d0fe6594bd29746e2d3496466c7 : FAILURE in 2h 09m 09s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/5b5c1d25db084ca1a7fdfffc5105cb8b : SUCCESS in 1h 16m 49s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/34736a37d1e4442f87e07210764ded17 : SUCCESS in 45m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"edc5e645f097b3f09d3b9dd1f57494f88194a870","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-15 11:07:54.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3adde199ac470028e6cf0a46915f2cf9fd9fb99f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-28 21:57:40.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"12eb7f0ba486941710cb7a20eee05be67f12d0b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 00:41:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71d8b7af3e964c16bc7afb436cdb0730 : SUCCESS in 6m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d91c1b8414d46bc96e3940efc95296d : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/781e3ec31be14d1486e8dcf804466c8b : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49b0910c402a425eb93f37b9a8a4f6b3 : SUCCESS in 10m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b7deeaf27bd5406ca914cf8b3b0892a2 : SUCCESS in 10m 16s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/cd929f506c054edaa879c09e9c74aff0 : FAILURE in 2h 03m 39s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/990252613a5f4c4aa69d3a0d1cf16613 : SUCCESS in 2h 00m 28s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/423b9930dab045e4be0658a605c02276 : SUCCESS in 1h 52m 13s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bde12a1290b94447bea7031db713e2ee : SUCCESS in 1h 47m 03s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/c9bbaae09dc1402483cffa5fd0efce43 : SUCCESS in 26m 32s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/4436a65b4f0f4d9c9222e52d220bd271 : SUCCESS in 22m 26s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4c1e7bf144bd4b9998f3ada45e523a66 : SUCCESS in 2h 37m 07s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/552abed9f91e45b0b062e83240c5a10d : FAILURE in 57m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bb02ad4b4034408a2399703afaaffda : SUCCESS in 12m 01s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a4f4b0940cb04dbaa76658ccc2dcf5ef : SUCCESS in 1h 59m 09s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/b84be0aa61944ac88d120b78c201f82f : SUCCESS in 1h 15m 55s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7eb93ae0f87040578d6b3926e2babad3 : SUCCESS in 38m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"74d0cb3043e2d7709e1988c9c949726bc1d73aa3","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-01-29 15:52:49.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"704798d0c5878304d328d9033579bb47ef44fca8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 15:53:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f1baaa13da03456e816ca9083ae0f4723e5fc660","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 19:15:04.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe794d3b4c99413380ef6e1ddee05a08 : SUCCESS in 6m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fe0b3ea54b644755954c2d8eeba392d2 : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7688fed19de1419a95ea2154e38c8c84 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/698d437e7474421bb4523f10da62ddb5 : SUCCESS in 10m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c94325d098f643eb8f297c624ea1fef3 : SUCCESS in 10m 00s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/796998638c544307884f679d5bf46a9c : SUCCESS in 1h 51m 30s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/21d047b3698b428f9cdb01506db92e62 : SUCCESS in 1h 56m 05s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5a6c97a9a2494e3f914bc84cce85f303 : SUCCESS in 1h 36m 49s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/ee2f6b9fd817430caf5bed7780c61370 : SUCCESS in 21m 15s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/81d9baedc6344655bb60e0a87caf7109 : SUCCESS in 22m 42s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3795aac8215941b5ac4cd41a7238a731 : TIMED_OUT in 3h 15m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c49ee930dced4acd876f1ebbe8c2c8a1 : SUCCESS in 10m 10s","accounts_in_message":[],"_revision_number":4},{"id":"42752abe847d7b8c7be68572da1350ce80809514","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-29 20:25:32.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1e4aaf2dfb9558a0c552df576e07d996c8d68c7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 23:33:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfc902c5438442fc81e725b77afb804a : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c56aaa526394dce8e936092e6e47afc : SUCCESS in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed7e475afda74f429b882e3dd85ddaea : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8dd2def26c3f4491a5e3ab3b609e1a5a : SUCCESS in 9m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/687ff2c1d4e94a949d83a857b4648675 : SUCCESS in 10m 24s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/8bc99933e6574cc7afe70bb4a3729397 : FAILURE in 1h 52m 01s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/7d4eb7cca9a443e89487e4a8161f3870 : SUCCESS in 2h 35m 34s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/f62f6540437b452da58e9463803c9f5c : SUCCESS in 3h 05m 42s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/af521050d1f049fa83520c0af0ff1abf : SUCCESS in 1h 46m 41s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/24dd8b6e975e4999b87425b3e08d915b : SUCCESS in 16m 15s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/e0b31e9eb0da4871802dab7163387120 : FAILURE in 16m 52s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/aa2d92fb3d174e83a0c4d3d937fd6153 : SUCCESS in 2h 22m 42s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/f6c929589e1e4163a4b8837fa60b6f3f : FAILURE in 45m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb505780fe7147efb154e4cd0cb170d4 : SUCCESS in 11m 39s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/655629295e6d4f889a431aa6960bd0f4 : SUCCESS in 1h 48m 09s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/a9c2e1813e2a4db6935f9315b31676b4 : SUCCESS in 1h 20m 55s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/841a080611164ca39584f9b8ba8429f8 : SUCCESS in 37m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"42d5c5fd0fda8961d82acc3089977a1fc33a498e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-30 03:28:05.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"3513607a0711ef96a4ae9af4020f2afb6974dda4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 06:13:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84469c8b6a81484db4bc79d833e2ca46 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b3ad04565d774722b36baeb5abc78488 : SUCCESS in 11m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6da399fc25e4d33bc57d5b098c3eac9 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee8ad74f62634a90a12d5b6324bdbb1d : SUCCESS in 11m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f26f01f6be641f9b99caea7114f0516 : SUCCESS in 8m 30s\n- networking-ovn-tempest-dsvm-ovs-master https://zuul.opendev.org/t/openstack/build/1e0c126c84c54f4fa24464097c308689 : FAILURE in 2h 17m 51s (non-voting)\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/79575e31ca32474cacad58cb19025ea2 : SUCCESS in 2h 16m 29s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/ce59ee5db682475e928d2ac50b73acff : SUCCESS in 2h 00m 48s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c7bce759f13642b68afeedfd5caf9dce : SUCCESS in 2h 10m 13s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/ed4e888e6e6f44ec899593eaa05c9429 : SUCCESS in 21m 01s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/1e0122ee819e400fba6a7c523ea9b275 : SUCCESS in 24m 12s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/090c3d5dce5847eba60f0e01e81618b1 : SUCCESS in 2h 39m 40s\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/aa93a16836f94fdfb4ca1a71dd963fa7 : FAILURE in 58m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6c10ce87aee4422811de7355ad9557a : SUCCESS in 11m 05s\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c57566e35e3b4869b1cfdd008fae9c82 : SUCCESS in 1h 47m 36s (non-voting)\n- networking-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/d5b2c0113d0849e1900e5956f70686cb : SUCCESS in 1h 29m 00s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f0d70f75503543ba9a81acaa1837f4ea : SUCCESS in 36m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"858b1245daf472110356bd4a80b07bf99dc0c3ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 06:13:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"621749a9617033784e699ff748c578fa3cd57c2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 08:53:43.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c741d288885412eb8833a2ca615c977 : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bc0982a65a794a3288a46e19ed08585a : SUCCESS in 10m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc8853f979994fa7a4e58c2b1eda8a12 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40f9adfc6a1b403ead97c55325691432 : SUCCESS in 9m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8afa7cc31fb4e0fa9879d5d3416fca7 : SUCCESS in 11m 30s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/dd4bc544c4c94cafa9612df5ab8abddb : SUCCESS in 2h 22m 47s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/7a34a8a40c684ac4b2080d2ca1dbc245 : SUCCESS in 2h 27m 11s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0dd3b02b986e451bb190ea927274289b : SUCCESS in 1h 44m 30s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/a6dde5803d5b45d88eb270f336453639 : SUCCESS in 19m 43s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/60ff16a123ab41b3ab19f9d9c1c17a85 : SUCCESS in 21m 03s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/930753efd6514c2cb6a9769bebef1126 : SUCCESS in 2h 33m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8636ab98f6a5438790ce7b4af36f4642 : SUCCESS in 11m 01s","accounts_in_message":[],"_revision_number":4},{"id":"fe3ed11ea003d8fe5b238a4b4a45a1e22ae157e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 08:53:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"d1822db37540b5316746590b0bcded2d5d81a2d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 08:55:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbc64c029d8542b198aea185468b66a3 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"283002079940272c9d34af8fada8cd2d3bc16f9b","revisions":{"174a405b4b82287f8dc1eec8eb19e065950173a2":{"kind":"REWORK","_number":1,"created":"2020-12-04 20:35:23.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/20/765620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/20/765620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/20/765620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/20/765620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/20/765620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/20/765620/1"}}},"commit":{"parents":[{"commit":"53738683629d0d0fac34b612cdef1a1e243623e4","subject":"Merge \"Fix get/update/delete of non-OVN agents\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/53738683629d0d0fac34b612cdef1a1e243623e4"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-12-04 20:34:03.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-12-04 20:34:03.000000000","tz":-300},"subject":"Fix leaked file descriptors by cleaning up objects","message":"Fix leaked file descriptors by cleaning up objects\n\nA reference to the ovsdbapp Connection object is stored on the the\nAPI impl class and it is causing the object to stick around and\nits pipe fds to never close. Ultimately, python-ovs\u0027s call to\nselect() is limited to 1024 file descriptors and will eventually\nerror out. This deletes the connection and api objects explicitly\nwhen we are done with them.\n\nConflicts:\n    netoworking_ovn/octavia/ovn_driver.py\n\nChange-Id: I97b27eaa293fb161724d34da88b4398f8b590c33\nCo-authored-by: Brian Haley \u003cbhaley@redhat.com\u003e\nCloses-Bug: #1894136\n(cherry picked from ovn-octavia-provider commit 5b0715d96700d1c288f601b3ecb2cad507c2c134)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/174a405b4b82287f8dc1eec8eb19e065950173a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/174a405b4b82287f8dc1eec8eb19e065950173a2"}]},"branch":"refs/heads/stable/train"},"361ebdd8d9f5e41294e1a39100652ac0cbd857a0":{"kind":"REWORK","_number":2,"created":"2020-12-07 18:53:14.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/20/765620/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/20/765620/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/20/765620/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/20/765620/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/20/765620/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/20/765620/2"}}},"commit":{"parents":[{"commit":"53738683629d0d0fac34b612cdef1a1e243623e4","subject":"Merge \"Fix get/update/delete of non-OVN agents\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/53738683629d0d0fac34b612cdef1a1e243623e4"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-12-04 20:34:03.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-12-07 18:52:52.000000000","tz":-300},"subject":"Fix leaked file descriptors by cleaning up objects","message":"Fix leaked file descriptors by cleaning up objects\n\nA reference to the ovsdbapp Connection object is stored on the the\nAPI impl class and it is causing the object to stick around and\nits pipe fds to never close. Ultimately, python-ovs\u0027s call to\nselect() is limited to 1024 file descriptors and will eventually\nerror out. This deletes the connection and api objects explicitly\nwhen we are done with them.\n\nConflicts:\n    networking_ovn/octavia/ovn_driver.py\n\nChange-Id: I97b27eaa293fb161724d34da88b4398f8b590c33\nCo-authored-by: Brian Haley \u003cbhaley@redhat.com\u003e\nCloses-Bug: #1894136\n(cherry picked from ovn-octavia-provider commit 5b0715d96700d1c288f601b3ecb2cad507c2c134)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/361ebdd8d9f5e41294e1a39100652ac0cbd857a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/361ebdd8d9f5e41294e1a39100652ac0cbd857a0"}]},"branch":"refs/heads/stable/train"},"e5e6cf309990eedb632eb9688a9dd3f86b4672c9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-14 22:30:31.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/20/765620/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/20/765620/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/20/765620/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/20/765620/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/20/765620/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/20/765620/3"}}},"commit":{"parents":[{"commit":"7ac35c033de463dda6be65380dc58018dc48c819","subject":"Fix gate failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7ac35c033de463dda6be65380dc58018dc48c819"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-12-04 20:34:03.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2021-01-14 22:30:31.000000000","tz":0},"subject":"Fix leaked file descriptors by cleaning up objects","message":"Fix leaked file descriptors by cleaning up objects\n\nA reference to the ovsdbapp Connection object is stored on the the\nAPI impl class and it is causing the object to stick around and\nits pipe fds to never close. Ultimately, python-ovs\u0027s call to\nselect() is limited to 1024 file descriptors and will eventually\nerror out. This deletes the connection and api objects explicitly\nwhen we are done with them.\n\nConflicts:\n    networking_ovn/octavia/ovn_driver.py\n\nChange-Id: I97b27eaa293fb161724d34da88b4398f8b590c33\nCo-authored-by: Brian Haley \u003cbhaley@redhat.com\u003e\nCloses-Bug: #1894136\n(cherry picked from ovn-octavia-provider commit 5b0715d96700d1c288f601b3ecb2cad507c2c134)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e5e6cf309990eedb632eb9688a9dd3f86b4672c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e5e6cf309990eedb632eb9688a9dd3f86b4672c9"}]},"branch":"refs/heads/stable/train","description":"Rebase"},"283002079940272c9d34af8fada8cd2d3bc16f9b":{"kind":"REWORK","_number":4,"created":"2021-01-15 02:04:18.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/20/765620/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/20/765620/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/20/765620/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/20/765620/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/20/765620/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/20/765620/4"}}},"commit":{"parents":[{"commit":"7ac35c033de463dda6be65380dc58018dc48c819","subject":"Fix gate failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7ac35c033de463dda6be65380dc58018dc48c819"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-12-04 20:34:03.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-01-15 02:04:08.000000000","tz":-300},"subject":"Fix leaked file descriptors by cleaning up objects","message":"Fix leaked file descriptors by cleaning up objects\n\nA reference to the ovsdbapp Connection object is stored on the the\nAPI impl class and it is causing the object to stick around and\nits pipe fds to never close. Ultimately, python-ovs\u0027s call to\nselect() is limited to 1024 file descriptors and will eventually\nerror out. This deletes the connection and api objects explicitly\nwhen we are done with them.\n\nConflicts:\n    networking_ovn/octavia/ovn_driver.py\n\nChange-Id: I97b27eaa293fb161724d34da88b4398f8b590c33\nCo-authored-by: Brian Haley \u003cbhaley@redhat.com\u003e\nCloses-Bug: #1894136\n(cherry picked from ovn-octavia-provider commit 5b0715d96700d1c288f601b3ecb2cad507c2c134)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/283002079940272c9d34af8fada8cd2d3bc16f9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/283002079940272c9d34af8fada8cd2d3bc16f9b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
