)]}'
{"id":"openstack%2Fnetworking-ovn~791439","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~Ib468c04779af7aaf4dedf84ba885bce71078248b","project":"openstack/networking-ovn","branch":"stable/train","topic":"bug/1928450","hashtags":[],"change_id":"Ib468c04779af7aaf4dedf84ba885bce71078248b","subject":"ovn-migration: Use DHCP nodes to configure dnsmasq","status":"MERGED","created":"2021-05-14 09:46:31.000000000","updated":"2021-05-14 21:57:15.000000000","submitted":"2021-05-14 21:55:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791439-bug/1928450","meta_rev_id":"fa4748e37a39905cc42ea9d3f7d23ba7598e7f52","_number":791439,"virtual_id_number":791439,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-14 21:55:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-05-14 14:29:57.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-05-14 21:41:16.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-05-14 21:41:16.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-14 10:26:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-14 14:29:57.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-05-14 21:41:16.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"0ad7735db0b13ac18867161192d043967b583f76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2021-05-14 09:46:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"046b18a5ce47042c9a63ff327a4173750b7ac4fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 10:26:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a486a4794c84441f8fa5c798f8670b1c : SUCCESS in 7m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1036120aecf64010be2821eed3eb506a : SUCCESS in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd78d4abb4145ecbf4bbcd82932ddcd : SUCCESS in 8m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60659fce447d495daae0b0979055c347 : SUCCESS in 9m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3c7cb94e6d3143a490f5fcd1d420c123 : SUCCESS in 10m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/907734ce300d465a9fccc26a61b429ca : SUCCESS in 11m 34s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e994d9d04c2342ab9c4bde615d16b311 : SUCCESS in 38m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aaf92d93fe114e6e712e3bab2e40beea6e03a74f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-14 14:29:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ad20932fb6fbcb654648d37a03b6fc4d6795687b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-14 21:41:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9d110e9197b92fe5b7d8d60cd1d24482b1b1f548","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 21:41:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a2bcf00745009684286edf98f99e16c4b42ab1d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 21:55:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27ce40368e204916a25eb83e0400fcac : SUCCESS in 6m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b18386fc1d0443bdb83e912936876f97 : SUCCESS in 11m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/678dc56c99b54506800767abdc6276f8 : SUCCESS in 10m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f1b38d831ac49598e80ce9885e411df : SUCCESS in 9m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84f8ef3ffa3547529eb128b25aa5d26c : SUCCESS in 10m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/914bb8267aaa4436b82a75822e1065e5 : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d1f05818806d1228b7111bd0728b29ed31d422ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 21:55:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fa4748e37a39905cc42ea9d3f7d23ba7598e7f52","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 21:57:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6156ada2d6b241e39ba079c112260f21 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6bc16a089aacf60da1598304f693ab04112c5e7","revisions":{"a6bc16a089aacf60da1598304f693ab04112c5e7":{"kind":"REWORK","_number":1,"created":"2021-05-14 09:46:31.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/39/791439/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/39/791439/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/791439/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/791439/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/39/791439/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/39/791439/1"}}},"commit":{"parents":[{"commit":"ae9f5cc345dfbb7e1cc83d7ca84fb18ac1338875","subject":"Fix docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ae9f5cc345dfbb7e1cc83d7ca84fb18ac1338875"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2021-03-19 11:07:12.000000000","tz":60},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2021-05-14 09:45:57.000000000","tz":0},"subject":"ovn-migration: Use DHCP nodes to configure dnsmasq","message":"ovn-migration: Use DHCP nodes to configure dnsmasq\n\nPreviously, the migration script used controller nodes to configure\ndnsmasq when configuring new MTU. Controller nodes may not run DHCP\nagents. The patch detects DHCP agents instead and uses those to\nconfigure dnsmasq.\n\nCloses-bug: #1928450\n\nChange-Id: Ib468c04779af7aaf4dedf84ba885bce71078248b\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n(cherry picked from commit 569aa4452a42d884f365fc7800164a9f7f984687)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a6bc16a089aacf60da1598304f693ab04112c5e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a6bc16a089aacf60da1598304f693ab04112c5e7"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
