)]}'
{"id":"openstack%2Fnetworking-ovn~796063","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~I8138b31cc865aff48c434f4f537bdedb88e4a2f3","project":"openstack/networking-ovn","branch":"stable/train","topic":"hm-backport","hashtags":[],"change_id":"I8138b31cc865aff48c434f4f537bdedb88e4a2f3","subject":"Add Health Monitor support","status":"ABANDONED","created":"2021-06-11 20:55:35.000000000","updated":"2021-08-12 14:28:01.000000000","total_comment_count":3,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"0074098e16291f8c954ef22ffcac1e53a566d31f","_number":796063,"virtual_id_number":796063,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-11 11:54:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":-1,"date":"2021-08-11 09:18:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-11 22:55:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-14 20:07:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-08-11 09:18:47.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"}],"messages":[{"id":"2480edb5e76356b86da467bbb0915c6320244eaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-11 20:55:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bcc6f91f3d000c09eca0236a801ffe0b2026358","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-11 20:56:18.000000000","message":"Patch Set 1: Workflow-1\n\nThis doesn\u0027t have any unit or functional tests, just trying to get code in a happy place first.","accounts_in_message":[],"_revision_number":1},{"id":"4aea3c00de69716859754967fba0168e06c2d199","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 22:55:38.000000000","message":"Patch Set 1: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5392d6dbe3174151b99c62446353038c : FAILURE in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ab554dfded942f388efd288aadcd63c : FAILURE in 12m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/828d9faea0234a808f4db1399905cb02 : SUCCESS in 7m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47f59b6d3ebc4c10ae9ea1aea54d315e : FAILURE in 13m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4e3dd7d1a314e3c92f4a9391bc531ce : FAILURE in 10m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6dad77d558954d2d902fe0a248526f92 : FAILURE in 13m 07s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/cc4f9c490e464fe98cb313e64ef35d37 : SUCCESS in 1h 52m 56s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/6473784673054d2787f3ab3e593697b1 : FAILURE in 29m 15s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3c714b1054f8449b80bb9b9b6c882c1e : SUCCESS in 1h 56m 18s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/952f19e0a2854be2a4344d88de7cc2dd : FAILURE in 21m 17s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f1d5f956d614409bf5f3213050e109e : FAILURE in 32m 43s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/20d0018f161d495399ce6efeda89e523 : FAILURE in 51m 53s (non-voting)\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/4b8f584e4184457e8e7a971a76255be3 : SUCCESS in 1h 17m 39s (non-voting)\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/2387f839269c458a8674b4bcf1933b1e : FAILURE in 1h 05m 54s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc558bab8e8f4c629ced884de00d064b : FAILURE in 31m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"816ad648e2f3092db8eb221dd87258613e17dd6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-12 19:29:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"06d684faa696a77b2af4411ba2a2e507c8c12be2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-12 19:44:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a34043f3cbe8dd3205e2fbab01d45e0c7a71a18a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-12 19:57:59.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"ecc0960d763e2e4239d992bfca70d9c0acafa757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-12 22:40:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71401f3d15284e9fb50f82aa9364dbf0 : SUCCESS in 6m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/485bf8673c1343879e9b2e47d04a8914 : SUCCESS in 9m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6577b64e17943ac8bb1768efc8b135d : SUCCESS in 6m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ac485f5b36b43418adaaf2fae12b8ef : SUCCESS in 11m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0b95f8b248b4c30b86effaca64b75a8 : SUCCESS in 10m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d52f0b56e4e4211b72cdb4830e0accd : SUCCESS in 10m 25s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/9cb5b18158d54e7dbb0a47a143f75dd6 : SUCCESS in 2h 30m 49s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/b70bb57a34cc41bea83c934d3c2fb0b1 : SUCCESS in 2h 13m 33s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/79b51d3590764a9ba6ea02aeea7c58e3 : SUCCESS in 1h 54m 12s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/d527a12157b54aa2944a7d8b8416304e : FAILURE in 23m 05s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/9aae7ef06de145f3aaf2e3a6a4b4aab0 : FAILURE in 25m 27s\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/40cee77fdfdf4ce189b6f90d02ce61d3 : SUCCESS in 2h 52m 14s (non-voting)\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/737f5211def7469b9448f3c01b86c2eb : SUCCESS in 1h 37m 26s (non-voting)\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e0dbf03ae1ba4edd94bede6bbd717409 : SUCCESS in 1h 57m 55s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d7b32a0ff43841719a523be76aa30cec : FAILURE in 38m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c6f55cbec77586e59c1c2cd55e74178c30b65942","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-14 16:23:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7539a03b15c223c9f31729500d37ed5ce49b359b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-14 16:42:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"514213fefc1ffbc4acd1e9eb5461a521974fbec0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-14 16:43:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"faf24cece73a56ea1d76b417f9c8c62793b2fd32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-14 17:15:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"10642d273991b9beedbf3bc33caf0ca56469347c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-14 17:34:22.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"bce4884a20bfcd65fba1dc4eab8c18a4abe51b44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-14 18:18:03.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"1a7ccc3b0ae340b9e823624880714f8eff723416","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-14 20:05:21.000000000","message":"Patch Set 9:\n\nFunctional job is using schema 5.16 which doesn\u0027t have support for Load_Balancer_Health_Check table, introduced in 5.23.","accounts_in_message":[],"_revision_number":9},{"id":"3be4a28005c348f72fb77a229eba160a116bde0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-14 20:05:48.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7710b639680f65dd8e1b7bc24b3f09d6b3b1c5cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 20:06:07.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-ovn on branch stable/train.  The error was:\n\n  extra keys not allowed @ data[\u0027devstack_localrc\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: networking-ovn-dsvm-functional-master\n      parent: networking-ovn-dsvm-functional\n      devstack_localrc:\n        OVN_BRANCH: master\n        OVS_BRANCH: master\n\n  in \"openstack/networking-ovn/zuul.d/legacy-networking-ovn-jobs.yaml@stable/train\", line 58, column 3","accounts_in_message":[],"_revision_number":10},{"id":"83f8ff20fb9638ae790a79878d5db4575ad0aea9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-14 20:08:10.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f0138e5e6b01b546c155e5d346d8d5e128c44b9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 20:08:29.000000000","message":"Patch Set 11: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-ovn on branch stable/train.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/networking-ovn/zuul.d/project.yaml@stable/train\", line 16, column 19","accounts_in_message":[],"_revision_number":11},{"id":"230af11094b1242248f59e0367824dd98e09ccb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-14 20:11:35.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"358567c70d2a13f2b77eb5ea36dfd9077c7c5ef5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 20:11:57.000000000","message":"Patch Set 12: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-ovn on branch stable/train.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/networking-ovn/zuul.d/project.yaml@stable/train\", line 16, column 19","accounts_in_message":[],"_revision_number":12},{"id":"7e602cc430051916a766979ac4b067ed22deaa02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-14 20:15:03.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"01ac2e29ce51b28da3b97230aa97dfb384c0cd9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 20:15:22.000000000","message":"Patch Set 13: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-ovn on branch stable/train.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/networking-ovn/zuul.d/project.yaml@stable/train\", line 16, column 19","accounts_in_message":[],"_revision_number":13},{"id":"0c824eceef073bf078603e1889d04898ca52474b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-14 20:16:37.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"625729608b012a8343550aa61690e6150d0a2807","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-14 21:11:28.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"742f3a251d59782adf0bdc4f373f13356fd85f88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-14 21:28:34.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"3765e48bfa1feeb6291d145fe44c3092131624cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-14 22:00:44.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"02d04ccf904dbc572d3b32a7ea02ceb9b5a6ff76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-14 22:16:11.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"028d90adfcdc0c6ed540ff13539afed4a0950b26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 01:13:20.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f908b65f985e48c28c1d7556404a2a69 : SUCCESS in 8m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2c329a82adb49fd8526b92bc6dbb6d3 : SUCCESS in 10m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00e383d8d13d48b786a67dd3cd261c86 : SUCCESS in 7m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76cf30b3c41a4712afae6439b2ef39d2 : SUCCESS in 13m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad9575b7c3844077b2dade2e50db674e : SUCCESS in 12m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c0daf76dd1c84d448fa7d05053d90cb3 : SUCCESS in 13m 52s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/b38d15b05f31452fbee912265aa783b9 : SUCCESS in 2h 19m 28s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/2cde9327b8ff48ad84a8fe4bf2ddd4ac : FAILURE in 2h 09m 47s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9d9dee3a11524cc4b5b6fb5ee1f79a8c : SUCCESS in 1h 42m 55s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/7d7ff806666e4e759ac605e2dd67d7b8 : FAILURE in 23m 00s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee185e0b69414a02ba4a939ad4e9c2a4 : FAILURE in 19m 55s\n- networking-ovn-dsvm-functional-master https://zuul.opendev.org/t/openstack/build/5b50d13072f6432c80c2d045cac10b7d : FAILURE in 19m 48s (non-voting)\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/676ebe75ad5241ce8e924670f3181f71 : SUCCESS in 2h 55m 32s (non-voting)\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/dda0a42cf7f64647b25f2b1b2c4aa653 : SUCCESS in 1h 40m 13s (non-voting)\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/114c694437eb44c69c9f71c68a248e11 : FAILURE in 1h 59m 03s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cdbb5081a7fd4de4b2b5b5b199d3f4b1 : FAILURE in 37m 51s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"cddd9fd2a82032fd9b34c5e4121f10fdaf5826fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-16 19:05:54.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"6a0008e9f940b49fd8b0b307d9d8ee3fe0c04a03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 21:55:02.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9359b71a6ce6425fa06d5ebf0e575d66 : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c4eb9c67bb7f46dfb1c4af0104ce145e : SUCCESS in 8m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f871cabba0f42db8706e4987faa8f35 : SUCCESS in 6m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fb06afd4586474aaaf898169c819f6c : SUCCESS in 11m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26bc63a40fae4050a5073051b0f360ca : SUCCESS in 9m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba3d6c1d5b6d468a943d2d469e77f6a8 : SUCCESS in 10m 59s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/75616d1de9ec4a3eb9bedd1e1e81da1d : SUCCESS in 2h 08m 52s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/cae72fd626a94640aadf6698ba91dec9 : SUCCESS in 1h 57m 39s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/06e7e6404b9b49bdb334ec4d7ad3e052 : SUCCESS in 1h 40m 02s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/9d334873e32d455d96ec72213b8da224 : FAILURE in 25m 45s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/2d332e17af2840609af070ff6a74d4ce : FAILURE in 19m 56s\n- networking-ovn-dsvm-functional-master https://zuul.opendev.org/t/openstack/build/55c30cb9585a4b5ba5ae433f2604f1eb : FAILURE in 22m 58s (non-voting)\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6e8fe504a84246fc8dfe453f0aa6892d : SUCCESS in 2h 44m 57s (non-voting)\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/9362f73c61da4924830b7544b2c34fc2 : SUCCESS in 1h 26m 33s (non-voting)\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/628cd21913f04cf39f4ffcd3eb6fbc56 : SUCCESS in 2h 13m 04s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2902096617104704a028d255837ab31c : FAILURE in 46m 16s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"9e8c65a818b15d7ffa91853422549e6a9943bf89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-21 19:01:51.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"e090b2bd993eaf22979fd066d9448547701046fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 21:27:47.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2302322617454db19941fb188e5fa537 : SUCCESS in 6m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/68152ade17fc451ea166473573446563 : FAILURE in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34832339ef314e7db959fffad954ff3f : SUCCESS in 6m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b59b26c31f654e67b7726e7d423bfe15 : FAILURE in 11m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/209d598806d6412ba1da907766fabe2a : FAILURE in 10m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81a2a4a0470646869918b9709399acc5 : FAILURE in 10m 36s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/ed61b69990464864b8eff103b6a79fe2 : SUCCESS in 2h 02m 39s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/f16d0c03311440d28189356471189345 : SUCCESS in 2h 24m 16s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/90ec28abba9549e9b04b69760e865210 : FAILURE in 1h 41m 33s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/c54e8f61a8c84ecaab2cd63812d6ea5d : FAILURE in 18m 54s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/ab29bef274d5432f9b39e32dd7adbd2a : FAILURE in 20m 07s\n- networking-ovn-dsvm-functional-master https://zuul.opendev.org/t/openstack/build/b970280cf776493f88719b45fddbb9a7 : FAILURE in 20m 29s (non-voting)\n- networking-ovn-tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/feb171eaf5294625a0ed3aa0fc58197b : FAILURE in 54m 45s (non-voting)\n- networking-ovn-dsvm-grenade https://zuul.opendev.org/t/openstack/build/cebc2da89c154c0e9cfb0ff534a1a680 : SUCCESS in 1h 48m 41s (non-voting)\n- networking-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7fe411623f1b41219c67f69f751f34ac : SUCCESS in 1h 48m 50s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc610133c25c42e0909ae02469e00e30 : FAILURE in 46m 09s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"82bdfab9898bad2e03c770b0a9b0cfa97a313889","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-22 02:38:15.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"6c4e793ae31046a41d3a2f1b8f264c93df7c4722","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-22 03:07:46.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"680ccdae152e54ea3606b40bf5d3b592186a6ebd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 05:39:13.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14f50d55390b4fc18f2f7ded4fd2c137 : SUCCESS in 6m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/af4c0a3f34fe4c43a566c1201cf79d79 : FAILURE in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6cf3f0279aa49a681946c1f9e6ebfac : SUCCESS in 8m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a453b61dc2184b008170718f069a4b1f : FAILURE in 11m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de88f088530a495cb07306d531d82e5c : FAILURE in 8m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bda23179e5894df7953ca30ba4ea00df : FAILURE in 9m 24s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c1e9287a37404f65af5e9c3d7843ffce : SUCCESS in 2h 23m 06s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/a106abf9058b4659a7ca1fb79be81a2a : FAILURE in 2h 14m 47s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2bcd2141501c4c4e987ae820deb1a29b : SUCCESS in 1h 49m 00s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/4e9c1cf9a39e468aabbff67aa0cf5b21 : FAILURE in 18m 33s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/bd05427ccb7a44b499b67850c67590d4 : FAILURE in 44m 58s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d07c237a704040539f8b2a2b5dca7f95 : FAILURE in 52m 16s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"dc9f383c9503578ce8d18a0285bf9b20e6232798","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-22 19:02:38.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"8fe09be77a20bc7c4bc8f339413d7dd1638c81ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-22 19:33:57.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"00212fa28f53c59c68e75ce287b8a3ee5634ea92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-22 20:36:18.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"83ef4f85b8ec0be25c4858117d7326296a7566b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 23:11:10.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f91cb2f6be8a4b36812c16a6712eb44c : SUCCESS in 6m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2858d472a874b78ad67f4e35d090c33 : SUCCESS in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/198024d1bfab4bd59a822759e9fc7df1 : SUCCESS in 6m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f206e2cb16a44578a312a7b0e2cac66f : SUCCESS in 12m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/712d8a5316094dc88e1b5c9f251cdfc5 : SUCCESS in 10m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/592b5f1d754f4637978dc7aed943956e : SUCCESS in 11m 01s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/b38ba9fee3314ab092930888d881dff0 : FAILURE in 2h 18m 18s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/89e68b9b3dcb4943aa1166f4ec924faa : SUCCESS in 2h 28m 31s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0194fc3ce75d4262a64477b7404b8202 : SUCCESS in 1h 53m 40s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/f8efc76166884970975ea27efda5878d : FAILURE in 26m 53s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/f03e96a12f8e4106ae554335c0245524 : FAILURE in 25m 07s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/36a6c16cb8cd4137ae86c797cb172805 : FAILURE in 55m 28s (non-voting)\n- networking-ovn-octavia-v2-dsvm-scenario-focal https://zuul.opendev.org/t/openstack/build/a974db4508804ba59238333506ea9ecc : RETRY_LIMIT in 6m 22s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"5c59ef4f5b63de544dabdac49f9152a5bec1bc3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-22 23:42:57.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"1450ee7c760afc2cd196dda86c6ca0b5b3ad274f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-23 00:04:23.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"c7be257227e610688babe7418f253cc88294750f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-23 00:29:34.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"81a1a8421274596bed9402e607f9b26ddff416e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 03:18:15.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ffc92263d8b4a5c91cf8f102a92bbbe : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/12a93e1009af467a9291751ea1d6af4b : FAILURE in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61dab186f50743f08bf5c8569ef24bd7 : SUCCESS in 7m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfd4c9c4ae264642a0312ce20cb1a725 : FAILURE in 10m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4f46bcce6d647039d4a47a72eadce01 : FAILURE in 8m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f64a9afabd14983ab52a1ed5f78be6f : FAILURE in 9m 13s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/94aa8635675a4c529c6c11c0190a6127 : SUCCESS in 2h 47m 25s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/7a0fd4f07cd9459899b0219fc04e58f3 : SUCCESS in 1h 45m 22s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bb94c88556984805bcf851477077621a : SUCCESS in 1h 50m 53s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/1ba54509e743400da124a6d581f3c998 : FAILURE in 17m 20s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/654b9fffff754709b63f61bff2f3dd0b : FAILURE in 31m 21s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2fbaf6e94bbd4974b0e0d8edce82cbc6 : FAILURE in 44m 17s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"5d913473924b402788ab07ad88f65e0bad3e8270","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-23 18:55:25.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"35996d24167973295b2db581a35377345eaf36ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-23 19:17:32.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"4ba974e517e28ba42a6948f040c4c50524d0cb73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-23 20:13:00.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"5974de31df29e0c53bfe9bcd3e79b1169e9294d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 23:25:17.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1be35c849924d208d651f81aaa04284 : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ccd8b521ba654303b416190b198ef22c : SUCCESS in 10m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f5689ba5f2f46608a0f8aae7bab228f : SUCCESS in 6m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d47733b587a343f19a3ddf84d4f9e239 : SUCCESS in 11m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/630897420dfa42f4b615490901047113 : SUCCESS in 9m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/82eec4d437d54a27b9b28697075977e3 : SUCCESS in 10m 58s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/d3c7ffc6f96344f0bf916eadf30bee44 : TIMED_OUT in 3h 10m 38s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/872f721b0ffd4334b7a58fbedf36abe2 : SUCCESS in 1h 56m 56s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1d82ba48f7ad41578f2d4a73f4ce2216 : SUCCESS in 1h 38m 39s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/6bfd0b6d76ae4520ace5d5a228965058 : FAILURE in 18m 16s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/234bab120a524d8eaae18fd8f52d1e5d : FAILURE in 18m 12s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da55eed9ec124ace8ae702edcb06c7c5 : FAILURE in 45m 01s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"658474e475ed921b058ed6efe5751d58f1ba0dff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-25 21:07:04.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"530910e6fc9b7c41f409113c9a639a710d70b7d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 23:12:19.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/237a0993e1e04a05aec4870320a16575 : SUCCESS in 7m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/37b1b28ffc8d4206bf412cdf6df871ef : SUCCESS in 10m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25b7224375e64067adf278205a39675f : SUCCESS in 8m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0773d63d8a944a59249a9f38abfa442 : SUCCESS in 11m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8fe8b6619ec4449b14c347d67324322 : SUCCESS in 11m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98cff83ba20b40e593cc6217c8e5e6c8 : SUCCESS in 15m 11s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/ecc53ba948244d54aa430644592c18b4 : SUCCESS in 1h 57m 57s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/ac9c4037604642f08217bf456c222268 : SUCCESS in 2h 04m 20s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/262fef733c4d4814a8bf71d08e436158 : SUCCESS in 1h 39m 43s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/e92290abd2054933aafcc5e9616329f0 : FAILURE in 19m 55s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/7412918450a04241add993d41e3ad762 : FAILURE in 20m 26s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/091f4556e12d448a9272ab1170b287bc : FAILURE in 45m 30s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"fed434db0911195d8d73582abdaa6f5d0ebd992e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-28 17:43:14.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"f04a199f809c591126b8fd9cbe92e0fd44aa79a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-28 19:41:20.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"9cfaff5c1a2a2815c7c9b19846a6f9e90eae0b40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-28 21:37:23.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"d5b3da96f4bd748cd5123e8677af21a53cbb7296","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 00:04:56.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2676dc89733c450b9f355d838c7330b6 : SUCCESS in 8m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7ac026fc8c64a61b8e826a7fca95ab2 : SUCCESS in 10m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73e1b9943f8945e3bb97b371ade1b270 : SUCCESS in 8m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8b3b318d7f84970a288bd6a9f6edc58 : SUCCESS in 13m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7471fd59483d41589ddd0c1df4e5be6c : SUCCESS in 11m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f47fd012083488abb7fdd443dbc0abb : SUCCESS in 11m 45s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c0010a84dae0412897d3128bc60f1fd8 : FAILURE in 2h 21m 13s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/c60a02235862405e9e16c43287626666 : SUCCESS in 2h 12m 44s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0e7b7afbe2a34a47952689be875c6c59 : SUCCESS in 2h 06m 46s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/232c7888806e4e63aa6143db82f1208d : FAILURE in 30m 23s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/287c18eff99343e581b09e1c27432411 : FAILURE in 21m 54s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/188384c1cf404edf8f025595f470c776 : FAILURE in 52m 23s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"0422f1dec75df2c1fd8d55fb8450c000c322b1cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-08 16:31:33.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"2173f62d92b16d91a9e66c0e5d09c72da9ff3244","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 19:03:59.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cc9c602b7b049739217fab0015acea2 : SUCCESS in 8m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf8c2529e3d04204a314ca8b0cf7f68b : FAILURE in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b018e6c0cf74026a14dca8802a20160 : SUCCESS in 6m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6df2a7ef6be4bf89161378ccb360644 : FAILURE in 12m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/200a8394ec524d0a96c951cd15403b9b : FAILURE in 8m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/273f67f7b8bb4679aeea1cb951e28952 : FAILURE in 12m 42s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/dcb8a18de78b4ed0bcd7e720f9fd2ae2 : SUCCESS in 2h 21m 20s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/e9b07c754ac3456c9221c41286ab7223 : SUCCESS in 2h 25m 13s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cac962ef601249db849919d684ee8afb : SUCCESS in 1h 52m 35s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/f0833bde85014c4d9d224b9a03ec1e55 : FAILURE in 22m 43s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae4f16130325478ba66c5fe91bb273ed : FAILURE in 19m 46s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d7e1a809c7934baab2d8fbd17ab812c4 : FAILURE in 43m 14s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"f61e5573bf60117508269221f6d90dc92fc37e78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-08 22:39:24.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"6766c0f0c0f9fcc131421d42035d2a1385e1e268","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-08 22:56:24.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"14c7e20dd01e45117c1b8770fb7fced49a17415b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 01:23:25.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecf3b35049074b10add733fc3ac5d313 : SUCCESS in 7m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1f464cb8bc944d3896a464fe542c5ffa : SUCCESS in 10m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf1b6240db704d2a88e5f7505f13f86d : SUCCESS in 7m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b81e64bd18d34268bceaa100b4a1fc5e : SUCCESS in 10m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/407b65abb70947be94d52308884533b0 : SUCCESS in 10m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f6fb8fb4914435bac862063669e32d1 : SUCCESS in 10m 39s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/f237c4035eb046b0a6ce8f222c79f626 : FAILURE in 2h 26m 24s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/a8ee2fe19ded4cb0b9567847cb5c8b26 : SUCCESS in 2h 17m 39s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/197f380516cf48fa9b47e65b084500f3 : SUCCESS in 1h 57m 59s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/11bad7619f934553bd11410ebae86847 : FAILURE in 24m 01s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/53f1836e3b8949d5807cefa47092edc5 : FAILURE in 24m 38s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d54fd8dd647499886bdc7cdcc9759b7 : FAILURE in 51m 53s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"87001a5a230706bc093156ff3c2f572742f551d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-12 14:56:32.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"20c8a373e9c46976e20fcf1c9a75c2d12a8cf163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 17:32:23.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d9160fd8baf436fa6f446b1d66e607c : SUCCESS in 6m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a8f42285af7b4c6d9e2001340d3c0257 : SUCCESS in 10m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bfc1f3991284fc9a4380520abfe4f9c : SUCCESS in 6m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30e6282989634396bdcacd2ea42c5f47 : SUCCESS in 15m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9da3b4d0721d44f0b12a5d8cb96600d4 : SUCCESS in 9m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38275bdfedc844ecb1b159e0b5116cee : SUCCESS in 11m 58s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/a41bdc7c47e14a908ab14300728bcbac : FAILURE in 2h 29m 00s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/ee90d2326c594e379743d8f73dac4c9e : SUCCESS in 2h 06m 51s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1806acba8b4c4bd6ac3deef3aa1bc6c0 : SUCCESS in 2h 11m 25s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/df7f3eee56f045d985876212ca4319aa : FAILURE in 28m 14s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/7c0f2ee06e514f65a1577875e66fd598 : FAILURE in 24m 35s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/762b99b163424dd4b765225eadb34a32 : FAILURE in 53m 16s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"4b9d58c7d7d4255c52a7b77c76ff3bca5284974c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-12 18:36:12.000000000","message":"Patch Set 39:\n\nrecheck","accounts_in_message":[],"_revision_number":39},{"id":"d763b4edecdfba7d63be3d97f2fe7eccfe380f07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 21:13:42.000000000","message":"Patch Set 39:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39a3f1ca29e94d86a2190723c37c5e60 : SUCCESS in 6m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2b2dde7c2d8f44118b79aaf5762331e2 : SUCCESS in 9m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfaab20a4f37417aa61904f261939634 : SUCCESS in 7m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e50427e75a794bb7a1ff75edb31a3b1a : SUCCESS in 12m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bf925e6a4e64ebaa15307bc6efe298b : SUCCESS in 11m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee24862827b840dca525f6203801e81a : SUCCESS in 11m 17s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/69a3697b357542dcba76e38b8316321b : FAILURE in 2h 29m 08s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/61f6ee4371924ffabec43ad89b335428 : SUCCESS in 2h 06m 24s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bbc9c351d7e446b5a263a934b11c0d7d : SUCCESS in 1h 51m 14s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/e9f26a059f20403996a8bf87864e4456 : FAILURE in 18m 06s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/bf825087136f42a894b5e37d6a2056a6 : FAILURE in 26m 14s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/73616c990c2a4c2d9d2eb84741b09ba2 : FAILURE in 48m 57s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"e5b84b4bc53cb76a02bae6855ca8eb14b1b6a2d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-13 01:35:24.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"2d55b2e5ba9c32524763856fbcf04ee7d6deaa76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 04:20:12.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/439882a2731c403ea045432f58b7eedf : SUCCESS in 6m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4de40148e9e64070b0d6fe981efe3249 : SUCCESS in 9m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3fabcb2a03c44108042a8065dc2d95f : SUCCESS in 6m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72a28553b1964b0688ad42b70854661a : SUCCESS in 12m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afd8cced94834bc1bc4745ccd2839bc4 : SUCCESS in 9m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2f808bca293443fb9dfadb048f37221 : SUCCESS in 10m 56s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/7baf6f3d8bb449dfaccd37e154b1a2ad : FAILURE in 2h 28m 33s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/8ae2f5c7164e49a4bc32791f9b0642b5 : SUCCESS in 2h 40m 20s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/700021fe277f48dda95d446df684fb31 : SUCCESS in 1h 44m 53s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/3e09269050a7423db2ed29dbe0a79bc4 : FAILURE in 24m 55s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/b97aefd9fb094367891c8ca9e7440727 : FAILURE in 26m 28s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/373633f67fc44001a374fabafb92dc78 : FAILURE in 53m 42s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"5322068df49059153b66ec9fbf8678c869264894","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-21 21:02:11.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"8cedc46c45b8a7bf4a87b5e6d7d26b9c655b3d09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 02:18:12.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/206e681834e84a6295392188aa0a97c0 : SUCCESS in 6m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/62c8c5c8c2924fa59505e394f4121a50 : SUCCESS in 10m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b2543c816824ca9ac5f4cd1404e66e1 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b46ec00728048bca77665c337693f70 : SUCCESS in 9m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7a1b3dcee32421d8ecf3f50651a20d6 : SUCCESS in 10m 28s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c2274cb8f0a440ab8d42fa1288644e6d : SUCCESS in 2h 00m 13s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/6e8157c18ca64253a08a30b75fb48baf : SUCCESS in 2h 58m 41s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/79018eedaffd429da266fc81cb30bfcd : SUCCESS in 1h 53m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0039bfdfca2c4343adbc1052504266e3 : SUCCESS in 11m 09s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7705e71f182347f581944672470b136d : FAILURE in 57m 21s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"f5ce3de6c6422b2c88fc236272028af6d1b50a2b","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-08-11 09:18:47.000000000","message":"Patch Set 41: Code-Review-1\n\n(3 comments)\n\nHi Brian, Flavio (just added u to the review)\n\nI have a few question about this backport. First I see that it has been backproted to Wallaby and Train [0] but what about stable/ussuri or stable/victoria ?\n\nI also noted that in the last zuul run (patch-set #41) there is no functional jobs in it, I couldn\u0027t figure out why is that but, I think we need to make sure those jobs are happy with this change.\n\n[0] https://review.opendev.org/q/8bbd8f1b62d7811433079bff8ce940fbdd041a0c","accounts_in_message":[],"_revision_number":41},{"id":"432ae97aff4461e60ee1be6b04a8fd34acd09f31","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-08-11 09:19:55.000000000","message":"Patch Set 41:\n\nrecheck\n\nI don\u0027t know why the functional jobs weren\u0027t triggered in the latest run. Rechecking it.","accounts_in_message":[],"_revision_number":41},{"id":"83591d4f626d3ac5ccb162f2b3cd1bd60ca9cf85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 11:54:49.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/292cad534ef045a294d769073f62b49e : SUCCESS in 7m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4751cf2441f84df4a27faa4075aa0c49 : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8998a01d2a1d42b68633c4d50bc286b0 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd2f586d0667420aa79b1922af988df0 : SUCCESS in 9m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f20f5f7e7982446990f0867c72902ea6 : SUCCESS in 10m 15s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/4959eabee4ba4c7bb2d911e7984be507 : FAILURE in 2h 32m 51s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/753f716f484b4e818592117d11e3ffef : SUCCESS in 2h 31m 05s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/077a4828a7914b338b7c0877b9f73c49 : SUCCESS in 1h 34m 24s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/e9e58058639a4691b649516fbc3f2d7c : FAILURE in 21m 29s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/b34d8d580dbd4c05beab917fa6348be1 : FAILURE in 20m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1bb4d94e1f54cedb4abe478643c2087 : SUCCESS in 10m 44s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f190746664c4608a28db85183450905 : FAILURE in 46m 13s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"392070d6a0087d7ac34bc50841b0dbcf8cf939e3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-08-11 14:35:36.000000000","message":"Patch Set 41:\n\nHi Lucas - I never intended this to merge, was more a stepping stone for the downstream change to see if it worked at all.","accounts_in_message":[],"_revision_number":41},{"id":"d3b35f9fce6c5644e10f4af520885721bc7de6bd","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-08-11 14:57:10.000000000","message":"Patch Set 41:\n\n\u003e Patch Set 41:\n\u003e \n\u003e Hi Lucas - I never intended this to merge, was more a stepping stone for the downstream change to see if it worked at all.\n\nAh cool thanks for the reply Brian. So apparently we would have this as a d/s only change instead of an upstream one.\n\nMaybe we can abandon this patch then ?\n\nThank you again!","accounts_in_message":[],"_revision_number":41},{"id":"e00dd8b8ac4863e7a214759e2e49a10569679fb1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-08-11 19:17:48.000000000","message":"Patch Set 41:\n\nLucas - yes, this can be abandoned, it will never pass functional as the OVS/OVN version is too old on the nodes in the Train gate and it trips over a schema issue from what I remember.  The downstream version was proposed already...","accounts_in_message":[],"_revision_number":41},{"id":"0074098e16291f8c954ef22ffcac1e53a566d31f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-08-12 14:28:01.000000000","message":"Abandoned\n\nThanks Brian! I will abandon this patch then, appreciate the help! Thanks much","accounts_in_message":[],"_revision_number":41}],"current_revision_number":41,"current_revision":"a3e64e7bd660e44795da80936e6f7b589fdfbb6d","revisions":{"c23bcac701172ea4dc058fc8569055c3e1ffc594":{"kind":"REWORK","_number":1,"created":"2021-06-11 20:55:35.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/1"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c23bcac701172ea4dc058fc8569055c3e1ffc594"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c23bcac701172ea4dc058fc8569055c3e1ffc594"}]},"branch":"refs/heads/stable/train"},"a298357ff07c67d3b3806288738a2ccdbb94ac00":{"kind":"REWORK","_number":2,"created":"2021-06-12 19:29:22.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/2"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-12 19:28:55.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a298357ff07c67d3b3806288738a2ccdbb94ac00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a298357ff07c67d3b3806288738a2ccdbb94ac00"}]},"branch":"refs/heads/stable/train"},"f2435af2ea8303893c9b61017a45e8d2bc332d0a":{"kind":"REWORK","_number":3,"created":"2021-06-12 19:44:40.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/3"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-12 19:44:26.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f2435af2ea8303893c9b61017a45e8d2bc332d0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f2435af2ea8303893c9b61017a45e8d2bc332d0a"}]},"branch":"refs/heads/stable/train"},"b6dcf0d2886f822247d5c866cb7a6d73d641d195":{"kind":"REWORK","_number":4,"created":"2021-06-14 16:23:25.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/4"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-14 16:23:09.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b6dcf0d2886f822247d5c866cb7a6d73d641d195"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b6dcf0d2886f822247d5c866cb7a6d73d641d195"}]},"branch":"refs/heads/stable/train"},"90d2b78862ddd22e07aa8acbf468a603bd28d073":{"kind":"REWORK","_number":5,"created":"2021-06-14 16:42:29.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/5"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-14 16:42:19.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/90d2b78862ddd22e07aa8acbf468a603bd28d073"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/90d2b78862ddd22e07aa8acbf468a603bd28d073"}]},"branch":"refs/heads/stable/train"},"a4ed0a056c5baa9cf8eca1c022274cb83b1199fb":{"kind":"REWORK","_number":6,"created":"2021-06-14 16:43:34.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/6"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-14 16:43:25.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a4ed0a056c5baa9cf8eca1c022274cb83b1199fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a4ed0a056c5baa9cf8eca1c022274cb83b1199fb"}]},"branch":"refs/heads/stable/train"},"1128c58c001bcb9d6bf0a35f55faeed80d68c03d":{"kind":"REWORK","_number":7,"created":"2021-06-14 17:15:20.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/7"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-14 17:13:53.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/1128c58c001bcb9d6bf0a35f55faeed80d68c03d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/1128c58c001bcb9d6bf0a35f55faeed80d68c03d"}]},"branch":"refs/heads/stable/train"},"1f5c67b322f01356acaecfacb2ac001d026646df":{"kind":"REWORK","_number":8,"created":"2021-06-14 17:34:22.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/8"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-14 17:34:07.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/1f5c67b322f01356acaecfacb2ac001d026646df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/1f5c67b322f01356acaecfacb2ac001d026646df"}]},"branch":"refs/heads/stable/train"},"e185dcc0f812d94498d121a1446c399804a349a5":{"kind":"REWORK","_number":9,"created":"2021-06-14 18:18:03.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/9"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-14 18:17:51.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e185dcc0f812d94498d121a1446c399804a349a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e185dcc0f812d94498d121a1446c399804a349a5"}]},"branch":"refs/heads/stable/train"},"e57b114e81cd39215822153c2b2a250248bfecdd":{"kind":"REWORK","_number":10,"created":"2021-06-14 20:05:48.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/10"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-14 20:05:34.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e57b114e81cd39215822153c2b2a250248bfecdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e57b114e81cd39215822153c2b2a250248bfecdd"}]},"branch":"refs/heads/stable/train"},"dac5b474954bbc10b2bad289d41daf362a09b7bb":{"kind":"REWORK","_number":11,"created":"2021-06-14 20:08:10.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/11"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-14 20:08:01.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/dac5b474954bbc10b2bad289d41daf362a09b7bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/dac5b474954bbc10b2bad289d41daf362a09b7bb"}]},"branch":"refs/heads/stable/train"},"f5a7cb1d9645b11f016206873a52fa124e6915a6":{"kind":"REWORK","_number":12,"created":"2021-06-14 20:11:35.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/12"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-14 20:11:22.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f5a7cb1d9645b11f016206873a52fa124e6915a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f5a7cb1d9645b11f016206873a52fa124e6915a6"}]},"branch":"refs/heads/stable/train"},"ba0bf551b8ac72e3c7e2c27c2ab224c2bc4b1f8f":{"kind":"REWORK","_number":13,"created":"2021-06-14 20:15:03.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/13"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-14 20:14:53.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ba0bf551b8ac72e3c7e2c27c2ab224c2bc4b1f8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ba0bf551b8ac72e3c7e2c27c2ab224c2bc4b1f8f"}]},"branch":"refs/heads/stable/train"},"73891da7d6af3b3e993a3cce17fc04ecabbc2a1d":{"kind":"REWORK","_number":14,"created":"2021-06-14 20:16:37.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/14"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-14 20:16:28.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/73891da7d6af3b3e993a3cce17fc04ecabbc2a1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/73891da7d6af3b3e993a3cce17fc04ecabbc2a1d"}]},"branch":"refs/heads/stable/train"},"798c581fda97c9f6099ad6915d11bb9997ee4c52":{"kind":"REWORK","_number":15,"created":"2021-06-14 21:11:28.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/15"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-14 21:11:15.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/798c581fda97c9f6099ad6915d11bb9997ee4c52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/798c581fda97c9f6099ad6915d11bb9997ee4c52"}]},"branch":"refs/heads/stable/train"},"5f2f93e00e8bfc4a17ab0a82e3d91bbfcb2a375e":{"kind":"REWORK","_number":16,"created":"2021-06-14 21:28:34.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/16"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-14 21:26:45.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5f2f93e00e8bfc4a17ab0a82e3d91bbfcb2a375e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5f2f93e00e8bfc4a17ab0a82e3d91bbfcb2a375e"}]},"branch":"refs/heads/stable/train"},"f654660a4d6c892a58a97834c32929624bd3c476":{"kind":"REWORK","_number":17,"created":"2021-06-14 22:00:44.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/17"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-14 22:00:31.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f654660a4d6c892a58a97834c32929624bd3c476"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f654660a4d6c892a58a97834c32929624bd3c476"}]},"branch":"refs/heads/stable/train"},"e09a9878b368ace2fe15075d82022ef75a0956d2":{"kind":"REWORK","_number":18,"created":"2021-06-14 22:16:11.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/18"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-14 22:15:56.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e09a9878b368ace2fe15075d82022ef75a0956d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e09a9878b368ace2fe15075d82022ef75a0956d2"}]},"branch":"refs/heads/stable/train"},"79b3c14600d3aabc21edb8511df575470ef7027a":{"kind":"REWORK","_number":19,"created":"2021-06-16 19:05:54.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/19"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-16 19:05:36.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/79b3c14600d3aabc21edb8511df575470ef7027a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/79b3c14600d3aabc21edb8511df575470ef7027a"}]},"branch":"refs/heads/stable/train"},"bff7ef94285eb2c578bf1f7ce9ca592a7ae729ec":{"kind":"REWORK","_number":20,"created":"2021-06-21 19:01:51.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/20"}}},"commit":{"parents":[{"commit":"def1c1e2b78533b080e78a1672fe2972c227f131","subject":"Merge \"Add log for request_handler events\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/def1c1e2b78533b080e78a1672fe2972c227f131"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-21 19:01:38.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/bff7ef94285eb2c578bf1f7ce9ca592a7ae729ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/bff7ef94285eb2c578bf1f7ce9ca592a7ae729ec"}]},"branch":"refs/heads/stable/train"},"5ecc97b0337b8639ec6f6e80331a6cecb0c0d7cc":{"kind":"REWORK","_number":21,"created":"2021-06-22 02:38:15.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/21"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-22 02:38:06.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5ecc97b0337b8639ec6f6e80331a6cecb0c0d7cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5ecc97b0337b8639ec6f6e80331a6cecb0c0d7cc"}]},"branch":"refs/heads/stable/train"},"dbae7bbafd059588742167e1ec346737e98c9700":{"kind":"REWORK","_number":22,"created":"2021-06-22 03:07:46.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/22"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-22 03:07:38.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/dbae7bbafd059588742167e1ec346737e98c9700"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/dbae7bbafd059588742167e1ec346737e98c9700"}]},"branch":"refs/heads/stable/train"},"746719c94dfaf68319a02f5a67c33a51674df26e":{"kind":"REWORK","_number":23,"created":"2021-06-22 19:02:38.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/23"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-22 19:02:28.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/746719c94dfaf68319a02f5a67c33a51674df26e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/746719c94dfaf68319a02f5a67c33a51674df26e"}]},"branch":"refs/heads/stable/train"},"c242c67b856ed5e0e635d67fcb763b3c5ace4dd4":{"kind":"REWORK","_number":24,"created":"2021-06-22 19:33:57.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/24"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-22 19:33:37.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c242c67b856ed5e0e635d67fcb763b3c5ace4dd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c242c67b856ed5e0e635d67fcb763b3c5ace4dd4"}]},"branch":"refs/heads/stable/train"},"d4441e7ab8eabba56176332c42126ccecb7d57eb":{"kind":"REWORK","_number":25,"created":"2021-06-22 20:36:18.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/25"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-22 20:36:09.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/d4441e7ab8eabba56176332c42126ccecb7d57eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/d4441e7ab8eabba56176332c42126ccecb7d57eb"}]},"branch":"refs/heads/stable/train"},"ad5ede9dcb7c7557ce7a562803e091c6f13fc967":{"kind":"REWORK","_number":26,"created":"2021-06-22 23:42:57.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/26"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-22 23:42:48.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ad5ede9dcb7c7557ce7a562803e091c6f13fc967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ad5ede9dcb7c7557ce7a562803e091c6f13fc967"}]},"branch":"refs/heads/stable/train"},"c4fd5a7a521865e798c593b7a57b470d8324cf72":{"kind":"REWORK","_number":27,"created":"2021-06-23 00:04:23.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/27"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-23 00:04:15.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c4fd5a7a521865e798c593b7a57b470d8324cf72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c4fd5a7a521865e798c593b7a57b470d8324cf72"}]},"branch":"refs/heads/stable/train"},"5eb0296aeb3a8927c499e067298e7c74c74e24ea":{"kind":"REWORK","_number":28,"created":"2021-06-23 00:29:34.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/28"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-23 00:29:24.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5eb0296aeb3a8927c499e067298e7c74c74e24ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/5eb0296aeb3a8927c499e067298e7c74c74e24ea"}]},"branch":"refs/heads/stable/train"},"f10fa800ed78ebe47ad6c7acd65fbb5050d403b8":{"kind":"REWORK","_number":29,"created":"2021-06-23 18:55:25.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/29"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-23 18:55:14.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f10fa800ed78ebe47ad6c7acd65fbb5050d403b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/f10fa800ed78ebe47ad6c7acd65fbb5050d403b8"}]},"branch":"refs/heads/stable/train"},"69f71f2b85b8c14656eca667be6cdfbe83738dad":{"kind":"REWORK","_number":30,"created":"2021-06-23 19:17:32.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/30"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-23 19:17:22.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/69f71f2b85b8c14656eca667be6cdfbe83738dad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/69f71f2b85b8c14656eca667be6cdfbe83738dad"}]},"branch":"refs/heads/stable/train"},"6086bb8285b017d788722b63e096b4ec94773f81":{"kind":"REWORK","_number":31,"created":"2021-06-23 20:13:00.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/31"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-23 20:12:35.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6086bb8285b017d788722b63e096b4ec94773f81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/6086bb8285b017d788722b63e096b4ec94773f81"}]},"branch":"refs/heads/stable/train"},"efceb4c3120309ed3429a213989a8854914ca205":{"kind":"REWORK","_number":32,"created":"2021-06-25 21:07:04.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/32"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-25 21:06:41.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN distributed DHCP port as the\nsource IP for messages by default, if one doesn\u0027t exist then\na port will be created on each given subnet. The list of member\nports to monitor is updated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/efceb4c3120309ed3429a213989a8854914ca205"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/efceb4c3120309ed3429a213989a8854914ca205"}]},"branch":"refs/heads/stable/train"},"7a53508b0c7825233b779cf8df3ea2a67f590cca":{"kind":"REWORK","_number":33,"created":"2021-06-28 17:43:14.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/33"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-28 17:40:42.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN DHCP port as the source IP\nfor messages, one must exist or the create or update\noperation will fail.  The list of member ports to monitor is\nupdated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7a53508b0c7825233b779cf8df3ea2a67f590cca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/7a53508b0c7825233b779cf8df3ea2a67f590cca"}]},"branch":"refs/heads/stable/train"},"b7717fd126d6640b86fff74f67bbdd6435a818f9":{"kind":"REWORK","_number":34,"created":"2021-06-28 19:41:20.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/34"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-28 19:41:10.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN DHCP port as the source IP\nfor messages, one must exist or the create or update\noperation will fail.  The list of member ports to monitor is\nupdated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b7717fd126d6640b86fff74f67bbdd6435a818f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/b7717fd126d6640b86fff74f67bbdd6435a818f9"}]},"branch":"refs/heads/stable/train"},"c3e71d44d4902fa7fa65a8907bb61b249e7be421":{"kind":"REWORK","_number":35,"created":"2021-06-28 21:37:23.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/35"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-28 21:37:12.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN DHCP port as the source IP\nfor messages, one must exist or the create or update\noperation will fail.  The list of member ports to monitor is\nupdated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c3e71d44d4902fa7fa65a8907bb61b249e7be421"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c3e71d44d4902fa7fa65a8907bb61b249e7be421"}]},"branch":"refs/heads/stable/train"},"95f347c9219a715d5698f7b53609b26a3d3a9aee":{"kind":"REWORK","_number":36,"created":"2021-07-08 16:31:33.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/36"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-07-08 16:31:14.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN DHCP port as the source IP\nfor messages, one must exist or the create or update\noperation will fail.  The list of member ports to monitor is\nupdated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/95f347c9219a715d5698f7b53609b26a3d3a9aee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/95f347c9219a715d5698f7b53609b26a3d3a9aee"}]},"branch":"refs/heads/stable/train"},"c2bd895c1598566d3169fe170ba152bf48a6d5b4":{"kind":"REWORK","_number":37,"created":"2021-07-08 22:39:24.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/37"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-07-08 22:39:10.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN DHCP port as the source IP\nfor messages, one must exist or the create or update\noperation will fail.  The list of member ports to monitor is\nupdated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c2bd895c1598566d3169fe170ba152bf48a6d5b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/c2bd895c1598566d3169fe170ba152bf48a6d5b4"}]},"branch":"refs/heads/stable/train"},"db149edc34d43466fbd7ad224ad0bf9f66445dec":{"kind":"REWORK","_number":38,"created":"2021-07-08 22:56:24.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/38"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-07-08 22:56:14.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN DHCP port as the source IP\nfor messages, one must exist or the create or update\noperation will fail.  The list of member ports to monitor is\nupdated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/db149edc34d43466fbd7ad224ad0bf9f66445dec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/db149edc34d43466fbd7ad224ad0bf9f66445dec"}]},"branch":"refs/heads/stable/train"},"dba0f1db5f33f3a3a16570d3fdef6f62bd39fc4d":{"kind":"REWORK","_number":39,"created":"2021-07-12 14:56:32.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/39"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-07-12 14:56:05.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN DHCP port as the source IP\nfor messages, one must exist or the create or update\noperation will fail.  The list of member ports to monitor is\nupdated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/dba0f1db5f33f3a3a16570d3fdef6f62bd39fc4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/dba0f1db5f33f3a3a16570d3fdef6f62bd39fc4d"}]},"branch":"refs/heads/stable/train"},"ebf1af7bfc359609b158f081d43e4606127ec3ac":{"kind":"REWORK","_number":40,"created":"2021-07-13 01:35:24.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/40"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-07-13 01:34:48.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN DHCP port as the source IP\nfor messages, one must exist or the create or update\noperation will fail.  The list of member ports to monitor is\nupdated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ebf1af7bfc359609b158f081d43e4606127ec3ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ebf1af7bfc359609b158f081d43e4606127ec3ac"}]},"branch":"refs/heads/stable/train"},"a3e64e7bd660e44795da80936e6f7b589fdfbb6d":{"kind":"REWORK","_number":41,"created":"2021-07-21 21:02:11.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/63/796063/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/63/796063/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/63/796063/41"}}},"commit":{"parents":[{"commit":"fa14489a1631e38c326c928f7abad729489c8dbb","subject":"Merge \"[EM releases] Move non-voting jobs to the experimental queue\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/fa14489a1631e38c326c928f7abad729489c8dbb"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-06-11 20:54:20.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2021-07-21 21:01:48.000000000","tz":-240},"subject":"Add Health Monitor support","message":"Add Health Monitor support\n\nTCP and UDP Connect health monitors are now supported by the\nprovider driver, when the underlying OVN version supports\nthem.\n\nThe health monitor uses the OVN DHCP port as the source IP\nfor messages, one must exist or the create or update\noperation will fail.  The list of member ports to monitor is\nupdated whenever one is created or deleted.\n\nAdded support for SBDB notifications for Service Monitor\nevents generated when the OVN health monitor code detects\nissues with connectivity. This will trigger updates to the\nOctavia objects as members go on/offline so that the Octavia\nAPI reflects the current running state of the load balancer.\n\nAdded a lot of new unit and functional tests to cover all the\nnew code, and enabled the tempest health monitor feature so\nthose tests are run as well.\n\n(cherry-picked from ovn-octavia-provider commit 56eb4c8b0e266e820f4e0d9a00b4d02de78975e0)\n\nChange-Id: I8138b31cc865aff48c434f4f537bdedb88e4a2f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a3e64e7bd660e44795da80936e6f7b589fdfbb6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a3e64e7bd660e44795da80936e6f7b589fdfbb6d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
