)]}'
{"id":"openstack%2Fnetworking-ovn~828801","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~Ie4093dc6cd63b89e3a62363a4f805ef8287d15b9","project":"openstack/networking-ovn","branch":"stable/train","topic":"bug/1960006-stable/train","attention_set":{},"removed_from_attention_set":{"23804":{"account":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"last_update":"2022-03-07 15:03:12.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2022-03-07 15:03:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie4093dc6cd63b89e3a62363a4f805ef8287d15b9","subject":"[ovn] Prevent stale ports in the OVN database","status":"MERGED","created":"2022-02-10 22:17:22.000000000","updated":"2022-03-07 15:04:46.000000000","submitted":"2022-03-07 15:03:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828801-bug/1960006-stable/train","meta_rev_id":"9d57fe1248f3f2275950198371c893ddb30bf890","_number":828801,"virtual_id_number":828801,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-07 15:03:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-03-07 12:48:04.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-03-02 08:00:06.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2022-03-07 12:48:04.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-10 22:17:22.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2022-02-11 01:09:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-11 22:09:13.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"CC"},{"updated":"2022-02-12 02:05:29.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2022-03-02 08:00:06.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-03-07 09:33:11.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2022-03-07 12:48:04.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"227fe14cf60f985bda9c16c8b37f3aad5580df51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-02-10 22:17:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe86fb275b99534369b4f93ab7d74985041d292c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 01:09:21.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9174747ccd44df2a43c431d2c640db9 : FAILURE in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85fe6ddab3b84d94af421326fdca9668 : FAILURE in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ba35094fc7d455fa3779429de7adad3 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6d50cc8d85c452faac2cbe5080005ca : FAILURE in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db8aa71235ad45098e89ceedb965a9a4 : FAILURE in 5m 56s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/b81c06c5cf5846c98776e4c65b0f43a8 : SUCCESS in 2h 32m 57s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/89358e1447094eab812aecf26317a7af : SUCCESS in 2h 45m 19s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6484c7014be0454aa3ec3b3f0a468696 : SUCCESS in 1h 53m 18s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/0a216403a888477098eab882999fee9a : SUCCESS in 27m 43s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/53db82ff7d0745dc9ac0565a4935e25b : SUCCESS in 29m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c1222df17fb48da8b4e2dcf2af1f7df : FAILURE in 4m 46s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce4d26c8986344f0ab961c9155640bad : SUCCESS in 1h 04m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"125b455c308e4edcd7b6e03eb21e9e3d29e21d96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-02-11 22:09:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8619c69133eb4aed9de26573649eca6dd9b04419","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-12 00:45:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c73ac636a4d542d78bee7074e98e4488 : SUCCESS in 6m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/448ae53983e14f17b1d7663ec2bc1343 : FAILURE in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4963863765fb409f941c5182faf5ecbb : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e86402373db0410ba7b29e3f91d77285 : FAILURE in 8m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/36ff57dd99a54cc6bde58ba767a17e9d : FAILURE in 9m 31s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/72601f02561048d7affa7c107e56a428 : FAILURE in 2h 23m 13s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/f5ac91a26d7d42acb5e9cfcd8537c092 : SUCCESS in 2h 29m 54s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7f40a50739e34b08becbb72e9a27e1e6 : TIMED_OUT in 2h 14m 42s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/3f869abd246743d396db60363231eb1f : SUCCESS in 28m 35s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/ca217bcc6cf445c29f4f7aac9f2e21db : FAILURE in 31m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b3e14f6225b49788a25ca94406652ff : FAILURE in 11m 16s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ef0715e566345428ee57a259494a7c4 : SUCCESS in 46m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"86f9d9b4382e240cc733837bcec711d0507c8104","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-02-12 02:05:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5307da0b3fa7b923676a32fba6bef7d5a35e8b8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-12 04:41:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71a5cbe2cc2c47b9a593dc7193ac7552 : SUCCESS in 7m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a232fa80d5cb49f1a017b60b33b76953 : SUCCESS in 12m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ff974bd370845c28e297f4f62dd3169 : SUCCESS in 8m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e580363ac4a4b6690434659b97aea5e : SUCCESS in 12m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee8b5f7a2c194491a5a3dd1be31afc8c : SUCCESS in 12m 55s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/1ca6e593c73544fb909f7f9d9b340051 : FAILURE in 1h 59m 26s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/7b5a744353f4464c9fa3ca266b4fc41a : SUCCESS in 2h 09m 47s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/045177a4d0034623bd4481bf2bd6b4f8 : SUCCESS in 1h 46m 24s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/7e094a9722194892898bf94e26848aee : SUCCESS in 29m 04s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/0eaa1d9c63fb408295278cbb5471826b : SUCCESS in 21m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a46ced4e20f04d61b21f1d2fbb08efd2 : SUCCESS in 13m 24s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/38c4d2abee43422188d22f83eac8b74f : SUCCESS in 46m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"171b7b74bf67440939eff933c6e4688eb060ed60","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-02 08:00:06.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aa556126ace04bdcaeabe4defaaa85e3334d0f2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 10:27:18.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39fcfb837441480d8f6f8bee0054d517 : SUCCESS in 7m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c278bc62f5e3435388f8fb095d607319 : SUCCESS in 10m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9ff7897a41c4e66a123554c7a5549d8 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ea73b55e7be46c8a8dd4f57ba411698 : SUCCESS in 10m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a3b757c435734d7fb0a5188bbfca4306 : SUCCESS in 12m 26s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/6e96cf5cb599411ba31b202fd284a0cf : FAILURE in 1h 45m 53s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/0111be25181b4d5fa8ca63be8cebfda8 : SUCCESS in 2h 20m 48s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5d89d95cbfd44aaf9bbcd1432fdf52ba : SUCCESS in 2h 13m 40s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/b1b8fd8551fb4afc8f71881099c1c54d : SUCCESS in 24m 49s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/59bb8877ba8d4dc196ad4471b254e324 : SUCCESS in 20m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3832df36b9a241959f99dc35dffa4913 : SUCCESS in 10m 45s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f9a18099ac204fac90affaa9e2d4fdd1 : SUCCESS in 37m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e980751a0bb102ad9ccef06703b8dd5fcfcd289d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-03-07 09:33:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3dd1de6a080e83bf95b150aad5aa8e250de85a2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 11:51:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffe42031a12d47d3bbf20ca02fe7772d : SUCCESS in 6m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/86abd2266850477b8d72dd9e963e7d83 : SUCCESS in 12m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf3de77d91ff41559ca4f406ed10d7e6 : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1124881cf11347709d6bd5f3f34ba2e2 : SUCCESS in 10m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8302c2c069414bd4a23c39afb97328e4 : SUCCESS in 10m 27s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/c539cd8e0a424a52ae14c7c29c8cc236 : SUCCESS in 2h 12m 31s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/e4ef63cb929042379383fbcec895bca5 : SUCCESS in 2h 02m 05s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6d9e381cf19a4597b0f9b50c19c35162 : SUCCESS in 1h 32m 50s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/e629f0578f1b4ef1980497ba72e023c4 : SUCCESS in 21m 40s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/f1554c86a4a14e75931097b5caafac51 : SUCCESS in 23m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22131dcb833a40daa15eafc882adea84 : SUCCESS in 12m 09s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6449675b9cb4d4db57b5c5a2c61bf94 : SUCCESS in 44m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c9f9bfbb0ab1654c8f5bfcaae2843343287a8bef","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-03-07 12:48:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fb39d8ecf9e772f911235616d135c728a9f15392","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 12:49:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9e93cea0ac4e54b2e1b82a0cd7737fc13205d836","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 15:03:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af644fcbaa9d4d589dd26ba472cdd55e : SUCCESS in 7m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e54cc01bf9042678a066c92ed69c35f : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af14275ffd1343b3be99b4786a0fa669 : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9656cb3d52c443d9b8023c729320f8e : SUCCESS in 10m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78756f9a283e434095aff42ae0ccbff0 : SUCCESS in 11m 39s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/f21184000439411b8b224b7b5b42cf8e : SUCCESS in 1h 42m 56s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/185e869ebad347a9b790f2cfea7bfe67 : SUCCESS in 2h 13m 41s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2c86fc991baa4d6b905bb4ad6de55577 : SUCCESS in 1h 43m 25s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/c96f7df6666f4554b4474105bc1bdd44 : SUCCESS in 22m 31s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/63fcc0c397b64e4da48edcf1df5f6228 : SUCCESS in 21m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfc085c4ab674c9aba76a3351437a2e8 : SUCCESS in 10m 38s","accounts_in_message":[],"_revision_number":3},{"id":"b636b93250da9a04823680a3706628fa712177f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 15:03:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9d57fe1248f3f2275950198371c893ddb30bf890","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 15:04:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06e6132cf97b4e3098e960c988a32f2d : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3c78234924eea9a75576629f084b7f03f7f9bce2","revisions":{"ffb11692ff101b740dafdb217389f21e2f8a25a7":{"kind":"REWORK","_number":1,"created":"2022-02-10 22:17:22.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/01/828801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/01/828801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/01/828801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/01/828801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/01/828801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/01/828801/1"}}},"commit":{"parents":[{"commit":"a2eba10c22af8d31a4214d2a41786fabe0e17a66","subject":"Merge \"Always update router static route\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a2eba10c22af8d31a4214d2a41786fabe0e17a66"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2022-02-04 10:32:47.000000000","tz":60},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2022-02-10 22:15:52.000000000","tz":-300},"subject":"[ovn] Prevent stale ports in the OVN database","message":"[ovn] Prevent stale ports in the OVN database\n\nUnder a lot of load, there can be situations where all the Neutron\nworkers have not updated their in-memory copy of the NB database\nin time before certain operations.\n\nThis scenario can lead to stale resources when a somewhat recently\ncreated port is attempted to be deleted, but the worker handling\nthis deletion doesn\u0027t know about the OVN port yet.\n\nThis patch detects this condition and allows some time (at least one\nmaintenance task cycle) before it deletes the OVN revision number.\nIf the port then shows up in the OVN database within that window, then\nit will be deleted later by the maintenance task avoiding the stale\nports. If not, the revision number row will be deleted and we won\u0027t\nstale these entries either.\n\nConflicts:\n\tneutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n\tnetworking_ovn/common/maintenance.py\n\nCloses-Bug: #1960006\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: Ie4093dc6cd63b89e3a62363a4f805ef8287d15b9\n(cherry picked from commit be7331c8169c53e3900c9c1a08e12808cf5ed2ec)\n(cherry picked from commit 8164c27d9ad1de198a5943d75ceff473d0218b3d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ffb11692ff101b740dafdb217389f21e2f8a25a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/ffb11692ff101b740dafdb217389f21e2f8a25a7"}]},"branch":"refs/heads/stable/train"},"35fed64813cef4ef13b9304f03ebd827f1cba983":{"kind":"REWORK","_number":2,"created":"2022-02-11 22:09:13.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/01/828801/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/01/828801/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/01/828801/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/01/828801/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/01/828801/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/01/828801/2"}}},"commit":{"parents":[{"commit":"a2eba10c22af8d31a4214d2a41786fabe0e17a66","subject":"Merge \"Always update router static route\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a2eba10c22af8d31a4214d2a41786fabe0e17a66"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2022-02-04 10:32:47.000000000","tz":60},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2022-02-11 22:08:47.000000000","tz":-300},"subject":"[ovn] Prevent stale ports in the OVN database","message":"[ovn] Prevent stale ports in the OVN database\n\nUnder a lot of load, there can be situations where all the Neutron\nworkers have not updated their in-memory copy of the NB database\nin time before certain operations.\n\nThis scenario can lead to stale resources when a somewhat recently\ncreated port is attempted to be deleted, but the worker handling\nthis deletion doesn\u0027t know about the OVN port yet.\n\nThis patch detects this condition and allows some time (at least one\nmaintenance task cycle) before it deletes the OVN revision number.\nIf the port then shows up in the OVN database within that window, then\nit will be deleted later by the maintenance task avoiding the stale\nports. If not, the revision number row will be deleted and we won\u0027t\nstale these entries either.\n\nConflicts:\n\tnetworking_ovn/common/ovn_client.py\n\tnetworking_ovn/common/maintenance.py\n        networking_ovn/tests/unit/ml2/test_mech_driver.py\n\nCloses-Bug: #1960006\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: Ie4093dc6cd63b89e3a62363a4f805ef8287d15b9\n(cherry picked from commit be7331c8169c53e3900c9c1a08e12808cf5ed2ec)\n(cherry picked from commit 8164c27d9ad1de198a5943d75ceff473d0218b3d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/35fed64813cef4ef13b9304f03ebd827f1cba983"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/35fed64813cef4ef13b9304f03ebd827f1cba983"}]},"branch":"refs/heads/stable/train"},"3c78234924eea9a75576629f084b7f03f7f9bce2":{"kind":"REWORK","_number":3,"created":"2022-02-12 02:05:29.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/01/828801/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/01/828801/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/01/828801/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/01/828801/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/01/828801/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/01/828801/3"}}},"commit":{"parents":[{"commit":"a2eba10c22af8d31a4214d2a41786fabe0e17a66","subject":"Merge \"Always update router static route\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a2eba10c22af8d31a4214d2a41786fabe0e17a66"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2022-02-04 10:32:47.000000000","tz":60},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2022-02-12 02:05:24.000000000","tz":-300},"subject":"[ovn] Prevent stale ports in the OVN database","message":"[ovn] Prevent stale ports in the OVN database\n\nUnder a lot of load, there can be situations where all the Neutron\nworkers have not updated their in-memory copy of the NB database\nin time before certain operations.\n\nThis scenario can lead to stale resources when a somewhat recently\ncreated port is attempted to be deleted, but the worker handling\nthis deletion doesn\u0027t know about the OVN port yet.\n\nThis patch detects this condition and allows some time (at least one\nmaintenance task cycle) before it deletes the OVN revision number.\nIf the port then shows up in the OVN database within that window, then\nit will be deleted later by the maintenance task avoiding the stale\nports. If not, the revision number row will be deleted and we won\u0027t\nstale these entries either.\n\nConflicts:\n\tnetworking_ovn/common/ovn_client.py\n\tnetworking_ovn/common/maintenance.py\n        networking_ovn/tests/unit/ml2/test_mech_driver.py\n\nCloses-Bug: #1960006\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\nChange-Id: Ie4093dc6cd63b89e3a62363a4f805ef8287d15b9\n(cherry picked from commit be7331c8169c53e3900c9c1a08e12808cf5ed2ec)\n(cherry picked from commit 8164c27d9ad1de198a5943d75ceff473d0218b3d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/3c78234924eea9a75576629f084b7f03f7f9bce2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/3c78234924eea9a75576629f084b7f03f7f9bce2"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
