)]}'
{"id":"openstack%2Fnetworking-ovn~831349","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~I079119b50ab079e17c82066f135c2ef007dd09d3","project":"openstack/networking-ovn","branch":"stable/train","topic":"train","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-03-02 07:56:00.000000000","reason":"Rodolfo Alonso replied on the change"},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2022-03-02 16:48:30.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-03-02 16:48:30.000000000","reason":"Change was submitted"},"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2022-03-02 16:48:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I079119b50ab079e17c82066f135c2ef007dd09d3","subject":"Avoid race when trying to find a recently created loadbalancer","status":"MERGED","created":"2022-03-01 09:33:03.000000000","updated":"2022-03-02 16:50:26.000000000","submitted":"2022-03-02 16:48:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831349-train","meta_rev_id":"a2c14cf6c985c2a88ba0ac9736b03088ce922733","_number":831349,"virtual_id_number":831349,"owner":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-02 16:48:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-03-02 12:06:38.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":2,"date":"2022-03-02 14:30:07.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":1,"date":"2022-03-02 14:30:07.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-01 09:36:41.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2022-03-01 09:36:41.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-03-01 09:36:41.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-03-01 09:37:32.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"REVIEWER"},{"updated":"2022-03-01 12:05:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-01 16:05:47.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"0f37d4a23925e987ee8eba61a03e360540164264","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-03-01 09:33:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"668a0aae8f30b2face0a3736a995a5106ed9c5e9","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2022-03-01 09:37:32.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"28e68c641dfea1c8502dfeabb943ffcb4c73ff6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 12:05:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ca2ce9af8f74360bbf349378e4e6629 : SUCCESS in 7m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf0a986f080b46b2a0affad8c58ab9ab : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bfb9d26fbd645808e25f8399f248a86 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/378e0131fd37451382f72d73b8ad5add : SUCCESS in 11m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff3b2c2f44c04ac58a232444049eb4ee : SUCCESS in 11m 13s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/5bd1884a78e34d41bfee8d86237161b8 : SUCCESS in 2h 13m 24s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/d0f55d1b71fd4068878bc36536b0db6f : SUCCESS in 2h 25m 01s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c1f6a4ea2bdf4baa8db31cdf8d049894 : SUCCESS in 1h 58m 29s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/a027dba1c6274f669bc9bc51c9fa2bb0 : SUCCESS in 19m 57s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/bc380d0c109f4d3694183bd4db00a461 : SUCCESS in 20m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/957ffe2fe96d4cff8d403d5947a2f1c8 : SUCCESS in 10m 48s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ccbfde1332e422884b9bf797bf0d3a5 : SUCCESS in 37m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0f5b3f40b9689b32a4eea89d4eea26d899946fe","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-03-01 16:05:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c6442ac79af1399eff07b0ef289850929428578","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-02 07:56:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cef5ef62fc9a23686e42bf6383c0a4e853fc52e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-03-02 08:10:14.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d5a33a095487fbc3fbdf6f49c4b5ea827730acbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 11:12:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cce9ffdcb3744a1b9008be02d64a6f19 : SUCCESS in 6m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5efd679487574826b9b38d66dc1fab36 : SUCCESS in 14m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd2f0c9ed2974979a59dfc8974e16be5 : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33f8897cb79b4a5e87c6b271de875a92 : SUCCESS in 13m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b1312d06a604d0e873ddc129f4471b3 : SUCCESS in 12m 29s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/080159268dee464d8e78c7efea05f47d : FAILURE in 2h 53m 07s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/99bcd03311d8429a9039b7c92495b27f : FAILURE in 2h 37m 40s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e4229ddf2d954126b1f87281e8619ab1 : SUCCESS in 2h 13m 35s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/7751ef2fc3d541c78ef41c366c80dc93 : SUCCESS in 33m 51s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/e6185335b2de47eca6801180ab200ced : SUCCESS in 34m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/741f158808b04f2583406eb7db72acea : SUCCESS in 14m 55s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a0809a281539417ca9ef71fa086695aa : SUCCESS in 57m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"07d3034711df51bbe1da7cd143323b9edd781f7b","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-03-02 11:14:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"871d5f0a74d5292ab0232fd988b6e54e04086abf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-02 12:06:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"15e484701c47cd32ba72d5bcde92ffabd5035e4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 13:25:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b0b303ada8e4346b50da21bf59f47cb : SUCCESS in 7m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3fb9ec00be21484c993d1cf1ccbebaa1 : SUCCESS in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c559cfa5f99e4fa59e6dd00ba0e54931 : SUCCESS in 8m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bc82c52b7604e238fe2b902f09acbb4 : SUCCESS in 9m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e6461386595b48f38f0bdb12dcb6e623 : SUCCESS in 11m 07s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/6e75f4537e21473f9bf651ea6efd175b : SUCCESS in 1h 57m 10s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/359b4738df0b49bdb16b7d723ec1da56 : SUCCESS in 2h 08m 44s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4c418cd77acb4a369d475f8a08a49682 : SUCCESS in 1h 43m 03s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/ec9ee7c11f724f30b3198c2ae40a955a : SUCCESS in 19m 54s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/1d70912e4ed3436e86fa9ed6e4a22df1 : SUCCESS in 20m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cec937191e564696a9f0a799c683985b : SUCCESS in 10m 58s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/44d887b4b7e44d77a7b21d0de8f48e27 : SUCCESS in 42m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9e90fb3140c323620e99aa116914e608888546d6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-03-02 14:30:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"81f934647498ca5a7baa380e11001bc7b179a907","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 14:32:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e4c3bde648cc0b9a1d927490e1b665ada7711837","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 16:48:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abd927c080e6479bbad5034c3a2f34cf : SUCCESS in 9m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/62439f3e0adf4c928e698cf2cf2eb0e5 : SUCCESS in 12m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c965c056ca94576b14956c5e53cb6fe : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16be8117c12042bca564d9bb8d197ec9 : SUCCESS in 11m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4de3aa878e26463fa36a48f10248df2c : SUCCESS in 12m 06s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/f1a870f1346949efad9e481c48948c42 : SUCCESS in 2h 02m 13s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/3f4fbe9286fd4dd0807bb2ef5678e7de : SUCCESS in 2h 14m 51s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0b0ebb604fd84723a19f531da9358492 : SUCCESS in 1h 41m 57s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/45e9d46d566e445d9eecfb2983b84300 : SUCCESS in 24m 32s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/e4496fb459b64d7088aa166bae46947b : SUCCESS in 23m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8a7924de3c943208fecce6d292c53ce : SUCCESS in 14m 19s","accounts_in_message":[],"_revision_number":2},{"id":"144f98a99d16bfa27343cb6455578f0f9167b73b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 16:48:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a2c14cf6c985c2a88ba0ac9736b03088ce922733","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 16:50:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e9f1b5bba4b4b15b9d53b963cf50c76 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"799812ea6af243fb56190d71703b781144dc63c2","revisions":{"960dd12b2d1e89c33e4a3ec475aec6c75ca2712d":{"kind":"REWORK","_number":1,"created":"2022-03-01 09:33:03.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/49/831349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/49/831349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/831349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/831349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/831349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/49/831349/1"}}},"commit":{"parents":[{"commit":"75f38b0137056a10184efc7c58d094ce61378a3b","subject":"Merge \"Allow to create ovn loadbalancer on dual-stack provider networks\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/75f38b0137056a10184efc7c58d094ce61378a3b"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2022-03-01 09:23:50.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2022-03-01 09:23:50.000000000","tz":60},"subject":"Avoid race when trying to find a recently created loadbalancer","message":"Avoid race when trying to find a recently created loadbalancer\n\nWith Kuryr, when a service of LoadBalancer type is created the\nprocess is the next:\n- Create a load balancer\n- Associate FIP to the load balancer VIP\n\nIn busy enviroments, with HA, there may be a race condition where\nthe method to associate the FIP to the loadbalancer fails to find\nthe recently created loadbalancer, therefore not doing the FIP to\nVIP association in the OVN NB DB.\n\nNote: This was fixed in more recent branches by commit:\nhttps://opendev.org/openstack/ovn-octavia-provider/commit/c6cee9207349a12e499cbc81fe0e5d4d5bfa015c\n\nHowever, this patch is doing many unrelated things, and this retry\nmechanism should not have been part of it. Thus making the backport\nof that part independent in this patch\n\nChange-Id: I079119b50ab079e17c82066f135c2ef007dd09d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/960dd12b2d1e89c33e4a3ec475aec6c75ca2712d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/960dd12b2d1e89c33e4a3ec475aec6c75ca2712d"}]},"branch":"refs/heads/stable/train"},"799812ea6af243fb56190d71703b781144dc63c2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-03-02 08:10:14.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/49/831349/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/49/831349/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/831349/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/831349/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/49/831349/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/49/831349/2"}}},"commit":{"parents":[{"commit":"75f38b0137056a10184efc7c58d094ce61378a3b","subject":"Merge \"Allow to create ovn loadbalancer on dual-stack provider networks\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/75f38b0137056a10184efc7c58d094ce61378a3b"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2022-03-01 09:23:50.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2022-03-02 08:10:14.000000000","tz":0},"subject":"Avoid race when trying to find a recently created loadbalancer","message":"Avoid race when trying to find a recently created loadbalancer\n\nWith Kuryr, when a service of LoadBalancer type is created the\nprocess is the next:\n- Create a load balancer\n- Associate FIP to the load balancer VIP\n\nIn busy enviroments, with HA, there may be a race condition where\nthe method to associate the FIP to the loadbalancer fails to find\nthe recently created loadbalancer, therefore not doing the FIP to\nVIP association in the OVN NB DB.\n\nNote: This was fixed in more recent branches by commit:\nhttps://opendev.org/openstack/ovn-octavia-provider/commit/c6cee9207349a12e499cbc81fe0e5d4d5bfa015c\n\nHowever, this patch is doing many unrelated things, and this retry\nmechanism should not have been part of it. Thus making the backport\nof that part independent in this patch\n\nCloses-Bug: #1962713\n\nChange-Id: I079119b50ab079e17c82066f135c2ef007dd09d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/799812ea6af243fb56190d71703b781144dc63c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/799812ea6af243fb56190d71703b781144dc63c2"}]},"branch":"refs/heads/stable/train","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
