)]}'
{"id":"openstack%2Fnetworking-ovn~833728","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~Ifc2e5b2cc64c0b3beafd0a2e6eb9b9f248970cc5","project":"openstack/networking-ovn","branch":"stable/train","topic":"train","attention_set":{},"removed_from_attention_set":{"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2022-03-16 11:15:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifc2e5b2cc64c0b3beafd0a2e6eb9b9f248970cc5","subject":"Ensure no GARPs are sent for Load Balancer VIPs on tenant networks","status":"MERGED","created":"2022-03-15 08:23:34.000000000","updated":"2022-03-16 11:16:40.000000000","submitted":"2022-03-16 11:15:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833728-train","meta_rev_id":"796d1044c4aa9695f3713c383e9b8f79bdea486a","_number":833728,"virtual_id_number":833728,"owner":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-16 11:15:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2022-03-16 08:46:47.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2022-03-15 17:02:34.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":1,"date":"2022-03-16 08:46:47.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-15 11:20:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-15 17:02:34.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-03-16 08:46:47.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"}],"messages":[{"id":"2e5c7e955e95f5d8c11a4ef206207aa85b0d32a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-03-15 08:23:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14060ef78e63fc44417325ee6af1eeeef886caca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 11:20:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/160da873b4254860b56416169f39bcd0 : SUCCESS in 7m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/113539a893f14e909bcc0dde5a23bda2 : SUCCESS in 10m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e67de3b40364bd9bcbb2934677cf637 : SUCCESS in 9m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9fbd01225bf48958922536c187dad07 : SUCCESS in 13m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d4837b65f5c4da7963d256a2e33c12b : SUCCESS in 10m 46s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/f6d489dfb7de42fbb2f745c0087841af : SUCCESS in 2h 54m 24s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/391976e033604f0ebf4050aa0910e6ea : SUCCESS in 2h 12m 14s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b092a72a47034cc19f18e4cef93e4181 : SUCCESS in 1h 54m 25s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/b4a5eb6ad04d4723ac6ac95aeabb1625 : SUCCESS in 18m 22s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/cee1c5299cb44eceb50ca2757d32e899 : SUCCESS in 25m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3040ff0335a4c31b67eb90ee60356e1 : SUCCESS in 10m 56s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0a5efd895032431bab14045eafb86bb9 : SUCCESS in 47m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8459bb3aefdb6ed97167b35b2576690483dfa148","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-03-15 17:02:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"87d563e84024d48b2a676382673c9463cebcaeb7","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-03-16 08:46:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2526a44b543c9057650c929c21064e1847f87adc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 08:48:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23a2a32df925c808ce08bd4b098430323f01ebf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 11:15:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66dbc3caf960454f96c986ce64306b03 : SUCCESS in 7m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/40f34bb616ad445f9257417c19f8c00c : SUCCESS in 12m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43fce3d6bfde46859c33f1f5e819b95a : SUCCESS in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b622cd4bf5048069e38572f5e63f0e9 : SUCCESS in 10m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c9fe17e17fe42a78ddd429f3e722558 : SUCCESS in 11m 13s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/180bde65d7244da9b3d975393b4ef71a : SUCCESS in 2h 22m 55s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/e4292b08568044489fbf9cc913b2a2b5 : SUCCESS in 2h 09m 32s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2407576bf42d4aac92e6bc926b2a31e1 : SUCCESS in 1h 47m 51s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/f46c0dfab6e145489dbac0d6e24a61b7 : SUCCESS in 26m 27s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/e316da95eef24d31b4714e6f5006f8b3 : SUCCESS in 26m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca3631cc14cd4d1ea723ad15acd7be4b : SUCCESS in 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e3cfe1e9a9aabb5f4aed8a82cb1329b73c0a250f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 11:15:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"796d1044c4aa9695f3713c383e9b8f79bdea486a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 11:16:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14aed2c38ead4f6f947357da2b4ba737 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb672b0c8573d0318aa2b7a15b8f43d16206460a","revisions":{"eb672b0c8573d0318aa2b7a15b8f43d16206460a":{"kind":"REWORK","_number":1,"created":"2022-03-15 08:23:34.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/28/833728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/28/833728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/28/833728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/28/833728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/28/833728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/28/833728/1"}}},"commit":{"parents":[{"commit":"e7e0796179e892df7719a82001e1c931a0e0062f","subject":"[ovn] Add neutron network to metadata namespace names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e7e0796179e892df7719a82001e1c931a0e0062f"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2022-03-15 08:21:42.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2022-03-15 08:23:07.000000000","tz":60},"subject":"Ensure no GARPs are sent for Load Balancer VIPs on tenant networks","message":"Ensure no GARPs are sent for Load Balancer VIPs on tenant networks\n\nWhen a loadbalancer is created in tenant network (VIP and members),\nand that tenant network is connected to a router, which is connected\nto the provider network, the ovn loadbalancer gets\nassociated to the ovn logical router. If the \"router\" option is used for\n\"nat-addresses\" (as currently done), ovn-controller will send GARPs for\nthat VIP.\n\nAs there is nothing blocking different tenants in openstack to create a\nsubnet with the same CIDR and then a loadbalancer with the same VIP,\nthere may be several ovn-controllers generating GARPs on the provider\nnetwork for the same IP, each one with the MAC of the logical router\nport belonging to each tenant.\nThis could be a problem for the physical network infrastructure.\n\nThis patch fixes it by setting a new option added in OVN[1], named\n\"exclude-lb-vips-from-garp\", on the router gateway port, ensuring\nno GARPs are sent for the load balancer VIPs.\n\n[1] https://patchwork.ozlabs.org/project/ovn/patch/183edfc446633c5c38d7d7361089d34432c527dd.1645793899.git.lorenzo.bianconi@redhat.com/\n\nCloses-Bug: #1964901\n\n(manually cherry picked from commit 9d4e6edd5b4c92c88fbe8d06f6a176fae1b35f25)\n\nChange-Id: Ifc2e5b2cc64c0b3beafd0a2e6eb9b9f248970cc5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/eb672b0c8573d0318aa2b7a15b8f43d16206460a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/eb672b0c8573d0318aa2b7a15b8f43d16206460a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
