)]}'
{"id":"openstack%2Fnetworking-ovn~859629","triplet_id":"openstack%2Fnetworking-ovn~stable%2Ftrain~I85265a2ab5e75d244a6aa85263d951540540e123","project":"openstack/networking-ovn","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"34451":{"account":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"last_update":"2022-10-03 11:25:55.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-09-29 15:11:27.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}}},"hashtags":[],"change_id":"I85265a2ab5e75d244a6aa85263d951540540e123","subject":"Avoid LB ERROR status on delete when LS/LR are already deleted","status":"MERGED","created":"2022-09-28 15:06:12.000000000","updated":"2022-10-03 11:26:59.000000000","submitted":"2022-10-03 11:25:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859629","meta_rev_id":"2599db701ec2f6aea48dbc05947296c71b142bfd","_number":859629,"virtual_id_number":859629,"owner":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-03 11:25:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2022-10-03 09:26:39.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2022-09-29 15:11:27.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-29 07:17:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":1,"date":"2022-10-03 09:26:39.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-28 15:55:07.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-09-28 18:16:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-29 07:17:32.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2022-10-03 09:26:39.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"}],"messages":[{"id":"d37b6a1c585616c4f3d37f5201355addcc29c8c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2022-09-28 15:06:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5210d55da62248445f987a4bffedf800e03da08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2022-09-28 15:54:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"64bcf1ce664de29273a0e9183314cf104851af1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 18:16:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64c8cafa7be54aa8be8afb6eed7fd4d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae470b92a72c45bd835a77a71757b65b : SUCCESS in 8m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a77eb3dfd7c0414d9a2da35d84792383 : SUCCESS in 13m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7f409bf83c14de1a3ca72cde6d3406d : SUCCESS in 8m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8574a48921e64347a3e44402d7b53cc4 : SUCCESS in 11m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae166b8dbb254d8796cd0519f10afa5a : SUCCESS in 13m 54s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/2c1d5963f8df46c484bb1b3b20cec444 : SUCCESS in 2h 12m 37s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/d14422bf9d0440e9a7ab9157975d0fa8 : SUCCESS in 2h 17m 47s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4a158bc79a764d2bb066a4b49987363b : SUCCESS in 1h 45m 29s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/05f645f687444763b907b44b2bc76d29 : SUCCESS in 20m 36s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/1d59eec97d3549eab2b311c98c85158c : SUCCESS in 19m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7af8aa0e15b645198c4c86eec835b17c : SUCCESS in 10m 34s\n- networking-ovn-octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3eefb4d753aa46d98e950d86e68eb924 : SUCCESS in 48m 12s","accounts_in_message":[],"_revision_number":2},{"id":"56d6a00279255922252e81842203e3ab2c49dfcb","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-09-29 07:17:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c1c8d44428e36c705e119821a70e1c930dc81dd7","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-09-29 15:11:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4b41938b8001a3670f2336b70b2e4de36f12d25d","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-10-03 09:26:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"29244f3675f77b7e2f44b767a6dc818f07523dbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 09:27:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b29ca26b43216ac17509d5b21c131f3e95f146ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 11:25:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcb45bfe49d44737a0a82b4c18587022\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/821c1bf79bd94c09a579f432302629ad : SUCCESS in 6m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/236ba225aeda4956a2a821c24677035a : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8315c1f169144cf3b6bb1a65212aef25 : SUCCESS in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d059e8be24449b997ddea02690a37bd : SUCCESS in 8m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40a40c4800974135bc7963a2d8b56aee : SUCCESS in 10m 06s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/80d1e3c16ce140048a67d6d05236733b : SUCCESS in 1h 57m 19s\n- networking-ovn-tempest-dsvm-ovs-release-python2 https://zuul.opendev.org/t/openstack/build/13031b97404d46d8b21844e0eb47b911 : SUCCESS in 1h 51m 15s\n- networking-ovn-rally-task https://zuul.opendev.org/t/openstack/build/229fac841e1c4fa68447925fdaab494c : SUCCESS in 1h 37m 24s\n- networking-ovn-dsvm-functional-py27 https://zuul.opendev.org/t/openstack/build/2c356cdd28b74a4bbf824ac7b173f6f2 : SUCCESS in 19m 47s\n- networking-ovn-dsvm-functional https://zuul.opendev.org/t/openstack/build/283512e40f564aee9f15866ef4c0590d : SUCCESS in 20m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4682f3d6f68a4ec4b77cba1ec9738014 : SUCCESS in 9m 38s","accounts_in_message":[],"_revision_number":2},{"id":"3bb521a2623141241fcf9068aa6fee0fb65a9766","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 11:25:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2599db701ec2f6aea48dbc05947296c71b142bfd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 11:26:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d80b9a0f0ca49aba89d78be3625cd3a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf884cd707124b59801175df22476dba : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e0b4aba39d97c00223fa65632cb12b067376f28b","revisions":{"045717aeff66d282e3cbb6c5b1e79d7627a00334":{"kind":"REWORK","_number":1,"created":"2022-09-28 15:06:12.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/29/859629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/29/859629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/29/859629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/29/859629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/29/859629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/29/859629/1"}}},"commit":{"parents":[{"commit":"a1c590c0954d9de379ca44a04a90c28e9b9522b3","subject":"[OVN] Rate limit the \"Disallow caching\" log from hash ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a1c590c0954d9de379ca44a04a90c28e9b9522b3"}]}],"author":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2022-09-28 15:05:37.000000000","tz":120},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2022-09-28 15:06:10.000000000","tz":120},"subject":"Avoid LB ERROR status on delete when LS/LR are already deleted","message":"Avoid LB ERROR status on delete when LS/LR are already deleted\n\nThe delete operation of an LB lists all references to it (LS and\nLR) and in a single transaction deletes them together with the\nLB itself. If any of those deletions raise a RowNotFound exception,\nthe execution will report ERROR status.\n\nThis patch will reduce those situations, rechecking the command\nthat triggers the RowNotFound exception and passing in case a LS\nor a LR was the culprit, deleting finally the LB and reporting as\na DELETED status. In other case, the exception is raised and\nfinally the LB will be reported as ERROR status.\n\nCloses-Bug:#1990129\n\n(manually cherry picked from commit f9ccb4c48627a200ef27dde60b01bccdccac4e45)\n\nChange-Id: I85265a2ab5e75d244a6aa85263d951540540e123\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/045717aeff66d282e3cbb6c5b1e79d7627a00334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/045717aeff66d282e3cbb6c5b1e79d7627a00334"}]},"branch":"refs/heads/stable/train"},"e0b4aba39d97c00223fa65632cb12b067376f28b":{"kind":"REWORK","_number":2,"created":"2022-09-28 15:54:31.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/29/859629/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-ovn","ref":"refs/changes/29/859629/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/29/859629/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/29/859629/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-ovn refs/changes/29/859629/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-ovn refs/changes/29/859629/2"}}},"commit":{"parents":[{"commit":"a1c590c0954d9de379ca44a04a90c28e9b9522b3","subject":"[OVN] Rate limit the \"Disallow caching\" log from hash ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/a1c590c0954d9de379ca44a04a90c28e9b9522b3"}]}],"author":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2022-09-28 15:05:37.000000000","tz":120},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2022-09-28 15:54:13.000000000","tz":120},"subject":"Avoid LB ERROR status on delete when LS/LR are already deleted","message":"Avoid LB ERROR status on delete when LS/LR are already deleted\n\nThe delete operation of an LB lists all references to it (LS and\nLR) and in a single transaction deletes them together with the\nLB itself. If any of those deletions raise a RowNotFound exception,\nthe execution will report ERROR status.\n\nThis patch will reduce those situations, rechecking the command\nthat triggers the RowNotFound exception and passing in case a LS\nor a LR was the culprit, deleting finally the LB and reporting as\na DELETED status. In other case, the exception is raised and\nfinally the LB will be reported as ERROR status.\n\nCloses-Bug:#1990129\n\n(manually cherry picked from commit f9ccb4c48627a200ef27dde60b01bccdccac4e45)\n\nChange-Id: I85265a2ab5e75d244a6aa85263d951540540e123\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e0b4aba39d97c00223fa65632cb12b067376f28b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-ovn/commit/e0b4aba39d97c00223fa65632cb12b067376f28b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}}]}],"submit_requirements":[]}
