)]}'
{"id":"openstack%2Fnetworking-powervm~682805","triplet_id":"openstack%2Fnetworking-powervm~master~I514ddef2ddfbb01d7af93cb1fb4d4458d0b71a79","project":"openstack/networking-powervm","branch":"master","hashtags":[],"change_id":"I514ddef2ddfbb01d7af93cb1fb4d4458d0b71a79","subject":"fix \"How To Contribute\" url","status":"ABANDONED","created":"2019-09-18 06:40:31.000000000","updated":"2019-10-31 09:12:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6d03160e2488e7fe67b71db531289a0080bb186d","_number":682805,"virtual_id_number":682805,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-09-18 07:18:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"disliked":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":-1,"date":"2019-10-30 20:44:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":-1,"date":"2019-10-31 00:57:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-09 06:25:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 07:18:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-30 20:44:14.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-10-31 00:57:50.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"}],"messages":[{"id":"2b025a2a2cb21cf582b616c22b76d1586aec3bc0","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2019-09-18 06:40:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e829c5ee1b4a6a37dd392fe248fbf8a0de62277","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 07:18:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/184ea052389e499e8b06b00259734c1f : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cab0194fcbd468f8bc52c14bf43b887 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1e688a33967ce724cee96d87ea17f8f5c62d0390","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2019-10-09 06:25:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"481b0a6c86f4bff533e40febc70b856d9f54cd45","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-30 20:44:14.000000000","message":"Patch Set 1: Code-Review-1\n\nSee comments in https://review.opendev.org/#/c/682802/","accounts_in_message":[],"_revision_number":1},{"id":"69ad40764f652bffd26579cbdb11b3ee2b1f0647","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-30 20:44:27.000000000","message":"Patch Set 1:\n\nAlso, you shouldn\u0027t vote on your own patch.","accounts_in_message":[],"_revision_number":1},{"id":"0cb751898a1d3ee2728133e4a6f844e1b7fafcfb","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-10-31 00:57:50.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"6d03160e2488e7fe67b71db531289a0080bb186d","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2019-10-31 09:12:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69b90e6cf700f15806648adaa55c4a7c66b9c21d","revisions":{"69b90e6cf700f15806648adaa55c4a7c66b9c21d":{"kind":"REWORK","_number":1,"created":"2019-09-18 06:40:31.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/05/682805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-powervm","ref":"refs/changes/05/682805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-powervm refs/changes/05/682805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-powervm refs/changes/05/682805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-powervm refs/changes/05/682805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-powervm refs/changes/05/682805/1"}}},"commit":{"parents":[{"commit":"50b81c3b2d874ecc0d81049cbcb265618534e4ba","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-powervm/commit/50b81c3b2d874ecc0d81049cbcb265618534e4ba"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2019-09-18 06:38:19.000000000","tz":480},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2019-09-18 06:40:20.000000000","tz":480},"subject":"fix \"How To Contribute\" url","message":"fix \"How To Contribute\" url\n\nChange-Id: I514ddef2ddfbb01d7af93cb1fb4d4458d0b71a79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-powervm/commit/69b90e6cf700f15806648adaa55c4a7c66b9c21d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-powervm/commit/69b90e6cf700f15806648adaa55c4a7c66b9c21d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
