)]}'
{"id":"openstack%2Fnetworking-powervm~701375","triplet_id":"openstack%2Fnetworking-powervm~master~Ic4d97db6c947264806e5ef9e35f21f764cbecf46","project":"openstack/networking-powervm","branch":"master","topic":"fix-networking-powervm-gate","hashtags":[],"change_id":"Ic4d97db6c947264806e5ef9e35f21f764cbecf46","subject":"Adjust version constraint of hacking","status":"MERGED","created":"2020-01-07 14:49:21.000000000","updated":"2020-01-10 07:15:49.000000000","submitted":"2020-01-10 07:15:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701375-1578640549560-46a14a57","meta_rev_id":"b930fb831c54b208f8b7374d618045a29c56de7a","_number":701375,"virtual_id_number":701375,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-07 15:31:51.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":0,"_account_id":14581,"name":"ymadhavi@in.ibm.com","email":"ymadhavi@in.ibm.com","username":"ymadhavi"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":0,"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-10 07:15:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":0,"_account_id":14581,"name":"ymadhavi@in.ibm.com","email":"ymadhavi@in.ibm.com","username":"ymadhavi"},{"value":2,"date":"2020-01-07 22:53:47.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":0,"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2020-01-09 14:23:28.000000000","_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":0,"date":"2020-01-09 14:23:49.000000000","_account_id":14581,"name":"ymadhavi@in.ibm.com","email":"ymadhavi@in.ibm.com","username":"ymadhavi"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2020-01-09 14:23:04.000000000","_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":1,"date":"2020-01-10 06:36:18.000000000","_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-09 14:23:39.000000000","_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14581,"name":"ymadhavi@in.ibm.com","email":"ymadhavi@in.ibm.com","username":"ymadhavi"},{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-07 22:53:47.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-01-09 14:23:04.000000000","updated_by":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"reviewer":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"state":"REVIEWER"},{"updated":"2020-01-09 14:23:28.000000000","updated_by":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"reviewer":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"state":"REVIEWER"},{"updated":"2020-01-09 14:23:39.000000000","updated_by":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"reviewer":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"state":"REVIEWER"},{"updated":"2020-01-09 14:23:49.000000000","updated_by":{"_account_id":14581,"name":"ymadhavi@in.ibm.com","email":"ymadhavi@in.ibm.com","username":"ymadhavi"},"reviewer":{"_account_id":14581,"name":"ymadhavi@in.ibm.com","email":"ymadhavi@in.ibm.com","username":"ymadhavi"},"state":"REVIEWER"},{"updated":"2020-01-10 06:36:18.000000000","updated_by":{"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},"reviewer":{"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},"state":"REVIEWER"},{"updated":"2020-01-10 07:15:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"846cb3a210ea3cf41cfc702d78b428ec13ccc61b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-01-07 14:49:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31ea7d78f3c8167ba1a322dc4208b2feb683d9c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 15:14:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18d0df09103b4358b0d8fd47227eb7a5 : SUCCESS in 5m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0ee286c026814a4aba98d7d6187c1e0f : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9739e0fb3e9462b9406cf21772f3583 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbda01c1e58346689d9a9da0a345eee5 : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/212c088494c94813806cae459211a3dd : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f922ad9357046eb9825d9b7340fd247 : SUCCESS in 7m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a92dc5ffaa07456aaa52684c279f0557 : SUCCESS in 7m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ce11af9994562f74617cfbce2cd646671674d17c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-01-07 15:20:35.000000000","message":"Topic set to fix-networking-powervm-gate","accounts_in_message":[],"_revision_number":1},{"id":"bc3674081c985828348883323db99dee71c7ca0d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-01-07 15:29:29.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit c9dbe657ace877ca369729e96d01f3928ca2f41b","accounts_in_message":[],"_revision_number":1},{"id":"02bc67cf41d6fb1a88c9dedcf87d2318c4830429","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-01-07 15:31:51.000000000","message":"Patch Set 1:\n\nexample for the failure from today\u0027s periodic:\n\nhttps://zuul.opendev.org/t/openstack/build/e1a76156d35a4fa5a7e78037e4ce24d3\nhttps://zuul.opendev.org/t/openstack/build/ad39ba57114f402ea645f57625fe19be\nhttps://zuul.opendev.org/t/openstack/build/c39fc9f63c894724813f9ef4103bcada","accounts_in_message":[],"_revision_number":1},{"id":"06a4f2361e9394c32bb6e8862542c2ed32b33b8e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-07 22:53:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2e81c2461cedfd5571ee460fbed3bf6fea73083e","author":{"_account_id":16551,"name":"Arun Mani","email":"arun.mani@in.ibm.com","username":"arunmani"},"date":"2020-01-10 06:36:18.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ff0811da091cfcb5af43d08b79f25abbcaa0378","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 06:36:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06d2300376f1f38930b0a0063a885a4da26b4859","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 07:15:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99dda10d71784e52978667c82958de5a : SUCCESS in 5m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/996b574cd52647a8905aa63ab39b5835 : SUCCESS in 3m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6c3dc77756d495baf12879f42f5d934 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a03cfc2fbbe4addb96c65fda605698e : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74e3a1003b7d489b8301cd2c697ef295 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/421377066b414fcbb37006693d9c51cd : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21765715d4d945a693ac91a140c6c471 : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":1},{"id":"b930fb831c54b208f8b7374d618045a29c56de7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 07:15:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a00fc731b14bd3ef953fe4cffdc74c8e441c0e13","revisions":{"a00fc731b14bd3ef953fe4cffdc74c8e441c0e13":{"kind":"REWORK","_number":1,"created":"2020-01-07 14:49:21.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/75/701375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-powervm","ref":"refs/changes/75/701375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-powervm refs/changes/75/701375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-powervm refs/changes/75/701375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-powervm refs/changes/75/701375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-powervm refs/changes/75/701375/1"}}},"commit":{"parents":[{"commit":"74215aceb83c6390c3b48a6ca0afacdf24d2b4dc","subject":"Enable support for both python3.6 and python3.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-powervm/commit/74215aceb83c6390c3b48a6ca0afacdf24d2b4dc"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-01-07 14:45:08.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-01-07 14:45:08.000000000","tz":60},"subject":"Adjust version constraint of hacking","message":"Adjust version constraint of hacking\n\nNeutron constrained the version of hacking \u003c 1.2.0, but since then the\nunit tests fails with:\n\nERROR: Double requirement given: hacking\u003c1.2.0,\u003e\u003d1.1.0 (from -ri https://opendev.org/openstack/neutron/raw/branch/master/test-requirements.txt (line 4)) (already in hacking\u003e\u003d1.1.0 (from -r [..]networking-powervm/test-requirements.txt (line 4)), name\u003d\u0027hacking\u0027)\n\nSo let\u0027s simply use the same constraint as set in neutron.\n\nChange-Id: Ic4d97db6c947264806e5ef9e35f21f764cbecf46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-powervm/commit/a00fc731b14bd3ef953fe4cffdc74c8e441c0e13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-powervm/commit/a00fc731b14bd3ef953fe4cffdc74c8e441c0e13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
