)]}'
{"id":"openstack%2Fnetworking-powervm~734392","triplet_id":"openstack%2Fnetworking-powervm~master~I7f07bf3049b28ec660550d2ea5d89e93e003cad7","project":"openstack/networking-powervm","branch":"master","topic":"drop_mock","hashtags":[],"change_id":"I7f07bf3049b28ec660550d2ea5d89e93e003cad7","subject":"Use unittest.mock instead of mock","status":"ABANDONED","created":"2020-06-08 23:35:23.000000000","updated":"2020-06-22 10:54:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fa4d18977feb3aa5a959fd4a9e32da14560ba792","_number":734392,"virtual_id_number":734392,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-06-22 07:06:25.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-10 09:55:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":-1,"date":"2020-06-11 10:55:25.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 09:55:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-22 07:06:25.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"c0def8bae8f218aa924b5d781004559bda6a3901","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-08 23:35:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e1e51dcd70dbfd9fd4686ab76f78624645d65bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 23:47:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d33d506bf9d43f998755f29073e44cf : SUCCESS in 4m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0bbf11d509c54cd6824411d7d6aced7e : FAILURE in 4m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1cf0beb4d1c409e85c1f94fdd9e206c : FAILURE in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a0f709233f44554a8d3c0e6d2602796 : FAILURE in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9c06b21ffb604454ad515a62f57eaa38 : FAILURE in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7579b1883f5346fcb9c3e576fe8af750 : FAILURE in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4755e8ac956344fbbab3e725bb3d08ca : FAILURE in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"5871dfcb088be782a77898683924da989cbca02d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-10 09:42:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2b8925dcdf26c95d9adf5edaf0c722b1ccbb1884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 09:55:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/093feb8e38794fc7b738a4d4f3f17a73 : SUCCESS in 4m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2e0f3c5d633a41028aac222443f5b228 : FAILURE in 4m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/466a910d9f3b464997294ea54f9eaadc : FAILURE in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c7620adb70d420bbc6f54ca172e2cad : FAILURE in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9151096250f8456bbdf364bab8a2d8ca : FAILURE in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2398d53e98554b26a7f98c1bb3ee6f65 : FAILURE in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d5bbb8bccc5473ca67269ec03da7dc4 : FAILURE in 4m 57s","accounts_in_message":[],"_revision_number":2},{"id":"bfbdcfc1be6b173b6b9246e249835d9bb217ede8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-11 10:55:25.000000000","message":"Patch Set 2: Workflow-1\n\nstill support py27","accounts_in_message":[],"_revision_number":2},{"id":"9cb5109f6a09353fb8a2dfb74aefcb0bf0ab4468","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-22 07:06:25.000000000","message":"Patch Set 2:\n\nduplicate of https://review.opendev.org/720959","accounts_in_message":[],"_revision_number":2},{"id":"c9a80116cb8cb963d5f89d281c4748d48b9fc6ec","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 10:54:36.000000000","message":"Abandoned\n\nhttps://review.opendev.org/#/c/720959/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d70ca8e40cbecd9f0cb2b48518741faff50bf2d6","revisions":{"e999ebe94128f9a652f46bc11eafd98d47fd2f7a":{"kind":"REWORK","_number":1,"created":"2020-06-08 23:35:23.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/92/734392/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-powervm","ref":"refs/changes/92/734392/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-powervm refs/changes/92/734392/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-powervm refs/changes/92/734392/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-powervm refs/changes/92/734392/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-powervm refs/changes/92/734392/1"}}},"commit":{"parents":[{"commit":"a00fc731b14bd3ef953fe4cffdc74c8e441c0e13","subject":"Adjust version constraint of hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-powervm/commit/a00fc731b14bd3ef953fe4cffdc74c8e441c0e13"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-08 23:35:17.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-08 23:35:17.000000000","tz":120},"subject":"Use unittest.mock instead of mock","message":"Use unittest.mock instead of mock\n\nThe mock third party library was needed for mock support in py2\nruntimes. Since we now only support py36 and later, we can use the\nstandard lib unittest.mock module instead.\n\nChange-Id: I7f07bf3049b28ec660550d2ea5d89e93e003cad7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-powervm/commit/e999ebe94128f9a652f46bc11eafd98d47fd2f7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-powervm/commit/e999ebe94128f9a652f46bc11eafd98d47fd2f7a"}]},"branch":"refs/heads/master"},"d70ca8e40cbecd9f0cb2b48518741faff50bf2d6":{"kind":"REWORK","_number":2,"created":"2020-06-10 09:42:00.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/92/734392/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-powervm","ref":"refs/changes/92/734392/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-powervm refs/changes/92/734392/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-powervm refs/changes/92/734392/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-powervm refs/changes/92/734392/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-powervm refs/changes/92/734392/2"}}},"commit":{"parents":[{"commit":"a00fc731b14bd3ef953fe4cffdc74c8e441c0e13","subject":"Adjust version constraint of hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-powervm/commit/a00fc731b14bd3ef953fe4cffdc74c8e441c0e13"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-08 23:35:17.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-10 09:41:58.000000000","tz":120},"subject":"Use unittest.mock instead of mock","message":"Use unittest.mock instead of mock\n\nThe mock third party library was needed for mock support in py2\nruntimes. Since we now only support py36 and later, we can use the\nstandard lib unittest.mock module instead.\n\nChange-Id: I7f07bf3049b28ec660550d2ea5d89e93e003cad7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-powervm/commit/d70ca8e40cbecd9f0cb2b48518741faff50bf2d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-powervm/commit/d70ca8e40cbecd9f0cb2b48518741faff50bf2d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
