)]}'
{"id":"openstack%2Fnetworking-sfc~338741","triplet_id":"openstack%2Fnetworking-sfc~master~I0887d7cc14f4496894df915dfe0ba1615e40c711","project":"openstack/networking-sfc","branch":"master","topic":"bug/1599037","hashtags":[],"change_id":"I0887d7cc14f4496894df915dfe0ba1615e40c711","subject":"Make port-aggregation more efficient","status":"ABANDONED","created":"2016-07-07 07:39:52.000000000","updated":"2016-10-24 17:04:48.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"94bc4fc8685188415f1a51e0b366348aa860e215","_number":338741,"virtual_id_number":338741,"owner":{"_account_id":18668,"name":"ding bo","email":"longddropt@163.com","username":"longddropt"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":-1,"date":"2016-09-12 18:23:00.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":11907,"name":"cathy","email":"cathy.h.zhang@huawei.com","username":"cathy"},{"_account_id":17540,"name":"xiaodongwang991481","email":"xiaodongwang991481@gmail.com","username":"xiaodongwang991481"},{"_account_id":11364,"name":"Fei Rao","email":"milo.frao@gmail.com","username":"milo.frao"},{"date":"2016-10-24 17:04:48.000000000","_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"_account_id":18668,"name":"ding bo","email":"longddropt@163.com","username":"longddropt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":17540,"name":"xiaodongwang991481","email":"xiaodongwang991481@gmail.com","username":"xiaodongwang991481"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11907,"name":"cathy","email":"cathy.h.zhang@huawei.com","username":"cathy"},{"value":1,"date":"2016-09-12 18:24:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17540,"name":"xiaodongwang991481","email":"xiaodongwang991481@gmail.com","username":"xiaodongwang991481"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11364,"name":"Fei Rao","email":"milo.frao@gmail.com","username":"milo.frao"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18668,"name":"ding bo","email":"longddropt@163.com","username":"longddropt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":11907,"name":"cathy","email":"cathy.h.zhang@huawei.com","username":"cathy"},{"_account_id":17540,"name":"xiaodongwang991481","email":"xiaodongwang991481@gmail.com","username":"xiaodongwang991481"},{"_account_id":11364,"name":"Fei Rao","email":"milo.frao@gmail.com","username":"milo.frao"},{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":18668,"name":"ding bo","email":"longddropt@163.com","username":"longddropt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":11907,"name":"cathy","email":"cathy.h.zhang@huawei.com","username":"cathy"},{"_account_id":17540,"name":"xiaodongwang991481","email":"xiaodongwang991481@gmail.com","username":"xiaodongwang991481"},{"_account_id":11364,"name":"Fei Rao","email":"milo.frao@gmail.com","username":"milo.frao"},{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"_account_id":18668,"name":"ding bo","email":"longddropt@163.com","username":"longddropt"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},{"_account_id":11364,"name":"Fei Rao","email":"milo.frao@gmail.com","username":"milo.frao"},{"_account_id":11907,"name":"cathy","email":"cathy.h.zhang@huawei.com","username":"cathy"},{"_account_id":17540,"name":"xiaodongwang991481","email":"xiaodongwang991481@gmail.com","username":"xiaodongwang991481"},{"_account_id":18668,"name":"ding bo","email":"longddropt@163.com","username":"longddropt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-07-07 07:42:30.000000000","updated_by":{"_account_id":11364,"name":"Fei Rao","email":"milo.frao@gmail.com","username":"milo.frao"},"reviewer":{"_account_id":11364,"name":"Fei Rao","email":"milo.frao@gmail.com","username":"milo.frao"},"state":"REVIEWER"},{"updated":"2016-07-07 07:43:17.000000000","updated_by":{"_account_id":11907,"name":"cathy","email":"cathy.h.zhang@huawei.com","username":"cathy"},"reviewer":{"_account_id":11907,"name":"cathy","email":"cathy.h.zhang@huawei.com","username":"cathy"},"state":"REVIEWER"},{"updated":"2016-09-12 18:23:00.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-09-12 18:24:35.000000000","updated_by":{"_account_id":17540,"name":"xiaodongwang991481","email":"xiaodongwang991481@gmail.com","username":"xiaodongwang991481"},"reviewer":{"_account_id":17540,"name":"xiaodongwang991481","email":"xiaodongwang991481@gmail.com","username":"xiaodongwang991481"},"state":"REVIEWER"},{"updated":"2016-10-24 17:04:48.000000000","updated_by":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"reviewer":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"state":"REVIEWER"}],"messages":[{"id":"a6de3d428146001ef19a6e88e6be43610ff225ea","author":{"_account_id":18668,"name":"ding bo","email":"longddropt@163.com","username":"longddropt"},"date":"2016-07-07 07:39:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4645707861c4c31d022f9898125054556f3e7e6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-07 08:10:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-sfc-docs http://docs-draft.openstack.org/41/338741/1/check/gate-networking-sfc-docs/74424c9//doc/build/html/ : SUCCESS in 4m 22s\n- gate-networking-sfc-pep8 http://logs.openstack.org/41/338741/1/check/gate-networking-sfc-pep8/1cc9164/ : FAILURE in 3m 03s\n- gate-networking-sfc-python27 http://logs.openstack.org/41/338741/1/check/gate-networking-sfc-python27/98d231d/ : SUCCESS in 4m 36s\n- gate-networking-sfc-python34 http://logs.openstack.org/41/338741/1/check/gate-networking-sfc-python34/5c2431c/ : SUCCESS in 10m 20s\n- gate-networking-sfc-python35-nv http://logs.openstack.org/41/338741/1/check/gate-networking-sfc-python35-nv/2875ec7/ : SUCCESS in 7m 09s (non-voting)\n- gate-tempest-dsvm-networking-sfc-nv http://logs.openstack.org/41/338741/1/check/gate-tempest-dsvm-networking-sfc-nv/0662105/ : SUCCESS in 28m 58s (non-voting)\n- gate-functional-dsvm-networking-sfc-nv http://logs.openstack.org/41/338741/1/check/gate-functional-dsvm-networking-sfc-nv/f14219c/ : FAILURE in 3m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"60f2451087664fe09ad9f2a77e30b1ac27fff6f5","author":{"_account_id":18668,"name":"ding bo","email":"longddropt@163.com","username":"longddropt"},"date":"2016-07-07 08:45:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6f8be4cf3ff7aabb7eaffdc6b3f77dc4ee7615a6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-07 09:15:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-sfc-docs http://docs-draft.openstack.org/41/338741/2/check/gate-networking-sfc-docs/78b2bdd//doc/build/html/ : SUCCESS in 6m 14s\n- gate-networking-sfc-pep8 http://logs.openstack.org/41/338741/2/check/gate-networking-sfc-pep8/7311bfb/ : FAILURE in 6m 05s\n- gate-networking-sfc-python27 http://logs.openstack.org/41/338741/2/check/gate-networking-sfc-python27/322ef8b/ : SUCCESS in 7m 27s\n- gate-networking-sfc-python34 http://logs.openstack.org/41/338741/2/check/gate-networking-sfc-python34/0ada866/ : SUCCESS in 12m 26s\n- gate-networking-sfc-python35-nv http://logs.openstack.org/41/338741/2/check/gate-networking-sfc-python35-nv/0d9cfdd/ : SUCCESS in 5m 37s (non-voting)\n- gate-tempest-dsvm-networking-sfc-nv http://logs.openstack.org/41/338741/2/check/gate-tempest-dsvm-networking-sfc-nv/dbfbff0/ : SUCCESS in 26m 25s (non-voting)\n- gate-functional-dsvm-networking-sfc-nv http://logs.openstack.org/41/338741/2/check/gate-functional-dsvm-networking-sfc-nv/383e95a/ : FAILURE in 9m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a59bce0dcd1666983eb967cadc61ef69121ed81b","author":{"_account_id":18668,"name":"ding bo","email":"longddropt@163.com","username":"longddropt"},"date":"2016-07-07 11:52:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"46987a81badddd4c92f4c20d4e1d5fc56a11244e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-07 12:26:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-sfc-docs http://docs-draft.openstack.org/41/338741/3/check/gate-networking-sfc-docs/bf6359a//doc/build/html/ : SUCCESS in 4m 56s\n- gate-networking-sfc-pep8 http://logs.openstack.org/41/338741/3/check/gate-networking-sfc-pep8/2655353/ : SUCCESS in 8m 05s\n- gate-networking-sfc-python27 http://logs.openstack.org/41/338741/3/check/gate-networking-sfc-python27/341f492/ : SUCCESS in 6m 21s\n- gate-networking-sfc-python34 http://logs.openstack.org/41/338741/3/check/gate-networking-sfc-python34/6f7ce86/ : SUCCESS in 13m 19s\n- gate-networking-sfc-python35-nv http://logs.openstack.org/41/338741/3/check/gate-networking-sfc-python35-nv/60811cf/ : SUCCESS in 5m 29s (non-voting)\n- gate-tempest-dsvm-networking-sfc-nv http://logs.openstack.org/41/338741/3/check/gate-tempest-dsvm-networking-sfc-nv/e43f786/ : SUCCESS in 28m 01s (non-voting)\n- gate-functional-dsvm-networking-sfc-nv http://logs.openstack.org/41/338741/3/check/gate-functional-dsvm-networking-sfc-nv/96cbdd6/ : FAILURE in 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f1d1fd445821e961a04f327f9d80833c83a7c38d","author":{"_account_id":17540,"name":"xiaodongwang991481","email":"xiaodongwang991481@gmail.com","username":"xiaodongwang991481"},"date":"2016-07-07 16:45:20.000000000","message":"Patch Set 3:\n\nAre you sure it will generate correct results in each corner case like ports\u003d[0, 1, 2] or [1,2,3,4]. Second I do not think it change the BigO.","accounts_in_message":[],"_revision_number":3},{"id":"db5194cbd23475a8b1465f78d715ec8744bdcc5e","author":{"_account_id":17540,"name":"xiaodongwang991481","email":"xiaodongwang991481@gmail.com","username":"xiaodongwang991481"},"date":"2016-07-07 17:03:12.000000000","message":"Patch Set 3:\n\nlike the case 0b11001100 to 0b11011001","accounts_in_message":[],"_revision_number":3},{"id":"9a77441905ce323ce306632e2984ca5c36914e60","author":{"_account_id":18668,"name":"ding bo","email":"longddropt@163.com","username":"longddropt"},"date":"2016-07-08 01:04:22.000000000","message":"Patch Set 3:\n\n(3 comments)\n\nYes, corner case handled at line 51. Also I explain why this method is O(n) in my comments too","accounts_in_message":[],"_revision_number":3},{"id":"914499368029cf4b2a2149f66f4270b2d7c57234","author":{"_account_id":17540,"name":"xiaodongwang991481","email":"xiaodongwang991481@gmail.com","username":"xiaodongwang991481"},"date":"2016-07-08 01:42:36.000000000","message":"Patch Set 3:\n\nplease add testcases to cover the example I give.","accounts_in_message":[],"_revision_number":3},{"id":"7fc6727665c8c48bb94808c3e07d7b6db53bab27","author":{"_account_id":18668,"name":"ding bo","email":"longddropt@163.com","username":"longddropt"},"date":"2016-07-08 01:53:29.000000000","message":"Patch Set 3:\n\nresult of original code:\n[\u00270xcc/0xfffc\u0027, \u00270xd0/0xfff8\u0027, \u00270xd8/0xfffe\u0027]\n\nresult of this method:\n[\u00270xcc/0xfffc\u0027, \u00270xd8/0xfffe\u0027, \u00270xd0/0xfff8\u0027]","accounts_in_message":[],"_revision_number":3},{"id":"5cc85112068ff40aeead96cc7978019a8b16e46a","author":{"_account_id":18668,"name":"ding bo","email":"longddropt@163.com","username":"longddropt"},"date":"2016-07-08 02:43:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0cc10b0deb10a464eefe47b5a8d393b67a09009d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-08 03:13:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-sfc-docs http://docs-draft.openstack.org/41/338741/4/check/gate-networking-sfc-docs/a566e26//doc/build/html/ : SUCCESS in 2m 54s\n- gate-networking-sfc-pep8 http://logs.openstack.org/41/338741/4/check/gate-networking-sfc-pep8/80f7284/ : SUCCESS in 5m 41s\n- gate-networking-sfc-python27 http://logs.openstack.org/41/338741/4/check/gate-networking-sfc-python27/c2568b8/ : SUCCESS in 7m 33s\n- gate-networking-sfc-python34 http://logs.openstack.org/41/338741/4/check/gate-networking-sfc-python34/99a13b2/ : SUCCESS in 14m 18s\n- gate-networking-sfc-python35-nv http://logs.openstack.org/41/338741/4/check/gate-networking-sfc-python35-nv/bf734c3/ : SUCCESS in 7m 43s (non-voting)\n- gate-tempest-dsvm-networking-sfc-nv http://logs.openstack.org/41/338741/4/check/gate-tempest-dsvm-networking-sfc-nv/399e2e5/ : SUCCESS in 26m 14s (non-voting)\n- gate-functional-dsvm-networking-sfc-nv http://logs.openstack.org/41/338741/4/check/gate-functional-dsvm-networking-sfc-nv/e5cac90/ : FAILURE in 8m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"80bf273c30e2e137676f565400cd52e9984dd706","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-12 18:23:00.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"05d2330a233a852ff69284dbd61f3303097eec28","author":{"_account_id":17540,"name":"xiaodongwang991481","email":"xiaodongwang991481@gmail.com","username":"xiaodongwang991481"},"date":"2016-09-12 18:24:35.000000000","message":"Patch Set 4: Code-Review+1\n\nI am neutral for this change. If others think it is OK, I am OK.","accounts_in_message":[],"_revision_number":4},{"id":"2188c555ea5dd087546d97c88c1a046dc6a4ccda","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2016-10-24 17:04:48.000000000","message":"Patch Set 4:\n\nThis review is \u003e 4 weeks without comment, and failed Jenkins the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":4},{"id":"ad8d7ba7bbd166f3822b9751a62cbf3126c27dd0","author":{"_account_id":748,"name":"Armando Migliaccio","email":"armamig@gmail.com","username":"armando-migliaccio"},"date":"2016-10-24 17:04:48.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Jenkins the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"86d081986044c63ea996b3815647f23c30807d28","revisions":{"9c383023c8453f10173247ee91e293812cbb6bb5":{"kind":"REWORK","_number":1,"created":"2016-07-07 07:39:52.000000000","uploader":{"_account_id":18668,"name":"ding bo","email":"longddropt@163.com","username":"longddropt"},"ref":"refs/changes/41/338741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/41/338741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/41/338741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/41/338741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/41/338741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/41/338741/1"}}},"commit":{"parents":[{"commit":"d07ac5ee16913ee833c8e9e81b9a74cd78ee0b3e","subject":"Fix tox unit test issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/d07ac5ee16913ee833c8e9e81b9a74cd78ee0b3e"}]}],"author":{"name":"longddropt","email":"longddropt@163.com","date":"2016-07-07 07:32:37.000000000","tz":-240},"committer":{"name":"longddropt","email":"longddropt@163.com","date":"2016-07-07 07:32:37.000000000","tz":-240},"subject":"Make port-aggregation more efficient","message":"Make port-aggregation more efficient\n\nCurrently, get_port_mask takes O(n^2) to calculate port\naggregation list. This patch reduce the time complexity to O(n)\nby first determine the boundary, and aggregate lower part and\nupper part respectively. n refers to the bit number of port in the\nabove context.\n\nChange-Id: I0887d7cc14f4496894df915dfe0ba1615e40c711\nCloses-Bug: 1599037\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/9c383023c8453f10173247ee91e293812cbb6bb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/9c383023c8453f10173247ee91e293812cbb6bb5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d07ac5ee16913ee833c8e9e81b9a74cd78ee0b3e","is_merged_in_target_branch":true,"change_id":"I7e6df407c7c083b4d98b631615161bcda82059a3","change_number":337614,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"18b9785c7ec4ef0807b3ac92214c92c6429f4789":{"kind":"REWORK","_number":2,"created":"2016-07-07 08:45:54.000000000","uploader":{"_account_id":18668,"name":"ding bo","email":"longddropt@163.com","username":"longddropt"},"ref":"refs/changes/41/338741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/41/338741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/41/338741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/41/338741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/41/338741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/41/338741/2"}}},"commit":{"parents":[{"commit":"d07ac5ee16913ee833c8e9e81b9a74cd78ee0b3e","subject":"Fix tox unit test issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/d07ac5ee16913ee833c8e9e81b9a74cd78ee0b3e"}]}],"author":{"name":"longddropt","email":"longddropt@163.com","date":"2016-07-07 07:32:37.000000000","tz":-240},"committer":{"name":"longddropt","email":"longddropt@163.com","date":"2016-07-07 08:45:16.000000000","tz":-240},"subject":"Make port-aggregation more efficient","message":"Make port-aggregation more efficient\n\nCurrently, get_port_mask takes O(n^2) to calculate port\naggregation list. This patch reduce the time complexity to O(n)\nby first determine the boundary, and aggregate lower part and\nupper part respectively. n refers to the bit number of port in the\nabove context.\n\nChange-Id: I0887d7cc14f4496894df915dfe0ba1615e40c711\nCloses-Bug: 1599037\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/18b9785c7ec4ef0807b3ac92214c92c6429f4789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/18b9785c7ec4ef0807b3ac92214c92c6429f4789"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d07ac5ee16913ee833c8e9e81b9a74cd78ee0b3e","is_merged_in_target_branch":true,"change_id":"I7e6df407c7c083b4d98b631615161bcda82059a3","change_number":337614,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"21b63d76f1a60d86874c1c610ded654196ee77e7":{"kind":"REWORK","_number":3,"created":"2016-07-07 11:52:13.000000000","uploader":{"_account_id":18668,"name":"ding bo","email":"longddropt@163.com","username":"longddropt"},"ref":"refs/changes/41/338741/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/41/338741/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/41/338741/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/41/338741/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/41/338741/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/41/338741/3"}}},"commit":{"parents":[{"commit":"d07ac5ee16913ee833c8e9e81b9a74cd78ee0b3e","subject":"Fix tox unit test issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/d07ac5ee16913ee833c8e9e81b9a74cd78ee0b3e"}]}],"author":{"name":"longddropt","email":"longddropt@163.com","date":"2016-07-07 07:32:37.000000000","tz":-240},"committer":{"name":"longddropt","email":"longddropt@163.com","date":"2016-07-07 11:47:47.000000000","tz":-240},"subject":"Make port-aggregation more efficient","message":"Make port-aggregation more efficient\n\nCurrently, get_port_mask takes O(n^2) to calculate port\naggregation list. This patch reduce the time complexity to O(n)\nby first determine the boundary, and aggregate lower part and\nupper part respectively. n refers to the bit number of port in the\nabove context.\n\nChange-Id: I0887d7cc14f4496894df915dfe0ba1615e40c711\nCloses-Bug: 1599037\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/21b63d76f1a60d86874c1c610ded654196ee77e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/21b63d76f1a60d86874c1c610ded654196ee77e7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d07ac5ee16913ee833c8e9e81b9a74cd78ee0b3e","is_merged_in_target_branch":true,"change_id":"I7e6df407c7c083b4d98b631615161bcda82059a3","change_number":337614,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"86d081986044c63ea996b3815647f23c30807d28":{"kind":"REWORK","_number":4,"created":"2016-07-08 02:43:52.000000000","uploader":{"_account_id":18668,"name":"ding bo","email":"longddropt@163.com","username":"longddropt"},"ref":"refs/changes/41/338741/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/41/338741/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/41/338741/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/41/338741/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/41/338741/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/41/338741/4"}}},"commit":{"parents":[{"commit":"d07ac5ee16913ee833c8e9e81b9a74cd78ee0b3e","subject":"Fix tox unit test issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/d07ac5ee16913ee833c8e9e81b9a74cd78ee0b3e"}]}],"author":{"name":"longddropt","email":"longddropt@163.com","date":"2016-07-07 07:32:37.000000000","tz":-240},"committer":{"name":"longddropt","email":"longddropt@163.com","date":"2016-07-08 02:43:29.000000000","tz":-240},"subject":"Make port-aggregation more efficient","message":"Make port-aggregation more efficient\n\nCurrently, get_port_mask takes O(n^2) to calculate port\naggregation list. This patch reduce the time complexity to O(n)\nby first determine the boundary, and aggregate lower part and\nupper part respectively. n refers to the bit number of port in the\nabove context.\n\nChange-Id: I0887d7cc14f4496894df915dfe0ba1615e40c711\nCloses-Bug: 1599037\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/86d081986044c63ea996b3815647f23c30807d28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/86d081986044c63ea996b3815647f23c30807d28"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d07ac5ee16913ee833c8e9e81b9a74cd78ee0b3e","is_merged_in_target_branch":true,"change_id":"I7e6df407c7c083b4d98b631615161bcda82059a3","change_number":337614,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
