)]}'
{"id":"openstack%2Fnetworking-sfc~617554","triplet_id":"openstack%2Fnetworking-sfc~master~Idf4e1c30fef5d5c4928dcfd5cedfe333293b07e8","project":"openstack/networking-sfc","branch":"master","hashtags":[],"change_id":"Idf4e1c30fef5d5c4928dcfd5cedfe333293b07e8","subject":"Remove those copy words occured twice times in api.rst","status":"MERGED","created":"2018-11-13 10:11:44.000000000","updated":"2018-12-12 10:03:11.000000000","submitted":"2018-12-12 10:03:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"617554-1544608991068-5f063c75","meta_rev_id":"4e8049e8dc9a6d97331fd4ec2913794daec84079","_number":617554,"virtual_id_number":617554,"owner":{"_account_id":27366,"name":"guotao","email":"guotao.bj@inspur.com","username":"taoguo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2018-12-12 10:03:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"recommended":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":1,"date":"2018-12-09 14:57:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2018-12-04 19:42:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":2,"date":"2018-12-12 07:57:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":1,"date":"2018-12-12 07:57:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-12-04 19:42:40.000000000","updated_by":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"reviewer":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"state":"REVIEWER"},{"updated":"2018-12-09 14:57:03.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2018-12-12 07:57:50.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2018-12-12 10:03:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5dffda5ba24a42d041ebf974fbdc72ff285a0f86","author":{"_account_id":27366,"name":"guotao","email":"guotao.bj@inspur.com","username":"taoguo"},"date":"2018-11-13 10:11:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e58ef05daa0bc38bc9df92ccb3c6a453df5edf0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-13 12:05:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/54/617554/1/check/openstack-tox-pep8/f79e4b9/ : SUCCESS in 6m 01s\n- build-openstack-sphinx-docs http://logs.openstack.org/54/617554/1/check/build-openstack-sphinx-docs/13368e0/html/ : SUCCESS in 6m 41s\n- networking-sfc-functional-dsvm http://logs.openstack.org/54/617554/1/check/networking-sfc-functional-dsvm/70745a6/ : SUCCESS in 14m 14s\n- networking-sfc-tempest-dsvm http://logs.openstack.org/54/617554/1/check/networking-sfc-tempest-dsvm/cdc0d8f/ : SUCCESS in 1h 44m 51s\n- networking-sfc-tempest-dsvm-multinode http://logs.openstack.org/54/617554/1/check/networking-sfc-tempest-dsvm-multinode/6edc42c/ : SUCCESS in 1h 50m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f76b2c202675c60e7b41e424a6e3d7d3abec5b8","author":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"date":"2018-12-04 19:42:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"32a70755602584a61c378eaa769f4b54405f04cb","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2018-12-09 14:57:03.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)\n\nI am okay if the author does not care his typos or grammars.","accounts_in_message":[],"_revision_number":1},{"id":"82f2959c9b0c309c45cad581a8cf71e0fb95998a","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2018-12-12 07:57:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d2bcb9a67def5c0565212749c452c1e446486f26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-12 07:58:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6dbda92c77cffd15bcb4e0bfa397b5889381498c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-12 10:03:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/54/617554/1/gate/openstack-tox-pep8/b7cfe95/ : SUCCESS in 7m 53s\n- openstack-tox-docs http://logs.openstack.org/54/617554/1/gate/openstack-tox-docs/aa25888/html/ : SUCCESS in 5m 14s\n- networking-sfc-functional-dsvm http://logs.openstack.org/54/617554/1/gate/networking-sfc-functional-dsvm/810e8bd/ : SUCCESS in 15m 11s\n- networking-sfc-tempest-dsvm http://logs.openstack.org/54/617554/1/gate/networking-sfc-tempest-dsvm/0d5085a/ : SUCCESS in 1h 46m 47s","accounts_in_message":[],"_revision_number":1},{"id":"4e8049e8dc9a6d97331fd4ec2913794daec84079","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-12 10:03:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"645946fe7efc5b8da32afa2db329cd7f8d2f152c","revisions":{"645946fe7efc5b8da32afa2db329cd7f8d2f152c":{"kind":"REWORK","_number":1,"created":"2018-11-13 10:11:44.000000000","uploader":{"_account_id":27366,"name":"guotao","email":"guotao.bj@inspur.com","username":"taoguo"},"ref":"refs/changes/54/617554/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/54/617554/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/54/617554/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/54/617554/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/54/617554/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/54/617554/1"}}},"commit":{"parents":[{"commit":"369d119549ab4f4af4191d22da5b69a32eef8863","subject":"Merge \"import zuul job settings from project-config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/369d119549ab4f4af4191d22da5b69a32eef8863"}]}],"author":{"name":"guotao.bj","email":"guotao.bj@inspur.com","date":"2018-11-13 10:05:39.000000000","tz":480},"committer":{"name":"guotao.bj","email":"guotao.bj@inspur.com","date":"2018-11-13 10:11:01.000000000","tz":480},"subject":"Remove those copy words occured twice times in api.rst","message":"Remove those copy words occured twice times in api.rst\n\nChange-Id: Idf4e1c30fef5d5c4928dcfd5cedfe333293b07e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/645946fe7efc5b8da32afa2db329cd7f8d2f152c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/645946fe7efc5b8da32afa2db329cd7f8d2f152c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"369d119549ab4f4af4191d22da5b69a32eef8863","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
