)]}'
{"id":"openstack%2Fnetworking-sfc~653747","triplet_id":"openstack%2Fnetworking-sfc~master~Ib30382f6af434edde4b372e6af015e538768ddbd","project":"openstack/networking-sfc","branch":"master","topic":"networking-sfc-tests-move","hashtags":[],"change_id":"Ib30382f6af434edde4b372e6af015e538768ddbd","subject":"Complete move of networking-sfc tempest tests to tempest plugin","status":"MERGED","created":"2019-04-18 14:39:22.000000000","updated":"2019-06-06 16:18:48.000000000","submitted":"2019-06-06 16:18:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653747-1559837928727-e9fba48b","meta_rev_id":"1cfb42f007d4bc4d92ce4f9151d050fe508b16ef","_number":653747,"virtual_id_number":653747,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":2,"date":"2019-06-06 16:18:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-06-04 16:50:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-06-04 16:50:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":1,"date":"2019-06-05 07:49:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-18 22:46:20.000000000","updated_by":{"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"},"reviewer":{"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"},"state":"REVIEWER"},{"updated":"2019-05-21 09:16:02.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-05-28 23:20:24.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-06-04 09:54:41.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2019-06-04 09:54:52.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-06-05 07:49:21.000000000","updated_by":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"reviewer":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"state":"REVIEWER"},{"updated":"2019-06-06 16:18:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ccba1cf33a2543934e8e334d4731c0d8a295efb","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-04-18 14:39:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74395ab9224a69165024efc4f8f9676e2989ebe5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 14:39:37.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron-tempest-plugin on branch master.  The error\nwas:\n\n  Job neutron-tempest-plugin-sfc in openstack/neutron-tempest-plugin is not permitted to shadow job neutron-tempest-plugin-sfc in openstack/networking-sfc\n\nThe error appears in the following job stanza:\n\n  job:\n      name: neutron-tempest-plugin-sfc\n      parent: neutron-tempest-plugin\n      timeout: 10800\n      required-projects:\n        - openstack-infra/devstack-gate\n        - openstack/networking-sfc\n        - openstack/neutron\n        - openstack/neutron-tempest-plugin\n        - openstack/tempest\n      vars:\n        tempest_test_regex: ^neutron_tempest_plugin\\.sfc\n        tox_envlist: all-plugin\n        devstack_plugins:\n          networking-sfc: https://git.openstack.org/openstack/networking-sfc\n          neutron-tempest-plugin: https://git.openstack.org/openstack/neutron-tempest-plugin\n        network_api_extensions_sfc:\n          - flow_classifier\n          - sfc\n        devstack_localrc:\n          NETWORK_API_EXTENSIONS: \"{{ (network_api_extensions_common + network_api_extensions_sfc) | join(\u0027,\u0027) }}\"\n      files:\n        - ^neutron_tempest_plugin/sfc/.*$\n\n  in \"openstack/neutron-tempest-plugin/.zuul.yaml@master\", line 693, column 3","accounts_in_message":[],"_revision_number":1},{"id":"ad4f6abc1c2dd409d2f8aff3beaf5d37f3b3d981","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-04-18 14:40:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dbdfb7cc3d51fff2836709a414c2a65f1108c87b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-04-18 14:41:23.000000000","message":"Patch Set 2: Workflow-1\n\nInitial WIP, work focused in parent review for now","accounts_in_message":[],"_revision_number":2},{"id":"1c78f69e94ebd4ac098ea8de641cc41e7a5a986c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-04-18 14:42:01.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e008f2c45cbc31b7d08c1c27f7c6f31ed7876d81","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-04-18 14:42:13.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"73e2b4e7e38f75d74d4d416e77cb067342a7b4c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 15:03:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/47/653747/3/check/openstack-tox-lower-constraints/a509db2/ : FAILURE in 4m 17s\n- openstack-tox-pep8 http://logs.openstack.org/47/653747/3/check/openstack-tox-pep8/6571092/ : FAILURE in 6m 43s\n- openstack-tox-py27 http://logs.openstack.org/47/653747/3/check/openstack-tox-py27/45bb927/ : FAILURE in 5m 58s\n- openstack-tox-docs http://logs.openstack.org/47/653747/3/check/openstack-tox-docs/d2f98d8/html/ : SUCCESS in 5m 20s\n- requirements-check http://logs.openstack.org/47/653747/3/check/requirements-check/1965f6f/ : FAILURE in 3m 43s\n- openstack-tox-py35 http://logs.openstack.org/47/653747/3/check/openstack-tox-py35/e9fe517/ : FAILURE in 6m 29s\n- openstack-tox-py36 http://logs.openstack.org/47/653747/3/check/openstack-tox-py36/9b356c9/ : FAILURE in 6m 11s\n- openstack-tox-py37 http://logs.openstack.org/47/653747/3/check/openstack-tox-py37/46a52f2/ : FAILURE in 11m 50s\n- networking-sfc-functional-dsvm http://logs.openstack.org/47/653747/3/check/networking-sfc-functional-dsvm/ad9e8de/ : SUCCESS in 15m 08s","accounts_in_message":[],"_revision_number":3},{"id":"515acb5bcd15c5c446560e22c906aa19ff899e45","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-04-19 13:09:42.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"fe08664d53d430b69aeb7aec7ee7ddcf7de6a6fc","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-04-19 13:09:54.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"f1e4ef3907c5b8b67a30a406efc0910fb8241bd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 14:23:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/47/653747/4/check/openstack-tox-lower-constraints/a47c45d/ : SUCCESS in 10m 32s\n- openstack-tox-pep8 http://logs.openstack.org/47/653747/4/check/openstack-tox-pep8/ff44949/ : SUCCESS in 9m 24s\n- openstack-tox-py27 http://logs.openstack.org/47/653747/4/check/openstack-tox-py27/f7f8a02/ : SUCCESS in 8m 56s\n- openstack-tox-docs http://logs.openstack.org/47/653747/4/check/openstack-tox-docs/6f38f42/html/ : SUCCESS in 5m 29s\n- requirements-check http://logs.openstack.org/47/653747/4/check/requirements-check/85cf251/ : FAILURE in 5m 45s\n- openstack-tox-py35 http://logs.openstack.org/47/653747/4/check/openstack-tox-py35/384a9a7/ : SUCCESS in 10m 35s\n- openstack-tox-py36 http://logs.openstack.org/47/653747/4/check/openstack-tox-py36/69705c3/ : SUCCESS in 9m 56s\n- openstack-tox-py37 http://logs.openstack.org/47/653747/4/check/openstack-tox-py37/6369fe8/ : SUCCESS in 9m 56s\n- networking-sfc-functional-dsvm http://logs.openstack.org/47/653747/4/check/networking-sfc-functional-dsvm/35a616a/ : SUCCESS in 18m 46s","accounts_in_message":[],"_revision_number":4},{"id":"fe2d2ff4987b89ada4c600968afbaace20cdef68","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-04-23 08:14:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4c813fb44a4976c66a5990f0dbce4d9ab0f595b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 08:14:34.000000000","message":"Patch Set 5: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-sfc on branch master.  The error was:\n\n  Job neutron-tempest-plugin-sfc not defined\n\nThe error appears in the following job stanza:\n\n  job:\n      name: neutron-tempest-plugin-sfc-periodic\n      parent: neutron-tempest-plugin-sfc\n      branches: master\n\n  in \"openstack/networking-sfc/zuul.d/jobs.yaml@master\", line 17, column 3","accounts_in_message":[],"_revision_number":5},{"id":"74521ad6057e60bcdcedea46054040851cb485f4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-04-23 08:55:38.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"6f1e6478424409d1a81f7d03a7acbbc32cacdb2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 09:17:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/47/653747/6/check/openstack-tox-lower-constraints/21bd509/ : SUCCESS in 9m 35s\n- openstack-tox-pep8 http://logs.openstack.org/47/653747/6/check/openstack-tox-pep8/7016b39/ : SUCCESS in 6m 39s\n- openstack-tox-py27 http://logs.openstack.org/47/653747/6/check/openstack-tox-py27/7d26a50/ : SUCCESS in 9m 19s\n- openstack-tox-docs http://logs.openstack.org/47/653747/6/check/openstack-tox-docs/242a9cb/html/ : SUCCESS in 5m 48s\n- requirements-check http://logs.openstack.org/47/653747/6/check/requirements-check/6c5f03c/ : FAILURE in 3m 45s\n- openstack-tox-py36 http://logs.openstack.org/47/653747/6/check/openstack-tox-py36/3474704/ : SUCCESS in 9m 01s\n- openstack-tox-py37 http://logs.openstack.org/47/653747/6/check/openstack-tox-py37/6386eae/ : SUCCESS in 9m 10s\n- networking-sfc-functional-dsvm http://logs.openstack.org/47/653747/6/check/networking-sfc-functional-dsvm/5be1c6e/ : SUCCESS in 14m 09s","accounts_in_message":[],"_revision_number":6},{"id":"d7566c7efe7a2e88dc153e15a80134b752b1d9dc","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-04-23 09:49:39.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"10def67ffbcd8b4000ba35d26edbbdfef6f4058a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 10:17:29.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/47/653747/7/check/test-release-openstack/5af99c4/ : SUCCESS in 3m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/653747/7/check/openstack-tox-lower-constraints/073195e/ : SUCCESS in 9m 38s\n- openstack-tox-pep8 http://logs.openstack.org/47/653747/7/check/openstack-tox-pep8/51c9177/ : SUCCESS in 7m 01s\n- openstack-tox-py27 http://logs.openstack.org/47/653747/7/check/openstack-tox-py27/d2741e0/ : SUCCESS in 12m 51s\n- openstack-tox-docs http://logs.openstack.org/47/653747/7/check/openstack-tox-docs/4a4a776/html/ : SUCCESS in 6m 17s\n- requirements-check http://logs.openstack.org/47/653747/7/check/requirements-check/927fe04/ : SUCCESS in 6m 13s\n- openstack-tox-py36 http://logs.openstack.org/47/653747/7/check/openstack-tox-py36/338c961/ : SUCCESS in 11m 01s\n- openstack-tox-py37 http://logs.openstack.org/47/653747/7/check/openstack-tox-py37/021c629/ : SUCCESS in 10m 37s\n- build-openstack-releasenotes http://logs.openstack.org/47/653747/7/check/build-openstack-releasenotes/693932c/html/ : SUCCESS in 6m 10s\n- networking-sfc-functional-dsvm http://logs.openstack.org/47/653747/7/check/networking-sfc-functional-dsvm/6c92539/ : SUCCESS in 18m 45s","accounts_in_message":[],"_revision_number":7},{"id":"1504df763ae6e3412e83ef34a85fc3b1cc699d90","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-04-23 11:53:21.000000000","message":"Patch Set 7: Workflow-1\n\nI do not see the tempest jobs run","accounts_in_message":[],"_revision_number":7},{"id":"1403c5dcf23ade1531183f83fc6ae1699b9f2da7","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-04-26 14:53:44.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"107d3c6c636831cd80aee87bc75da81a97423c75","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-04-26 14:54:48.000000000","message":"Patch Set 7:\n\nOK I think I see what is missing (just after hitting recheck...): relevant files defined in usptream job are not the same as in this repo","accounts_in_message":[],"_revision_number":7},{"id":"70fe2086a47022f057f216ea0f5ca395672df25c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 15:19:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/47/653747/7/check/test-release-openstack/09923a1/ : SUCCESS in 3m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/653747/7/check/openstack-tox-lower-constraints/29f726d/ : SUCCESS in 8m 23s\n- openstack-tox-pep8 http://logs.openstack.org/47/653747/7/check/openstack-tox-pep8/85f4104/ : SUCCESS in 6m 06s\n- openstack-tox-py27 http://logs.openstack.org/47/653747/7/check/openstack-tox-py27/1442772/ : SUCCESS in 8m 32s\n- openstack-tox-docs http://logs.openstack.org/47/653747/7/check/openstack-tox-docs/bc117f3/html/ : SUCCESS in 4m 52s\n- requirements-check http://logs.openstack.org/47/653747/7/check/requirements-check/37552bf/ : SUCCESS in 3m 26s\n- openstack-tox-py36 http://logs.openstack.org/47/653747/7/check/openstack-tox-py36/33105a7/ : SUCCESS in 8m 33s\n- openstack-tox-py37 http://logs.openstack.org/47/653747/7/check/openstack-tox-py37/46f171c/ : SUCCESS in 8m 25s\n- build-openstack-releasenotes http://logs.openstack.org/47/653747/7/check/build-openstack-releasenotes/6c7198d/html/ : SUCCESS in 4m 31s\n- networking-sfc-functional-dsvm http://logs.openstack.org/47/653747/7/check/networking-sfc-functional-dsvm/9964b72/ : SUCCESS in 18m 54s","accounts_in_message":[],"_revision_number":7},{"id":"bbcd8395ea9f4013ca2eb3899642d63e241d8251","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-01 00:10:19.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ee663ab57659e56c3850eedd38a18ff8ec682821","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-01 00:21:10.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"16d768c1b15a513bf428023681e40e9d40230718","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-01 00:24:52.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"8bb3315dabfeaa702459414b370c6245d9707054","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 02:16:53.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/47/653747/10/check/test-release-openstack/7cd8739/ : SUCCESS in 4m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/653747/10/check/openstack-tox-lower-constraints/a8e05a6/ : SUCCESS in 9m 07s\n- openstack-tox-pep8 http://logs.openstack.org/47/653747/10/check/openstack-tox-pep8/fdfe8ce/ : SUCCESS in 7m 31s\n- openstack-tox-py27 http://logs.openstack.org/47/653747/10/check/openstack-tox-py27/c93930d/ : SUCCESS in 9m 13s\n- openstack-tox-docs http://logs.openstack.org/47/653747/10/check/openstack-tox-docs/536bd27/html/ : SUCCESS in 5m 55s\n- requirements-check http://logs.openstack.org/47/653747/10/check/requirements-check/7eb702d/ : SUCCESS in 4m 13s\n- openstack-tox-py36 http://logs.openstack.org/47/653747/10/check/openstack-tox-py36/b69dcf1/ : SUCCESS in 8m 15s\n- openstack-tox-py37 http://logs.openstack.org/47/653747/10/check/openstack-tox-py37/2ec83e2/ : SUCCESS in 11m 34s\n- build-openstack-releasenotes http://logs.openstack.org/47/653747/10/check/build-openstack-releasenotes/6f307bd/html/ : SUCCESS in 6m 29s\n- networking-sfc-functional-dsvm http://logs.openstack.org/47/653747/10/check/networking-sfc-functional-dsvm/02327ed/ : SUCCESS in 22m 15s\n- networking-sfc-tempest http://logs.openstack.org/47/653747/10/check/networking-sfc-tempest/2c1a4ae/ : SUCCESS in 1h 46m 55s\n- networking-sfc-tempest-multinode http://logs.openstack.org/47/653747/10/check/networking-sfc-tempest-multinode/817ee71/ : FAILURE in 1h 19m 38s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"18776ed3a9c661f74c85f5f243ba47ce95039c67","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-03 17:13:40.000000000","message":"Patch Set 10:\n\nrecheck multinode nova error","accounts_in_message":[],"_revision_number":10},{"id":"93b5f35e2c14b5dc5f381e70a9a6dac212148653","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 19:07:55.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/47/653747/10/check/test-release-openstack/9151a28/ : SUCCESS in 3m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/653747/10/check/openstack-tox-lower-constraints/8876585/ : SUCCESS in 11m 58s\n- openstack-tox-pep8 http://logs.openstack.org/47/653747/10/check/openstack-tox-pep8/48196ef/ : SUCCESS in 6m 32s\n- openstack-tox-py27 http://logs.openstack.org/47/653747/10/check/openstack-tox-py27/b78d2c6/ : SUCCESS in 9m 44s\n- openstack-tox-docs http://logs.openstack.org/47/653747/10/check/openstack-tox-docs/8a0d793/html/ : SUCCESS in 5m 44s\n- requirements-check http://logs.openstack.org/47/653747/10/check/requirements-check/e78ffa7/ : SUCCESS in 4m 07s\n- openstack-tox-py36 http://logs.openstack.org/47/653747/10/check/openstack-tox-py36/7a1ceef/ : SUCCESS in 9m 11s\n- openstack-tox-py37 http://logs.openstack.org/47/653747/10/check/openstack-tox-py37/36e090d/ : SUCCESS in 10m 33s\n- build-openstack-releasenotes http://logs.openstack.org/47/653747/10/check/build-openstack-releasenotes/3e85a6b/html/ : SUCCESS in 4m 42s\n- networking-sfc-functional-dsvm http://logs.openstack.org/47/653747/10/check/networking-sfc-functional-dsvm/30c24b4/ : SUCCESS in 15m 56s\n- networking-sfc-tempest http://logs.openstack.org/47/653747/10/check/networking-sfc-tempest/59ad848/ : SUCCESS in 1h 48m 57s\n- networking-sfc-tempest-multinode http://logs.openstack.org/47/653747/10/check/networking-sfc-tempest-multinode/fa719f0/ : FAILURE in 1h 13m 50s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5635cd34e0857f28a206f088d10eb4d512d08882","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-20 13:43:43.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"1d664eae584034af0592e9de58aea5bc4503ab92","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-20 13:44:31.000000000","message":"Patch Set 11:\n\nLast item to have this good to go: proper zuulv3 multinode job","accounts_in_message":[],"_revision_number":11},{"id":"066c0d359a0c042b3e05b168c3be23ff6a873ebd","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-20 15:31:46.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"8ce479cdad098c7ade9922e7c45f5db8f519ce03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 17:31:39.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/47/653747/12/check/test-release-openstack/0f91fa2/ : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/653747/12/check/openstack-tox-lower-constraints/bd9dafe/ : SUCCESS in 9m 42s\n- openstack-tox-pep8 http://logs.openstack.org/47/653747/12/check/openstack-tox-pep8/c763198/ : SUCCESS in 6m 12s\n- openstack-tox-py27 http://logs.openstack.org/47/653747/12/check/openstack-tox-py27/798dff1/ : SUCCESS in 8m 19s\n- openstack-tox-docs http://logs.openstack.org/47/653747/12/check/openstack-tox-docs/d91df90/html/ : SUCCESS in 4m 41s\n- requirements-check http://logs.openstack.org/47/653747/12/check/requirements-check/fd5a127/ : SUCCESS in 4m 37s\n- openstack-tox-py36 http://logs.openstack.org/47/653747/12/check/openstack-tox-py36/91543b9/ : SUCCESS in 9m 44s\n- openstack-tox-py37 http://logs.openstack.org/47/653747/12/check/openstack-tox-py37/55da9c7/ : SUCCESS in 10m 16s\n- build-openstack-releasenotes http://logs.openstack.org/47/653747/12/check/build-openstack-releasenotes/fc03b6a/html/ : SUCCESS in 4m 15s\n- networking-sfc-functional-dsvm http://logs.openstack.org/47/653747/12/check/networking-sfc-functional-dsvm/e0cd246/ : SUCCESS in 16m 05s\n- networking-sfc-tempest http://logs.openstack.org/47/653747/12/check/networking-sfc-tempest/01b35f2/ : SUCCESS in 1h 56m 39s\n- networking-sfc-tempest-multinode http://logs.openstack.org/47/653747/12/check/networking-sfc-tempest-multinode/6345f89/ : FAILURE in 1h 12m 39s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"9c89b7138fcabab6c5364d0282c097d8204afd1c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-21 09:16:02.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"cf115cf7f627dc71b28ae6b47e89cb10929afc7f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-21 09:24:56.000000000","message":"Patch Set 12: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"3e71635e95c210b6e8475afbe5e4282bb65d67d0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-21 09:29:11.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"dc6c7bfa895b9d7ff2bf64ddebe494a88f7799f9","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-21 11:11:45.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"77b6253226a927b39f3fbc5a74c88694780d8624","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 13:16:59.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/47/653747/14/check/test-release-openstack/13b1356/ : SUCCESS in 3m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/653747/14/check/openstack-tox-lower-constraints/468c7ac/ : SUCCESS in 9m 43s\n- openstack-tox-pep8 http://logs.openstack.org/47/653747/14/check/openstack-tox-pep8/28396a1/ : SUCCESS in 6m 32s\n- openstack-tox-py27 http://logs.openstack.org/47/653747/14/check/openstack-tox-py27/bbdc0d4/ : SUCCESS in 9m 20s\n- openstack-tox-docs http://logs.openstack.org/47/653747/14/check/openstack-tox-docs/cbac04e/html/ : SUCCESS in 4m 33s\n- requirements-check http://logs.openstack.org/47/653747/14/check/requirements-check/ace2e45/ : SUCCESS in 3m 26s\n- openstack-tox-py36 http://logs.openstack.org/47/653747/14/check/openstack-tox-py36/894297b/ : SUCCESS in 8m 39s\n- openstack-tox-py37 http://logs.openstack.org/47/653747/14/check/openstack-tox-py37/990b8a7/ : SUCCESS in 9m 07s\n- build-openstack-releasenotes http://logs.openstack.org/47/653747/14/check/build-openstack-releasenotes/57d1fb8/html/ : SUCCESS in 3m 54s\n- networking-sfc-functional-dsvm http://logs.openstack.org/47/653747/14/check/networking-sfc-functional-dsvm/bafe398/ : SUCCESS in 14m 38s\n- networking-sfc-tempest http://logs.openstack.org/47/653747/14/check/networking-sfc-tempest/42b7302/ : SUCCESS in 1h 55m 52s\n- networking-sfc-tempest-multinode http://logs.openstack.org/47/653747/14/check/networking-sfc-tempest-multinode/9151946/ : RETRY_LIMIT in 6m 46s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"abc06bdc12b43de3b5beaca971901a2d16f7db6f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-21 13:30:48.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"139208ac94e59c08de8ce7af8d2631c4af435530","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 15:37:38.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/47/653747/15/check/test-release-openstack/18d360a/ : SUCCESS in 3m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/653747/15/check/openstack-tox-lower-constraints/319227e/ : SUCCESS in 9m 51s\n- openstack-tox-pep8 http://logs.openstack.org/47/653747/15/check/openstack-tox-pep8/c2b8805/ : SUCCESS in 6m 27s\n- openstack-tox-py27 http://logs.openstack.org/47/653747/15/check/openstack-tox-py27/37b9ab5/ : SUCCESS in 8m 20s\n- openstack-tox-docs http://logs.openstack.org/47/653747/15/check/openstack-tox-docs/59b187d/html/ : SUCCESS in 4m 45s\n- requirements-check http://logs.openstack.org/47/653747/15/check/requirements-check/1508a14/ : SUCCESS in 3m 10s\n- openstack-tox-py36 http://logs.openstack.org/47/653747/15/check/openstack-tox-py36/34b37e4/ : SUCCESS in 11m 06s\n- openstack-tox-py37 http://logs.openstack.org/47/653747/15/check/openstack-tox-py37/5f7faeb/ : SUCCESS in 11m 35s\n- build-openstack-releasenotes http://logs.openstack.org/47/653747/15/check/build-openstack-releasenotes/8a17037/html/ : SUCCESS in 5m 06s\n- networking-sfc-functional-dsvm http://logs.openstack.org/47/653747/15/check/networking-sfc-functional-dsvm/b327776/ : SUCCESS in 14m 13s\n- networking-sfc-tempest http://logs.openstack.org/47/653747/15/check/networking-sfc-tempest/376ee25/ : SUCCESS in 1h 59m 51s\n- networking-sfc-tempest-multinode http://logs.openstack.org/47/653747/15/check/networking-sfc-tempest-multinode/4f54afe/ : RETRY_LIMIT in 5m 46s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"57cd98f326879c66446bfe490059977a9f498e13","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-23 09:18:21.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"a6154c2d6b0f395e376b509673955ac58f0ee564","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-23 09:19:01.000000000","message":"Patch Set 16: Workflow-1\n\nTesting by inheriting directly DVR multinode job","accounts_in_message":[],"_revision_number":16},{"id":"cb4dcbeba9154414e9a25fc79161fc5f0b13f0d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 11:46:47.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/47/653747/16/check/test-release-openstack/a39c6d7/ : SUCCESS in 3m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/653747/16/check/openstack-tox-lower-constraints/ca83f91/ : SUCCESS in 9m 06s\n- openstack-tox-pep8 http://logs.openstack.org/47/653747/16/check/openstack-tox-pep8/9c61bff/ : SUCCESS in 6m 47s\n- openstack-tox-py27 http://logs.openstack.org/47/653747/16/check/openstack-tox-py27/ae70201/ : SUCCESS in 11m 08s\n- openstack-tox-docs http://logs.openstack.org/47/653747/16/check/openstack-tox-docs/21df788/html/ : SUCCESS in 5m 01s\n- requirements-check http://logs.openstack.org/47/653747/16/check/requirements-check/d602574/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/47/653747/16/check/openstack-tox-py36/61c0f75/ : SUCCESS in 10m 27s\n- openstack-tox-py37 http://logs.openstack.org/47/653747/16/check/openstack-tox-py37/48cbf96/ : SUCCESS in 8m 13s\n- build-openstack-releasenotes http://logs.openstack.org/47/653747/16/check/build-openstack-releasenotes/d796b1e/html/ : SUCCESS in 4m 26s\n- networking-sfc-functional-dsvm http://logs.openstack.org/47/653747/16/check/networking-sfc-functional-dsvm/5ecd48c/ : SUCCESS in 12m 57s\n- networking-sfc-tempest http://logs.openstack.org/47/653747/16/check/networking-sfc-tempest/86f873d/ : SUCCESS in 1h 43m 29s\n- networking-sfc-tempest-multinode http://logs.openstack.org/47/653747/16/check/networking-sfc-tempest-multinode/d426179/ : FAILURE in 2h 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"066684a54a343c9d3c2521b39bb67dc3a0512f69","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-23 14:31:22.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"9515033715a7979561b7d04111ef4761a8490d2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 16:34:02.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/47/653747/17/check/test-release-openstack/ec370d8/ : SUCCESS in 3m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/653747/17/check/openstack-tox-lower-constraints/e2d3b68/ : SUCCESS in 10m 47s\n- openstack-tox-pep8 http://logs.openstack.org/47/653747/17/check/openstack-tox-pep8/7fa7648/ : SUCCESS in 7m 15s\n- openstack-tox-py27 http://logs.openstack.org/47/653747/17/check/openstack-tox-py27/e86f542/ : SUCCESS in 8m 25s\n- openstack-tox-docs http://logs.openstack.org/47/653747/17/check/openstack-tox-docs/5d1f755/html/ : SUCCESS in 4m 48s\n- requirements-check http://logs.openstack.org/47/653747/17/check/requirements-check/6c1df54/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/47/653747/17/check/openstack-tox-py36/de6ee77/ : SUCCESS in 9m 05s\n- openstack-tox-py37 http://logs.openstack.org/47/653747/17/check/openstack-tox-py37/1b9fe51/ : SUCCESS in 14m 22s\n- build-openstack-releasenotes http://logs.openstack.org/47/653747/17/check/build-openstack-releasenotes/2f7b332/html/ : SUCCESS in 3m 57s\n- networking-sfc-functional-dsvm http://logs.openstack.org/47/653747/17/check/networking-sfc-functional-dsvm/7d0068b/ : SUCCESS in 25m 59s\n- networking-sfc-tempest http://logs.openstack.org/47/653747/17/check/networking-sfc-tempest/88b17f4/ : SUCCESS in 1h 47m 04s\n- networking-sfc-tempest-multinode http://logs.openstack.org/47/653747/17/check/networking-sfc-tempest-multinode/5c1fb58/ : RETRY_LIMIT in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"f97a5177b6ef97e879e18a1ad15cb0deb704e727","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-23 16:54:43.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"94d8ff8b659a22c5bbee79aea3a7b6f12a0a5d87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 18:57:23.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/47/653747/18/check/test-release-openstack/313c84a/ : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/653747/18/check/openstack-tox-lower-constraints/e00c69f/ : SUCCESS in 8m 53s\n- openstack-tox-pep8 http://logs.openstack.org/47/653747/18/check/openstack-tox-pep8/282cb6c/ : SUCCESS in 7m 34s\n- openstack-tox-py27 http://logs.openstack.org/47/653747/18/check/openstack-tox-py27/dda3bdd/ : SUCCESS in 9m 35s\n- openstack-tox-docs http://logs.openstack.org/47/653747/18/check/openstack-tox-docs/abb6f68/html/ : SUCCESS in 5m 37s\n- requirements-check http://logs.openstack.org/47/653747/18/check/requirements-check/373330d/ : SUCCESS in 4m 24s\n- openstack-tox-py36 http://logs.openstack.org/47/653747/18/check/openstack-tox-py36/64d2eb0/ : SUCCESS in 9m 08s\n- openstack-tox-py37 http://logs.openstack.org/47/653747/18/check/openstack-tox-py37/7c5d6c7/ : SUCCESS in 9m 21s\n- build-openstack-releasenotes http://logs.openstack.org/47/653747/18/check/build-openstack-releasenotes/31d803d/html/ : SUCCESS in 4m 17s\n- networking-sfc-functional-dsvm http://logs.openstack.org/47/653747/18/check/networking-sfc-functional-dsvm/7d4ec69/ : SUCCESS in 14m 22s\n- networking-sfc-tempest http://logs.openstack.org/47/653747/18/check/networking-sfc-tempest/f51792c/ : SUCCESS in 1h 59m 44s\n- networking-sfc-tempest-multinode http://logs.openstack.org/47/653747/18/check/networking-sfc-tempest-multinode/1846e2d/ : FAILURE in 1h 16m 38s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"b71aae473462e52d5404ce32658d14cb9b594c2b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-24 13:38:04.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"5250f0047a3ff77947e5651391fedf923226b0ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 15:26:57.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/47/653747/19/check/test-release-openstack/6e02bb7/ : SUCCESS in 3m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/653747/19/check/openstack-tox-lower-constraints/67d5810/ : SUCCESS in 8m 46s\n- openstack-tox-pep8 http://logs.openstack.org/47/653747/19/check/openstack-tox-pep8/e74ae7c/ : SUCCESS in 15m 06s\n- openstack-tox-py27 http://logs.openstack.org/47/653747/19/check/openstack-tox-py27/80f4e1e/ : SUCCESS in 8m 06s\n- openstack-tox-docs http://logs.openstack.org/47/653747/19/check/openstack-tox-docs/705c003/html/ : SUCCESS in 5m 35s\n- requirements-check http://logs.openstack.org/47/653747/19/check/requirements-check/d6caef1/ : SUCCESS in 3m 28s\n- openstack-tox-py36 http://logs.openstack.org/47/653747/19/check/openstack-tox-py36/39c8ac4/ : SUCCESS in 8m 42s\n- openstack-tox-py37 http://logs.openstack.org/47/653747/19/check/openstack-tox-py37/e5030f0/ : SUCCESS in 8m 20s\n- build-openstack-releasenotes http://logs.openstack.org/47/653747/19/check/build-openstack-releasenotes/3359178/html/ : SUCCESS in 3m 53s\n- networking-sfc-functional-dsvm http://logs.openstack.org/47/653747/19/check/networking-sfc-functional-dsvm/fe74ca8/ : SUCCESS in 14m 40s\n- networking-sfc-tempest http://logs.openstack.org/47/653747/19/check/networking-sfc-tempest/bfb25a8/ : SUCCESS in 1h 46m 12s\n- networking-sfc-tempest-multinode http://logs.openstack.org/47/653747/19/check/networking-sfc-tempest-multinode/29419a0/ : FAILURE in 1h 42m 18s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"f8c66e82e2f026b9314ab54643d329bccdf2893a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-24 15:43:07.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"9a8bb2a09ca676b3d6642de5a8e3c81f5ee92638","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 17:38:58.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/47/653747/20/check/test-release-openstack/28aa400/ : SUCCESS in 3m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/653747/20/check/openstack-tox-lower-constraints/ff39dca/ : SUCCESS in 10m 00s\n- openstack-tox-pep8 http://logs.openstack.org/47/653747/20/check/openstack-tox-pep8/8acd4aa/ : SUCCESS in 6m 00s\n- openstack-tox-py27 http://logs.openstack.org/47/653747/20/check/openstack-tox-py27/3bb5669/ : SUCCESS in 9m 30s\n- openstack-tox-docs http://logs.openstack.org/47/653747/20/check/openstack-tox-docs/dc27d1c/html/ : SUCCESS in 4m 46s\n- requirements-check http://logs.openstack.org/47/653747/20/check/requirements-check/3dfd737/ : SUCCESS in 3m 44s\n- openstack-tox-py36 http://logs.openstack.org/47/653747/20/check/openstack-tox-py36/216b9f7/ : SUCCESS in 9m 26s\n- openstack-tox-py37 http://logs.openstack.org/47/653747/20/check/openstack-tox-py37/4afbc18/ : SUCCESS in 8m 37s\n- build-openstack-releasenotes http://logs.openstack.org/47/653747/20/check/build-openstack-releasenotes/845fe55/html/ : SUCCESS in 4m 05s\n- networking-sfc-functional-dsvm http://logs.openstack.org/47/653747/20/check/networking-sfc-functional-dsvm/a5767d1/ : SUCCESS in 14m 02s\n- networking-sfc-tempest http://logs.openstack.org/47/653747/20/check/networking-sfc-tempest/d11f13c/ : SUCCESS in 1h 48m 56s\n- networking-sfc-tempest-multinode http://logs.openstack.org/47/653747/20/check/networking-sfc-tempest-multinode/20d7e4a/ : RETRY_LIMIT in 10m 28s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"463d5117af7643e35546147b5bcdc4f20dd49651","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-25 07:24:07.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"eb92ac0c23e8823e5b66e096e204aa77b8dfdc05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 09:24:31.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/47/653747/21/check/test-release-openstack/2b5b5f8/ : SUCCESS in 3m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/653747/21/check/openstack-tox-lower-constraints/999baaf/ : SUCCESS in 8m 23s\n- openstack-tox-pep8 http://logs.openstack.org/47/653747/21/check/openstack-tox-pep8/d3b13f5/ : SUCCESS in 5m 19s\n- openstack-tox-py27 http://logs.openstack.org/47/653747/21/check/openstack-tox-py27/3ec6bdf/ : SUCCESS in 7m 49s\n- openstack-tox-docs http://logs.openstack.org/47/653747/21/check/openstack-tox-docs/013e25e/html/ : SUCCESS in 3m 56s\n- requirements-check http://logs.openstack.org/47/653747/21/check/requirements-check/97c3a75/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/47/653747/21/check/openstack-tox-py36/6d56576/ : SUCCESS in 7m 56s\n- openstack-tox-py37 http://logs.openstack.org/47/653747/21/check/openstack-tox-py37/e59d821/ : SUCCESS in 8m 46s\n- build-openstack-releasenotes http://logs.openstack.org/47/653747/21/check/build-openstack-releasenotes/ef5ef2a/html/ : SUCCESS in 3m 44s\n- networking-sfc-functional-dsvm http://logs.openstack.org/47/653747/21/check/networking-sfc-functional-dsvm/ba26caf/ : SUCCESS in 16m 16s\n- networking-sfc-tempest http://logs.openstack.org/47/653747/21/check/networking-sfc-tempest/6f78129/ : SUCCESS in 1h 47m 01s\n- networking-sfc-tempest-multinode http://logs.openstack.org/47/653747/21/check/networking-sfc-tempest-multinode/380a25d/ : SUCCESS in 1h 54m 12s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"6cd5450ea765328c91d4183b61d39ac5c4df616f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-26 15:46:53.000000000","message":"Patch Set 21:\n\nGood to go for review!\nhttp://logs.openstack.org/47/653747/21/check/networking-sfc-tempest-multinode/380a25d/controller/logs/tempest_log.txt.gz#_2019-05-25_08_08_47_421 says \"Running test on multi node\"","accounts_in_message":[],"_revision_number":21},{"id":"a8a550d969f83e51c618da56caade5c07208b1ed","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-05-28 23:20:24.000000000","message":"Patch Set 21: Code-Review+2","accounts_in_message":[],"_revision_number":21},{"id":"f65194383f89a9d10d4750e859f230b8fba19163","author":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"date":"2019-06-04 10:12:37.000000000","message":"Patch Set 21: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":21},{"id":"b47db8c6f9c8740b0e6220aaa5d4f42c7e397dda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 14:09:38.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"563e380b4977de0a5e8f5e52d8cba322f8529c07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 14:52:20.000000000","message":"Patch Set 21: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":21},{"id":"4621c3c375348bb5f0050a588c415accba83b65c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-06-04 16:50:42.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"c6f3e500c1192b7446b60e537504f69a1d085b85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 18:57:11.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/47/653747/22/check/test-release-openstack/d71a07e/ : SUCCESS in 4m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/653747/22/check/openstack-tox-lower-constraints/a12db50/ : SUCCESS in 9m 25s\n- openstack-tox-pep8 http://logs.openstack.org/47/653747/22/check/openstack-tox-pep8/c8ab1fa/ : SUCCESS in 6m 32s\n- openstack-tox-py27 http://logs.openstack.org/47/653747/22/check/openstack-tox-py27/01ee3bf/ : SUCCESS in 8m 53s\n- openstack-tox-docs http://logs.openstack.org/47/653747/22/check/openstack-tox-docs/19b852b/html/ : SUCCESS in 4m 39s\n- requirements-check http://logs.openstack.org/47/653747/22/check/requirements-check/d5108d9/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/47/653747/22/check/openstack-tox-py36/827f667/ : SUCCESS in 8m 45s\n- openstack-tox-py37 http://logs.openstack.org/47/653747/22/check/openstack-tox-py37/a433308/ : SUCCESS in 11m 45s\n- build-openstack-releasenotes http://logs.openstack.org/47/653747/22/check/build-openstack-releasenotes/5d7a2bf/html/ : SUCCESS in 5m 14s\n- networking-sfc-functional-dsvm http://logs.openstack.org/47/653747/22/check/networking-sfc-functional-dsvm/35b3c5a/ : SUCCESS in 22m 10s\n- networking-sfc-tempest http://logs.openstack.org/47/653747/22/check/networking-sfc-tempest/6d7e08e/ : SUCCESS in 1h 53m 06s\n- networking-sfc-tempest-multinode http://logs.openstack.org/47/653747/22/check/networking-sfc-tempest-multinode/50b926f/ : SUCCESS in 1h 56m 45s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"43c9afa4ef5849a655d4129e206ee95dd5fa0ccd","author":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"date":"2019-06-05 07:49:21.000000000","message":"Patch Set 22: Workflow+1","accounts_in_message":[],"_revision_number":22},{"id":"cf8976e0f409e341a593509ab39c160244f2823d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 13:48:41.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"a0e6fc3b2b6613ff997fe1387a2f4af0db6f32d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 16:18:48.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/47/653747/22/gate/test-release-openstack/b3f6cc6/ : SUCCESS in 5m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/653747/22/gate/openstack-tox-lower-constraints/17ec2c2/ : SUCCESS in 11m 43s\n- openstack-tox-pep8 http://logs.openstack.org/47/653747/22/gate/openstack-tox-pep8/ad2bc6d/ : SUCCESS in 5m 59s\n- openstack-tox-py27 http://logs.openstack.org/47/653747/22/gate/openstack-tox-py27/0db491f/ : SUCCESS in 10m 17s\n- openstack-tox-docs http://logs.openstack.org/47/653747/22/gate/openstack-tox-docs/bd1a13b/html/ : SUCCESS in 4m 43s\n- requirements-check http://logs.openstack.org/47/653747/22/gate/requirements-check/020e960/ : SUCCESS in 3m 17s\n- openstack-tox-py36 http://logs.openstack.org/47/653747/22/gate/openstack-tox-py36/825e34a/ : SUCCESS in 10m 28s\n- openstack-tox-py37 http://logs.openstack.org/47/653747/22/gate/openstack-tox-py37/5d30fe7/ : SUCCESS in 8m 47s\n- build-openstack-releasenotes http://logs.openstack.org/47/653747/22/gate/build-openstack-releasenotes/a2c06e8/html/ : SUCCESS in 3m 43s\n- networking-sfc-functional-dsvm http://logs.openstack.org/47/653747/22/gate/networking-sfc-functional-dsvm/c8f35d7/ : SUCCESS in 12m 10s\n- networking-sfc-tempest http://logs.openstack.org/47/653747/22/gate/networking-sfc-tempest/647cc0b/ : SUCCESS in 1h 56m 59s","accounts_in_message":[],"_revision_number":22},{"id":"d7462c395e2c221c8650a2c43db7ae3e830f78e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 16:18:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"e617fc9f760858f9673619112cf7cb89880384d7","revisions":{"0e5575fc96935ad55c0cde4a1a1275c4b0cb4a79":{"kind":"REWORK","_number":1,"created":"2019-04-18 14:39:22.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/1"}}},"commit":{"parents":[{"commit":"a96923819c0e94fa2c7a41c1649b8de54f90acee","subject":"Merge \"Replace openstack.org git:// URLs with https://\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/a96923819c0e94fa2c7a41c1649b8de54f90acee"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"subject":"Complete move of neutron-sfc tempest tests to tempest plugin","message":"Complete move of neutron-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.openstack.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/0e5575fc96935ad55c0cde4a1a1275c4b0cb4a79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/0e5575fc96935ad55c0cde4a1a1275c4b0cb4a79"}]},"branch":"refs/heads/master"},"0f9a084a529a6c2ed192e76de7a9904bec64e029":{"kind":"REWORK","_number":2,"created":"2019-04-18 14:40:45.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/2"}}},"commit":{"parents":[{"commit":"a96923819c0e94fa2c7a41c1649b8de54f90acee","subject":"Merge \"Replace openstack.org git:// URLs with https://\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/a96923819c0e94fa2c7a41c1649b8de54f90acee"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:40:27.000000000","tz":120},"subject":"Complete move of neutron-sfc tempest tests to tempest plugin","message":"Complete move of neutron-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.openstack.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/0f9a084a529a6c2ed192e76de7a9904bec64e029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/0f9a084a529a6c2ed192e76de7a9904bec64e029"}]},"branch":"refs/heads/master"},"7899b91fe9bc4478adaf2d9da66b447eefe42a9a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-04-18 14:42:01.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/3"}}},"commit":{"parents":[{"commit":"a96923819c0e94fa2c7a41c1649b8de54f90acee","subject":"Merge \"Replace openstack.org git:// URLs with https://\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/a96923819c0e94fa2c7a41c1649b8de54f90acee"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:41:47.000000000","tz":120},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.openstack.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/7899b91fe9bc4478adaf2d9da66b447eefe42a9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/7899b91fe9bc4478adaf2d9da66b447eefe42a9a"}]},"branch":"refs/heads/master"},"38f0ad8933dfdbd34c400aeb246218143208e59d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-04-19 13:09:42.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/4"}}},"commit":{"parents":[{"commit":"a42d5fad7952a98b59227d4ac26b02bf0d29080e","subject":"Combine two patches to fix gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/a42d5fad7952a98b59227d4ac26b02bf0d29080e"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-19 13:09:42.000000000","tz":0},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.openstack.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/38f0ad8933dfdbd34c400aeb246218143208e59d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/38f0ad8933dfdbd34c400aeb246218143208e59d"}]},"branch":"refs/heads/master"},"47571d37d85db4081fe80e9d3c426dc468360f4d":{"kind":"REWORK","_number":5,"created":"2019-04-23 08:14:20.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/5"}}},"commit":{"parents":[{"commit":"4bbc31ee65553e65f050bb3921726d502fd3eaa2","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/4bbc31ee65553e65f050bb3921726d502fd3eaa2"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-23 08:14:08.000000000","tz":120},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.openstack.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/47571d37d85db4081fe80e9d3c426dc468360f4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/47571d37d85db4081fe80e9d3c426dc468360f4d"}]},"branch":"refs/heads/master"},"75332aecf463eea76d9bc7bf1236b79e8a88296e":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-04-23 08:55:38.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/6"}}},"commit":{"parents":[{"commit":"4bbc31ee65553e65f050bb3921726d502fd3eaa2","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/4bbc31ee65553e65f050bb3921726d502fd3eaa2"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-23 08:55:20.000000000","tz":120},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.opendev.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/75332aecf463eea76d9bc7bf1236b79e8a88296e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/75332aecf463eea76d9bc7bf1236b79e8a88296e"}]},"branch":"refs/heads/master"},"d498b5137b1dd091bfd1f07391a1f93f378516e3":{"kind":"REWORK","_number":7,"created":"2019-04-23 09:49:39.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/7"}}},"commit":{"parents":[{"commit":"4bbc31ee65553e65f050bb3921726d502fd3eaa2","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/4bbc31ee65553e65f050bb3921726d502fd3eaa2"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-23 09:49:19.000000000","tz":120},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.opendev.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/d498b5137b1dd091bfd1f07391a1f93f378516e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/d498b5137b1dd091bfd1f07391a1f93f378516e3"}]},"branch":"refs/heads/master"},"4d4d701634a7b51b2b1c352975dea463111ae4e8":{"kind":"REWORK","_number":8,"created":"2019-05-01 00:10:19.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/8"}}},"commit":{"parents":[{"commit":"4bbc31ee65553e65f050bb3921726d502fd3eaa2","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/4bbc31ee65553e65f050bb3921726d502fd3eaa2"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-05-01 00:10:00.000000000","tz":120},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.opendev.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/4d4d701634a7b51b2b1c352975dea463111ae4e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/4d4d701634a7b51b2b1c352975dea463111ae4e8"}]},"branch":"refs/heads/master"},"1c556ea143ebda7ca9c9ad6dc11ff318417a11f5":{"kind":"REWORK","_number":9,"created":"2019-05-01 00:21:10.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/9"}}},"commit":{"parents":[{"commit":"4bbc31ee65553e65f050bb3921726d502fd3eaa2","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/4bbc31ee65553e65f050bb3921726d502fd3eaa2"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-05-01 00:20:51.000000000","tz":120},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.opendev.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/1c556ea143ebda7ca9c9ad6dc11ff318417a11f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/1c556ea143ebda7ca9c9ad6dc11ff318417a11f5"}]},"branch":"refs/heads/master"},"34cf4dc56e5ce85865f2613b9957a89a8ffcbe0e":{"kind":"REWORK","_number":10,"created":"2019-05-01 00:24:52.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/10"}}},"commit":{"parents":[{"commit":"4bbc31ee65553e65f050bb3921726d502fd3eaa2","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/4bbc31ee65553e65f050bb3921726d502fd3eaa2"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-05-01 00:24:37.000000000","tz":120},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.opendev.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/34cf4dc56e5ce85865f2613b9957a89a8ffcbe0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/34cf4dc56e5ce85865f2613b9957a89a8ffcbe0e"}]},"branch":"refs/heads/master"},"3f388a7131b316e310f579fbe6fd670345fcb802":{"kind":"REWORK","_number":11,"created":"2019-05-20 13:43:43.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/11"}}},"commit":{"parents":[{"commit":"4bbc31ee65553e65f050bb3921726d502fd3eaa2","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/4bbc31ee65553e65f050bb3921726d502fd3eaa2"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-05-20 13:43:33.000000000","tz":120},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.opendev.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/3f388a7131b316e310f579fbe6fd670345fcb802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/3f388a7131b316e310f579fbe6fd670345fcb802"}]},"branch":"refs/heads/master"},"afda8978fa6e76899396763f3389f50264f912a1":{"kind":"REWORK","_number":12,"created":"2019-05-20 15:31:46.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/12"}}},"commit":{"parents":[{"commit":"f0168d95a51a53550717844701f042bba66fc987","subject":"Update sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f0168d95a51a53550717844701f042bba66fc987"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-05-20 15:31:17.000000000","tz":120},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.opendev.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/afda8978fa6e76899396763f3389f50264f912a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/afda8978fa6e76899396763f3389f50264f912a1"}]},"branch":"refs/heads/master"},"bbc25ed09e416545a1994fcc7a5835b49eff57e4":{"kind":"REWORK","_number":13,"created":"2019-05-21 09:29:11.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/13"}}},"commit":{"parents":[{"commit":"f0168d95a51a53550717844701f042bba66fc987","subject":"Update sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f0168d95a51a53550717844701f042bba66fc987"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-05-21 09:28:17.000000000","tz":120},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.opendev.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/bbc25ed09e416545a1994fcc7a5835b49eff57e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/bbc25ed09e416545a1994fcc7a5835b49eff57e4"}]},"branch":"refs/heads/master"},"41533c082be87cc9a1cf7f51404b4bc8c3ad0e3b":{"kind":"REWORK","_number":14,"created":"2019-05-21 11:11:45.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/14"}}},"commit":{"parents":[{"commit":"f0168d95a51a53550717844701f042bba66fc987","subject":"Update sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f0168d95a51a53550717844701f042bba66fc987"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-05-21 11:11:19.000000000","tz":120},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\nThis also converts the tempest tests to zuulv3\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.opendev.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/41533c082be87cc9a1cf7f51404b4bc8c3ad0e3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/41533c082be87cc9a1cf7f51404b4bc8c3ad0e3b"}]},"branch":"refs/heads/master"},"6435ccb331d3350089f743236a0ddf996afc67dc":{"kind":"REWORK","_number":15,"created":"2019-05-21 13:30:48.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/15"}}},"commit":{"parents":[{"commit":"f0168d95a51a53550717844701f042bba66fc987","subject":"Update sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f0168d95a51a53550717844701f042bba66fc987"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-05-21 13:30:37.000000000","tz":120},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\nThis also converts the tempest tests to zuulv3\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.opendev.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/6435ccb331d3350089f743236a0ddf996afc67dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/6435ccb331d3350089f743236a0ddf996afc67dc"}]},"branch":"refs/heads/master"},"f0c57fe7498c7d46b1fd5a31350278da058986c9":{"kind":"REWORK","_number":16,"created":"2019-05-23 09:18:21.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/16"}}},"commit":{"parents":[{"commit":"f0168d95a51a53550717844701f042bba66fc987","subject":"Update sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f0168d95a51a53550717844701f042bba66fc987"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-05-23 09:17:32.000000000","tz":120},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\nThis also converts the tempest tests to zuulv3\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.opendev.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f0c57fe7498c7d46b1fd5a31350278da058986c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f0c57fe7498c7d46b1fd5a31350278da058986c9"}]},"branch":"refs/heads/master"},"5a7097ff9a8869301f3f31ec3c19fc5c741e98f4":{"kind":"REWORK","_number":17,"created":"2019-05-23 14:31:22.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/17"}}},"commit":{"parents":[{"commit":"f0168d95a51a53550717844701f042bba66fc987","subject":"Update sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f0168d95a51a53550717844701f042bba66fc987"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-05-23 14:30:38.000000000","tz":120},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\nThis also converts the tempest tests to zuulv3\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.opendev.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/5a7097ff9a8869301f3f31ec3c19fc5c741e98f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/5a7097ff9a8869301f3f31ec3c19fc5c741e98f4"}]},"branch":"refs/heads/master"},"adcb62fd9ff69059dab8ad61ccfe56b310600d38":{"kind":"REWORK","_number":18,"created":"2019-05-23 16:54:43.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/18"}}},"commit":{"parents":[{"commit":"f0168d95a51a53550717844701f042bba66fc987","subject":"Update sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f0168d95a51a53550717844701f042bba66fc987"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-05-23 16:47:57.000000000","tz":120},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\nThis also converts the tempest tests to zuulv3\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.opendev.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/adcb62fd9ff69059dab8ad61ccfe56b310600d38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/adcb62fd9ff69059dab8ad61ccfe56b310600d38"}]},"branch":"refs/heads/master"},"f50e56e01a2b3a454e407c1d068776f70505ea9a":{"kind":"REWORK","_number":19,"created":"2019-05-24 13:38:04.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/19"}}},"commit":{"parents":[{"commit":"f0168d95a51a53550717844701f042bba66fc987","subject":"Update sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f0168d95a51a53550717844701f042bba66fc987"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-05-24 13:35:33.000000000","tz":120},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\nThis also converts the tempest tests to zuulv3\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.opendev.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f50e56e01a2b3a454e407c1d068776f70505ea9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f50e56e01a2b3a454e407c1d068776f70505ea9a"}]},"branch":"refs/heads/master"},"cbb7818349fd2cd5daf28494ac486f33b9785394":{"kind":"REWORK","_number":20,"created":"2019-05-24 15:43:07.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/20"}}},"commit":{"parents":[{"commit":"f0168d95a51a53550717844701f042bba66fc987","subject":"Update sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f0168d95a51a53550717844701f042bba66fc987"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-05-24 15:41:31.000000000","tz":120},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\nThis also converts the tempest tests to zuulv3\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.opendev.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/cbb7818349fd2cd5daf28494ac486f33b9785394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/cbb7818349fd2cd5daf28494ac486f33b9785394"}]},"branch":"refs/heads/master"},"d18b8b840bff76a842476a7dffc9a8fc5b038a03":{"kind":"REWORK","_number":21,"created":"2019-05-25 07:24:07.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/21"}}},"commit":{"parents":[{"commit":"f0168d95a51a53550717844701f042bba66fc987","subject":"Update sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f0168d95a51a53550717844701f042bba66fc987"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-05-25 07:23:52.000000000","tz":120},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\nThis also converts the tempest tests to zuulv3\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.opendev.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/d18b8b840bff76a842476a7dffc9a8fc5b038a03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/d18b8b840bff76a842476a7dffc9a8fc5b038a03"}]},"branch":"refs/heads/master"},"e617fc9f760858f9673619112cf7cb89880384d7":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2019-06-04 16:50:42.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/47/653747/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/47/653747/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/47/653747/22"}}},"commit":{"parents":[{"commit":"7ff51c0c4d24099964e4f0a9dc4831088be4caa8","subject":"Update sphinx requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/7ff51c0c4d24099964e4f0a9dc4831088be4caa8"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-04-18 14:38:36.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-06-04 16:50:42.000000000","tz":0},"subject":"Complete move of networking-sfc tempest tests to tempest plugin","message":"Complete move of networking-sfc tempest tests to tempest plugin\n\nAs discussed in the neutron_ci meeting [1] the QA team would like to\nmove the tempest tests for the stadium projects from their repos to\nrepos specific to being tempest plugins. This is the second part of a two\nstage move, by removing the tempest tests that were copied to the\nneutron-tempest-plugin repo [2].\n\nThis also converts the tempest tests to zuulv3\n\n[1] http://eavesdrop.openstack.org/meetings/neutron_ci/2019/neutron_ci.2019-03-12-16.01.log.html#l-94\n[2] https://etherpad.openstack.org/p/neutron_stadium_move_to_tempest_plugin_repo\n\nChange-Id: Ib30382f6af434edde4b372e6af015e538768ddbd\nDepends-On: https://review.opendev.org/653012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/e617fc9f760858f9673619112cf7cb89880384d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/e617fc9f760858f9673619112cf7cb89880384d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
