)]}'
{"id":"openstack%2Fnetworking-sfc~660142","triplet_id":"openstack%2Fnetworking-sfc~master~I7484f260c6f58927088d9e891957be2ab355df6e","project":"openstack/networking-sfc","branch":"master","topic":"sphinx_requirements","hashtags":[],"change_id":"I7484f260c6f58927088d9e891957be2ab355df6e","subject":"Update sphinx requirements","status":"MERGED","created":"2019-05-20 14:53:32.000000000","updated":"2019-06-06 16:18:47.000000000","submitted":"2019-06-06 16:18:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660142-1559837927716-94b57270","meta_rev_id":"cac1c67f66dff97c20c6e4f0ff72a0f2bc71fcc7","_number":660142,"virtual_id_number":660142,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-04 15:09:27.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-06-06 16:18:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"recommended":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-06-04 16:22:41.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-06-04 15:08:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":2,"date":"2019-06-04 15:08:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-06-04 18:37:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-23 07:44:53.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-06-04 16:22:41.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-06-04 18:37:04.000000000","updated_by":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"reviewer":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"state":"REVIEWER"},{"updated":"2019-06-06 16:18:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"623116a75c3a56434c0524e3c04efc3cbadd0a18","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-05-20 14:53:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf27191809d4fbafd7a385f8ddac0a10b1373b51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 15:15:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/42/660142/1/check/openstack-tox-lower-constraints/d051799/ : SUCCESS in 9m 30s\n- openstack-tox-pep8 http://logs.openstack.org/42/660142/1/check/openstack-tox-pep8/d081ba5/ : SUCCESS in 6m 09s\n- openstack-tox-py27 http://logs.openstack.org/42/660142/1/check/openstack-tox-py27/f65c9f8/ : SUCCESS in 9m 59s\n- openstack-tox-docs http://logs.openstack.org/42/660142/1/check/openstack-tox-docs/668381b/html/ : SUCCESS in 4m 45s\n- requirements-check http://logs.openstack.org/42/660142/1/check/requirements-check/0faacd1/ : SUCCESS in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/42/660142/1/check/openstack-tox-py36/4b3035a/ : SUCCESS in 10m 04s\n- openstack-tox-py37 http://logs.openstack.org/42/660142/1/check/openstack-tox-py37/6deba54/ : SUCCESS in 8m 47s\n- build-openstack-releasenotes http://logs.openstack.org/42/660142/1/check/build-openstack-releasenotes/1d33b9a/html/ : SUCCESS in 4m 22s\n- networking-sfc-functional-dsvm http://logs.openstack.org/42/660142/1/check/networking-sfc-functional-dsvm/0ea4b40/ : SUCCESS in 15m 19s","accounts_in_message":[],"_revision_number":1},{"id":"d0a1477e4ae86e5d89c306b9bf9b4b29c41d559f","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-05-23 07:44:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"57674ad17a522dfc1e540b39656ae05a7fea390f","author":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"date":"2019-06-04 14:09:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"40b5abc0d230667f266a73a0f02405d5af0b9042","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 14:09:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a940855aaf574dcb5c8084960961565eb805e1bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 14:52:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/42/660142/1/gate/openstack-tox-lower-constraints/1bade05/ : FAILURE in 7m 15s\n- openstack-tox-pep8 http://logs.openstack.org/42/660142/1/gate/openstack-tox-pep8/d44f876/ : SUCCESS in 8m 09s\n- openstack-tox-py27 http://logs.openstack.org/42/660142/1/gate/openstack-tox-py27/23c629b/ : FAILURE in 5m 36s\n- openstack-tox-docs http://logs.openstack.org/42/660142/1/gate/openstack-tox-docs/8f2d0d5/html/ : SUCCESS in 4m 31s\n- requirements-check http://logs.openstack.org/42/660142/1/gate/requirements-check/012f292/ : SUCCESS in 5m 28s\n- openstack-tox-py36 http://logs.openstack.org/42/660142/1/gate/openstack-tox-py36/9f215c0/ : FAILURE in 6m 00s\n- openstack-tox-py37 http://logs.openstack.org/42/660142/1/gate/openstack-tox-py37/63e93dc/ : FAILURE in 7m 05s\n- build-openstack-releasenotes http://logs.openstack.org/42/660142/1/gate/build-openstack-releasenotes/4aa4c78/html/ : SUCCESS in 4m 47s\n- networking-sfc-functional-dsvm http://logs.openstack.org/42/660142/1/gate/networking-sfc-functional-dsvm/0f7f80b/ : SUCCESS in 19m 42s","accounts_in_message":[],"_revision_number":1},{"id":"3ea7030b6626148f1b6493eb07a6b53b31ef54d6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-06-04 15:08:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"04df4a120cabcf026471490ac866cd11fe4cf34d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-06-04 15:09:27.000000000","message":"Patch Set 2:\n\nRebased on top of https://review.opendev.org/#/c/662342/ (now also needed to pass gates)","accounts_in_message":[],"_revision_number":2},{"id":"5abb9ca952c75c9e939f209d754e54fabfad6f14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 15:42:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/42/660142/2/check/openstack-tox-lower-constraints/bbdce09/ : SUCCESS in 9m 15s\n- openstack-tox-pep8 http://logs.openstack.org/42/660142/2/check/openstack-tox-pep8/a162f69/ : SUCCESS in 6m 38s\n- openstack-tox-py27 http://logs.openstack.org/42/660142/2/check/openstack-tox-py27/870cd14/ : SUCCESS in 9m 57s\n- openstack-tox-docs http://logs.openstack.org/42/660142/2/check/openstack-tox-docs/991b88b/html/ : SUCCESS in 4m 29s\n- requirements-check http://logs.openstack.org/42/660142/2/check/requirements-check/cc49559/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/42/660142/2/check/openstack-tox-py36/61a2ca4/ : SUCCESS in 8m 45s\n- openstack-tox-py37 http://logs.openstack.org/42/660142/2/check/openstack-tox-py37/5a99694/ : SUCCESS in 9m 27s\n- build-openstack-releasenotes http://logs.openstack.org/42/660142/2/check/build-openstack-releasenotes/5432241/html/ : SUCCESS in 4m 43s\n- networking-sfc-functional-dsvm http://logs.openstack.org/42/660142/2/check/networking-sfc-functional-dsvm/28ef72a/ : SUCCESS in 29m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ab4fb11497779c8753e427d515d7631201583005","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-06-04 16:22:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a9642fd00cf053c32966f590c1e765868276e059","author":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"date":"2019-06-04 18:37:04.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dc056a0a6d9ea9cb1531647182fdffca9b8ab372","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 13:48:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"49f0a4cc4c01cb180023d424bd03a13c41b574ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 16:18:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/42/660142/2/gate/openstack-tox-lower-constraints/ac92cfd/ : SUCCESS in 10m 01s\n- openstack-tox-pep8 http://logs.openstack.org/42/660142/2/gate/openstack-tox-pep8/3187170/ : SUCCESS in 6m 04s\n- openstack-tox-py27 http://logs.openstack.org/42/660142/2/gate/openstack-tox-py27/b536b8f/ : SUCCESS in 8m 39s\n- openstack-tox-docs http://logs.openstack.org/42/660142/2/gate/openstack-tox-docs/321e2f7/html/ : SUCCESS in 4m 58s\n- requirements-check http://logs.openstack.org/42/660142/2/gate/requirements-check/2ef5bf6/ : SUCCESS in 4m 00s\n- openstack-tox-py36 http://logs.openstack.org/42/660142/2/gate/openstack-tox-py36/b84fbc9/ : SUCCESS in 8m 23s\n- openstack-tox-py37 http://logs.openstack.org/42/660142/2/gate/openstack-tox-py37/ac64256/ : SUCCESS in 8m 42s\n- build-openstack-releasenotes http://logs.openstack.org/42/660142/2/gate/build-openstack-releasenotes/57dc028/html/ : SUCCESS in 4m 33s\n- networking-sfc-functional-dsvm http://logs.openstack.org/42/660142/2/gate/networking-sfc-functional-dsvm/d6ab76d/ : SUCCESS in 14m 56s","accounts_in_message":[],"_revision_number":2},{"id":"5e0936394b8397f49b6cedca72393250b0f5237c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 16:18:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7ff51c0c4d24099964e4f0a9dc4831088be4caa8","revisions":{"f0168d95a51a53550717844701f042bba66fc987":{"kind":"REWORK","_number":1,"created":"2019-05-20 14:53:32.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/42/660142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/42/660142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/42/660142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/42/660142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/42/660142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/42/660142/1"}}},"commit":{"parents":[{"commit":"4bbc31ee65553e65f050bb3921726d502fd3eaa2","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/4bbc31ee65553e65f050bb3921726d502fd3eaa2"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-05-20 14:52:47.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-05-20 14:52:47.000000000","tz":120},"subject":"Update sphinx requirements","message":"Update sphinx requirements\n\nrequirements-check fails with:\nhttp://logs.openstack.org/47/653747/11/check/requirements-check/d637f52/job-output.txt.gz\n\nChange-Id: I7484f260c6f58927088d9e891957be2ab355df6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f0168d95a51a53550717844701f042bba66fc987"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f0168d95a51a53550717844701f042bba66fc987"}]},"branch":"refs/heads/master"},"7ff51c0c4d24099964e4f0a9dc4831088be4caa8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-04 15:08:21.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/42/660142/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/42/660142/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/42/660142/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/42/660142/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/42/660142/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/42/660142/2"}}},"commit":{"parents":[{"commit":"0014c99fadcd18da86bf8f961336fbd03f29682c","subject":"Switch to native openflow implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/0014c99fadcd18da86bf8f961336fbd03f29682c"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-05-20 14:52:47.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-06-04 15:08:21.000000000","tz":0},"subject":"Update sphinx requirements","message":"Update sphinx requirements\n\nrequirements-check fails with:\nhttp://logs.openstack.org/47/653747/11/check/requirements-check/d637f52/job-output.txt.gz\n\nChange-Id: I7484f260c6f58927088d9e891957be2ab355df6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/7ff51c0c4d24099964e4f0a9dc4831088be4caa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/7ff51c0c4d24099964e4f0a9dc4831088be4caa8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
