)]}'
{"id":"openstack%2Fnetworking-sfc~668109","triplet_id":"openstack%2Fnetworking-sfc~master~I000436c7bbda82a95f1f190b09053b9179ccaf1f","project":"openstack/networking-sfc","branch":"master","hashtags":[],"change_id":"I000436c7bbda82a95f1f190b09053b9179ccaf1f","subject":"Add libpq-dev to bindep.txt to fix gate error","status":"MERGED","created":"2019-06-28 07:54:05.000000000","updated":"2019-12-04 18:47:40.000000000","submitted":"2019-07-30 11:33:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668109-1564486408816-d14e3b94","meta_rev_id":"17636f5246a177fd4604aea39f9ec613bc3c1086","_number":668109,"virtual_id_number":668109,"owner":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-07-30 11:33:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"recommended":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":1,"date":"2019-07-08 06:53:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-07-05 15:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":2,"date":"2019-07-25 07:04:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"all":[{"value":0,"date":"2019-07-23 06:08:01.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"date":"2019-07-08 06:26:35.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2019-07-25 07:10:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-08 06:26:35.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-07-08 06:53:01.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-07-23 06:08:01.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-07-25 07:10:06.000000000","updated_by":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"reviewer":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"state":"REVIEWER"},{"updated":"2019-07-30 11:33:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1aeb9f98048591ae850cc6def1037e9c7561083d","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-06-28 07:54:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3614a15143db64b4a4726f60888c9b2d78fdbd7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 08:15:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/09/668109/1/check/test-release-openstack/6f0e7d0/ : SUCCESS in 1m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/668109/1/check/openstack-tox-lower-constraints/1709c7d/ : SUCCESS in 8m 51s\n- openstack-tox-pep8 http://logs.openstack.org/09/668109/1/check/openstack-tox-pep8/4c4868f/ : SUCCESS in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/09/668109/1/check/openstack-tox-py27/02ca61a/ : SUCCESS in 8m 39s\n- openstack-tox-docs http://logs.openstack.org/09/668109/1/check/openstack-tox-docs/8a0a42f/html/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/09/668109/1/check/openstack-tox-py36/0e5d90b/ : SUCCESS in 7m 39s\n- openstack-tox-py37 http://logs.openstack.org/09/668109/1/check/openstack-tox-py37/1703272/ : FAILURE in 3m 17s\n- networking-sfc-functional http://logs.openstack.org/09/668109/1/check/networking-sfc-functional/4018e90/ : SUCCESS in 15m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1bf8d8d59a893000884d4a6c451a01ea736596c8","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-07-02 06:13:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ce1af853d64cb86d34181307c2e2bd5f3c9903a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 06:34:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/09/668109/1/check/test-release-openstack/59c92ba/ : SUCCESS in 1m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/668109/1/check/openstack-tox-lower-constraints/1b60b6f/ : SUCCESS in 7m 48s\n- openstack-tox-pep8 http://logs.openstack.org/09/668109/1/check/openstack-tox-pep8/7efc9fc/ : SUCCESS in 6m 26s\n- openstack-tox-py27 http://logs.openstack.org/09/668109/1/check/openstack-tox-py27/60d60dd/ : SUCCESS in 7m 16s\n- openstack-tox-docs http://logs.openstack.org/09/668109/1/check/openstack-tox-docs/080389f/html/ : SUCCESS in 4m 55s\n- openstack-tox-py36 http://logs.openstack.org/09/668109/1/check/openstack-tox-py36/ae09361/ : SUCCESS in 9m 43s\n- openstack-tox-py37 http://logs.openstack.org/09/668109/1/check/openstack-tox-py37/af3c478/ : FAILURE in 2m 29s\n- networking-sfc-functional http://logs.openstack.org/09/668109/1/check/networking-sfc-functional/06f8855/ : SUCCESS in 13m 55s","accounts_in_message":[],"_revision_number":1},{"id":"24ea0a94ec306c7eeec03ed6158bd7e59bbec061","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-07-05 10:22:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ed788568571bdc23d446932de16af02e7ed8fcdc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 12:48:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/09/668109/2/check/openstack-tox-lower-constraints/c74bd5a/ : SUCCESS in 8m 00s\n- openstack-tox-pep8 http://logs.openstack.org/09/668109/2/check/openstack-tox-pep8/cbf9a8a/ : SUCCESS in 4m 53s\n- openstack-tox-py27 http://logs.openstack.org/09/668109/2/check/openstack-tox-py27/9e39222/ : SUCCESS in 7m 06s\n- openstack-tox-docs http://logs.openstack.org/09/668109/2/check/openstack-tox-docs/858fd9e/html/ : SUCCESS in 5m 14s\n- openstack-tox-py36 http://logs.openstack.org/09/668109/2/check/openstack-tox-py36/c605cf9/ : SUCCESS in 9m 22s\n- openstack-tox-py37 http://logs.openstack.org/09/668109/2/check/openstack-tox-py37/b52232f/ : SUCCESS in 7m 18s\n- build-openstack-releasenotes http://logs.openstack.org/09/668109/2/check/build-openstack-releasenotes/d7ae3fa/html/ : SUCCESS in 3m 24s\n- networking-sfc-functional http://logs.openstack.org/09/668109/2/check/networking-sfc-functional/845468a/ : SUCCESS in 15m 44s\n- networking-sfc-tempest http://logs.openstack.org/09/668109/2/check/networking-sfc-tempest/82f88a9/ : SUCCESS in 1h 53m 06s\n- networking-sfc-tempest-multinode http://logs.openstack.org/09/668109/2/check/networking-sfc-tempest-multinode/20959e4/ : SUCCESS in 2h 05m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9b0db75a9c83caa71265d11554f19086ff73e825","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-07-05 13:05:17.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"448a286ce96bbce6ba1e714ef06f204bb57a1f48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 15:14:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/09/668109/3/check/openstack-tox-lower-constraints/d110cfb/ : SUCCESS in 7m 51s\n- openstack-tox-pep8 http://logs.openstack.org/09/668109/3/check/openstack-tox-pep8/63f9ab1/ : SUCCESS in 5m 35s\n- openstack-tox-py27 http://logs.openstack.org/09/668109/3/check/openstack-tox-py27/1050eee/ : SUCCESS in 7m 20s\n- openstack-tox-docs http://logs.openstack.org/09/668109/3/check/openstack-tox-docs/7c3ac85/html/ : SUCCESS in 3m 33s\n- openstack-tox-py36 http://logs.openstack.org/09/668109/3/check/openstack-tox-py36/6a7ad1c/ : SUCCESS in 7m 35s\n- openstack-tox-py37 http://logs.openstack.org/09/668109/3/check/openstack-tox-py37/dbdd523/ : SUCCESS in 9m 03s\n- build-openstack-releasenotes http://logs.openstack.org/09/668109/3/check/build-openstack-releasenotes/99c42db/html/ : SUCCESS in 3m 38s\n- networking-sfc-functional http://logs.openstack.org/09/668109/3/check/networking-sfc-functional/788e9c8/ : SUCCESS in 18m 08s\n- networking-sfc-tempest http://logs.openstack.org/09/668109/3/check/networking-sfc-tempest/ea9ff66/ : SUCCESS in 1h 48m 53s\n- networking-sfc-tempest-multinode http://logs.openstack.org/09/668109/3/check/networking-sfc-tempest-multinode/770c2a5/ : SUCCESS in 2h 03m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6125636cd7c378814d3cea17155726c976528e66","author":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"date":"2019-07-05 15:51:09.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"06b6418a01a70217c87842002a0db8922f81940a","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-08 06:53:01.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a3d06e1920ceaf1acd526f0bc68c06ca9e9eb5e5","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-07-25 07:04:30.000000000","message":"Patch Set 3: Code-Review+2\n\n@Manuel Can we merge this patch, as it needed tofix gate job.","accounts_in_message":[],"_revision_number":3},{"id":"740b4066ae9068c224e4cfdb87d38b7501c64a46","author":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"date":"2019-07-25 07:10:06.000000000","message":"Patch Set 3: Workflow+1\n\n\u003e @Manuel Can we merge this patch, as it needed tofix gate job.\n\nyes! I was expecting another core to +2 it but let\u0027s do that","accounts_in_message":[],"_revision_number":3},{"id":"3bf27a598d314fee72df860fc084aaf78cf27a7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 07:10:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1a20ec90727e7c573739ff3e59945161a9bdcafe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 07:35:28.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/09/668109/3/gate/openstack-tox-lower-constraints/c94d927/ : SUCCESS in 9m 17s\n- openstack-tox-pep8 http://logs.openstack.org/09/668109/3/gate/openstack-tox-pep8/6925c19/ : SUCCESS in 6m 00s\n- openstack-tox-py27 http://logs.openstack.org/09/668109/3/gate/openstack-tox-py27/4815a45/ : SUCCESS in 8m 36s\n- openstack-tox-docs http://logs.openstack.org/09/668109/3/gate/openstack-tox-docs/2bf8845/html/ : SUCCESS in 3m 59s\n- openstack-tox-py36 http://logs.openstack.org/09/668109/3/gate/openstack-tox-py36/8a22c01/ : SUCCESS in 7m 13s\n- openstack-tox-py37 http://logs.openstack.org/09/668109/3/gate/openstack-tox-py37/15206e5/ : SUCCESS in 7m 49s\n- build-openstack-releasenotes http://logs.openstack.org/09/668109/3/gate/build-openstack-releasenotes/20bcfe8/html/ : SUCCESS in 3m 37s\n- networking-sfc-functional http://logs.openstack.org/09/668109/3/gate/networking-sfc-functional/9dd24ce/ : FAILURE in 15m 41s\n- networking-sfc-tempest http://logs.openstack.org/09/668109/3/gate/networking-sfc-tempest/8fd6dd8/ : FAILURE in 23m 27s","accounts_in_message":[],"_revision_number":3},{"id":"e42227195a2a17ea69e0b42d75b3ce09e1b33d82","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-07-30 05:12:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"231699ff2f14c556c9eb0eda321ca0b632ea6da0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 07:18:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/09/668109/3/check/openstack-tox-lower-constraints/cc0cd35/ : SUCCESS in 8m 43s\n- openstack-tox-pep8 http://logs.openstack.org/09/668109/3/check/openstack-tox-pep8/ad6d790/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/09/668109/3/check/openstack-tox-py27/3252062/ : SUCCESS in 7m 24s\n- openstack-tox-docs http://logs.openstack.org/09/668109/3/check/openstack-tox-docs/a490a3b/html/ : SUCCESS in 4m 00s\n- openstack-tox-py36 http://logs.openstack.org/09/668109/3/check/openstack-tox-py36/a14b39c/ : SUCCESS in 8m 04s\n- openstack-tox-py37 http://logs.openstack.org/09/668109/3/check/openstack-tox-py37/bafd4a6/ : SUCCESS in 8m 08s\n- build-openstack-releasenotes http://logs.openstack.org/09/668109/3/check/build-openstack-releasenotes/15cfa1e/html/ : SUCCESS in 3m 14s\n- networking-sfc-functional http://logs.openstack.org/09/668109/3/check/networking-sfc-functional/8b21349/ : SUCCESS in 16m 35s\n- networking-sfc-tempest http://logs.openstack.org/09/668109/3/check/networking-sfc-tempest/424ffea/ : FAILURE in 2h 01m 08s\n- networking-sfc-tempest-multinode http://logs.openstack.org/09/668109/3/check/networking-sfc-tempest-multinode/7de2579/ : SUCCESS in 1h 54m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ba4dc9100d65321299c8c52c3663fbd66f5c3568","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-07-30 07:48:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c8bfac8bbb3c5aead27cf053803b3a20e7002ff4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 09:48:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/09/668109/3/check/openstack-tox-lower-constraints/75284e7/ : SUCCESS in 8m 54s\n- openstack-tox-pep8 http://logs.openstack.org/09/668109/3/check/openstack-tox-pep8/e4a4aee/ : SUCCESS in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/09/668109/3/check/openstack-tox-py27/0084029/ : SUCCESS in 7m 42s\n- openstack-tox-docs http://logs.openstack.org/09/668109/3/check/openstack-tox-docs/a9ffea7/html/ : SUCCESS in 5m 29s\n- openstack-tox-py36 http://logs.openstack.org/09/668109/3/check/openstack-tox-py36/25f2e46/ : SUCCESS in 9m 58s\n- openstack-tox-py37 http://logs.openstack.org/09/668109/3/check/openstack-tox-py37/9633eee/ : SUCCESS in 8m 09s\n- build-openstack-releasenotes http://logs.openstack.org/09/668109/3/check/build-openstack-releasenotes/566c1fc/html/ : SUCCESS in 3m 34s\n- networking-sfc-functional http://logs.openstack.org/09/668109/3/check/networking-sfc-functional/65d8f71/ : SUCCESS in 14m 39s\n- networking-sfc-tempest http://logs.openstack.org/09/668109/3/check/networking-sfc-tempest/b0cccfb/ : SUCCESS in 1h 43m 33s\n- networking-sfc-tempest-multinode http://logs.openstack.org/09/668109/3/check/networking-sfc-tempest-multinode/5078d66/ : FAILURE in 1h 52m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5975c7d7d91cd66148ceb701496384bfb0bfe554","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 09:48:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"94f22f1fab04d7bd940d164bfe80373dff5c781e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 11:33:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/09/668109/3/gate/openstack-tox-lower-constraints/4c557c2/ : SUCCESS in 8m 13s\n- openstack-tox-pep8 http://logs.openstack.org/09/668109/3/gate/openstack-tox-pep8/df028ad/ : SUCCESS in 6m 32s\n- openstack-tox-py27 http://logs.openstack.org/09/668109/3/gate/openstack-tox-py27/1314035/ : SUCCESS in 8m 20s\n- openstack-tox-docs http://logs.openstack.org/09/668109/3/gate/openstack-tox-docs/e21e029/html/ : SUCCESS in 3m 43s\n- openstack-tox-py36 http://logs.openstack.org/09/668109/3/gate/openstack-tox-py36/a7b440a/ : SUCCESS in 7m 50s\n- openstack-tox-py37 http://logs.openstack.org/09/668109/3/gate/openstack-tox-py37/b288015/ : SUCCESS in 9m 14s\n- build-openstack-releasenotes http://logs.openstack.org/09/668109/3/gate/build-openstack-releasenotes/4006cfa/html/ : SUCCESS in 5m 06s\n- networking-sfc-functional http://logs.openstack.org/09/668109/3/gate/networking-sfc-functional/f2b903b/ : SUCCESS in 17m 34s\n- networking-sfc-tempest http://logs.openstack.org/09/668109/3/gate/networking-sfc-tempest/3d54678/ : SUCCESS in 1h 38m 49s","accounts_in_message":[],"_revision_number":3},{"id":"74d4e20ce5fd774d774c639f0663823647399251","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 11:33:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"17636f5246a177fd4604aea39f9ec613bc3c1086","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-04 18:47:40.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 817765152f618bd8aa76e23afc5a1c0490005e27","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"315eefc9b55df58060e4760fe90f7f545ef0753d","revisions":{"a050ce8d5cd2d9aabca313157a92bc9c37104be1":{"kind":"REWORK","_number":1,"created":"2019-06-28 07:54:05.000000000","uploader":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"ref":"refs/changes/09/668109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/09/668109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/09/668109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/09/668109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/09/668109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/09/668109/1"}}},"commit":{"parents":[{"commit":"6d62a74e483d64c5d7b22c00010dbad24a1f6696","subject":"Use opendev repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/6d62a74e483d64c5d7b22c00010dbad24a1f6696"}]}],"author":{"name":"dharmendra","email":"dharmendra.kushwaha@india.nec.com","date":"2019-06-28 07:53:24.000000000","tz":0},"committer":{"name":"dharmendra","email":"dharmendra.kushwaha@india.nec.com","date":"2019-06-28 07:54:03.000000000","tz":0},"subject":"DNM: Testing","message":"DNM: Testing\n\nChange-Id: I000436c7bbda82a95f1f190b09053b9179ccaf1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/a050ce8d5cd2d9aabca313157a92bc9c37104be1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/a050ce8d5cd2d9aabca313157a92bc9c37104be1"}]},"branch":"refs/heads/master"},"1620a725d84a64b3eacd3f3b9cd615d587bb6e17":{"kind":"REWORK","_number":2,"created":"2019-07-05 10:22:36.000000000","uploader":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"ref":"refs/changes/09/668109/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/09/668109/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/09/668109/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/09/668109/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/09/668109/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/09/668109/2"}}},"commit":{"parents":[{"commit":"6d62a74e483d64c5d7b22c00010dbad24a1f6696","subject":"Use opendev repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/6d62a74e483d64c5d7b22c00010dbad24a1f6696"}]}],"author":{"name":"dharmendra","email":"dharmendra.kushwaha@india.nec.com","date":"2019-07-05 10:21:59.000000000","tz":0},"committer":{"name":"dharmendra","email":"dharmendra.kushwaha@india.nec.com","date":"2019-07-05 10:21:59.000000000","tz":0},"subject":"DNM: Testing","message":"DNM: Testing\n\nChange-Id: I000436c7bbda82a95f1f190b09053b9179ccaf1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/1620a725d84a64b3eacd3f3b9cd615d587bb6e17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/1620a725d84a64b3eacd3f3b9cd615d587bb6e17"}]},"branch":"refs/heads/master"},"315eefc9b55df58060e4760fe90f7f545ef0753d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-07-05 13:05:17.000000000","uploader":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"ref":"refs/changes/09/668109/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/09/668109/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/09/668109/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/09/668109/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/09/668109/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/09/668109/3"}}},"commit":{"parents":[{"commit":"6d62a74e483d64c5d7b22c00010dbad24a1f6696","subject":"Use opendev repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/6d62a74e483d64c5d7b22c00010dbad24a1f6696"}]}],"author":{"name":"dharmendra","email":"dharmendra.kushwaha@india.nec.com","date":"2019-07-05 10:21:59.000000000","tz":0},"committer":{"name":"dharmendra kushwaha","email":"dharmendra.kushwaha@india.nec.com","date":"2019-07-05 13:05:05.000000000","tz":0},"subject":"Add libpq-dev to bindep.txt to fix gate error","message":"Add libpq-dev to bindep.txt to fix gate error\n\nSpotted log:\nhttp://logs.openstack.org/09/668109/1/check/openstack-tox-py37/af3c478/job-output.txt.gz\n\nChange-Id: I000436c7bbda82a95f1f190b09053b9179ccaf1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/315eefc9b55df58060e4760fe90f7f545ef0753d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/315eefc9b55df58060e4760fe90f7f545ef0753d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
