)]}'
{"id":"openstack%2Fnetworking-sfc~668848","triplet_id":"openstack%2Fnetworking-sfc~master~I799b976d1e8764de344e7dc8efe27767800611e0","project":"openstack/networking-sfc","branch":"master","hashtags":[],"change_id":"I799b976d1e8764de344e7dc8efe27767800611e0","subject":"Update the constraints url","status":"MERGED","created":"2019-07-03 09:40:33.000000000","updated":"2019-07-31 17:46:48.000000000","submitted":"2019-07-31 17:46:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668848-1564595208783-cc82055b","meta_rev_id":"58587c19937e0fe9c5e398972f19755a681fcab7","_number":668848,"virtual_id_number":668848,"owner":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-07-31 17:46:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"recommended":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":1,"date":"2019-07-08 06:52:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-07-05 15:48:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":2,"date":"2019-07-30 12:14:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-19 06:54:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"all":[{"value":0,"date":"2019-07-23 06:08:36.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"date":"2019-07-08 06:26:56.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2019-07-31 15:55:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-08 06:26:56.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-07-08 06:52:42.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-07-19 06:54:36.000000000","updated_by":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"reviewer":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"state":"REVIEWER"},{"updated":"2019-07-23 06:08:36.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-07-31 15:55:18.000000000","updated_by":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"reviewer":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"state":"REVIEWER"},{"updated":"2019-07-31 17:46:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b2d0e6137dd8fb6f0bd4fc978c38bfa88e325d2","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-07-03 09:40:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93ec3a9ac73ec7836f498815bacc86b4ed80ee36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 11:51:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/48/668848/1/check/openstack-tox-lower-constraints/09646cd/ : FAILURE in 27m 11s\n- openstack-tox-pep8 http://logs.openstack.org/48/668848/1/check/openstack-tox-pep8/6fc4358/ : FAILURE in 27m 36s\n- openstack-tox-py27 http://logs.openstack.org/48/668848/1/check/openstack-tox-py27/3a09c25/ : FAILURE in 27m 23s\n- openstack-tox-docs http://logs.openstack.org/48/668848/1/check/openstack-tox-docs/0a5bf69/ : POST_FAILURE in 27m 23s\n- openstack-tox-py36 http://logs.openstack.org/48/668848/1/check/openstack-tox-py36/6bfe417/ : SUCCESS in 7m 21s\n- openstack-tox-py37 http://logs.openstack.org/48/668848/1/check/openstack-tox-py37/d1f7fbd/ : FAILURE in 2m 25s\n- build-openstack-releasenotes http://logs.openstack.org/48/668848/1/check/build-openstack-releasenotes/b7d844e/html/ : SUCCESS in 2m 49s\n- networking-sfc-functional http://logs.openstack.org/48/668848/1/check/networking-sfc-functional/b449610/ : SUCCESS in 17m 19s\n- networking-sfc-tempest http://logs.openstack.org/48/668848/1/check/networking-sfc-tempest/aa3c76f/ : SUCCESS in 1h 52m 44s\n- networking-sfc-tempest-multinode http://logs.openstack.org/48/668848/1/check/networking-sfc-tempest-multinode/c9f49eb/ : SUCCESS in 2h 03m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34fdf2c385c96f65732d36da9cf3f91cb33a48ea","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-07-05 13:25:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e1c64a29f7a2bf8f11527422f91727609ff6c774","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 15:41:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/48/668848/2/check/openstack-tox-lower-constraints/b6af90c/ : SUCCESS in 9m 27s\n- openstack-tox-pep8 http://logs.openstack.org/48/668848/2/check/openstack-tox-pep8/823b25d/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/48/668848/2/check/openstack-tox-py27/856ee4a/ : SUCCESS in 7m 33s\n- openstack-tox-docs http://logs.openstack.org/48/668848/2/check/openstack-tox-docs/9abdd35/html/ : SUCCESS in 4m 24s\n- openstack-tox-py36 http://logs.openstack.org/48/668848/2/check/openstack-tox-py36/9e825e4/ : SUCCESS in 7m 21s\n- openstack-tox-py37 http://logs.openstack.org/48/668848/2/check/openstack-tox-py37/19e9222/ : SUCCESS in 8m 10s\n- build-openstack-releasenotes http://logs.openstack.org/48/668848/2/check/build-openstack-releasenotes/b67991a/html/ : SUCCESS in 3m 12s\n- networking-sfc-functional http://logs.openstack.org/48/668848/2/check/networking-sfc-functional/47abdd7/ : SUCCESS in 16m 50s\n- networking-sfc-tempest http://logs.openstack.org/48/668848/2/check/networking-sfc-tempest/5933b62/ : SUCCESS in 2h 08m 22s\n- networking-sfc-tempest-multinode http://logs.openstack.org/48/668848/2/check/networking-sfc-tempest-multinode/c7975dd/ : SUCCESS in 2h 02m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d737b9f9b290539c88fbc3bdb43f0b2f43e7bd43","author":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"date":"2019-07-05 15:48:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6bc64249d009f74bebee12bb7ff85804cf6bf35d","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-08 06:52:42.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4eddf3621eb8c95feeb5b295fcdb3db74546cfe6","author":{"_account_id":26222,"name":"Cong Phuoc Hoang","email":"hoangphuocbk2.07@gmail.com","username":"hoangphuoc"},"date":"2019-07-19 06:54:36.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d0dc6dbc7f2f3cc3236564875cba38628b4a5a55","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-07-30 12:14:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e1a9e10413f99c38887614c33084430d37ae1c99","author":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"date":"2019-07-31 15:55:18.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"977fbe9aa373dbd0478752ca490bbaccd185fda9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 15:55:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d59f099885ef0ae4f25ef7d39e3bf3149e2eac18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 17:46:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/48/668848/2/gate/openstack-tox-lower-constraints/ffe5ed2/ : SUCCESS in 11m 33s\n- openstack-tox-pep8 http://logs.openstack.org/48/668848/2/gate/openstack-tox-pep8/4a95f02/ : SUCCESS in 7m 03s\n- openstack-tox-py27 http://logs.openstack.org/48/668848/2/gate/openstack-tox-py27/187bbc8/ : SUCCESS in 11m 50s\n- openstack-tox-docs http://logs.openstack.org/48/668848/2/gate/openstack-tox-docs/6f9d985/html/ : SUCCESS in 5m 34s\n- openstack-tox-py36 http://logs.openstack.org/48/668848/2/gate/openstack-tox-py36/b935597/ : SUCCESS in 9m 52s\n- openstack-tox-py37 http://logs.openstack.org/48/668848/2/gate/openstack-tox-py37/2c1a52e/ : SUCCESS in 10m 34s\n- build-openstack-releasenotes http://logs.openstack.org/48/668848/2/gate/build-openstack-releasenotes/3aeb996/html/ : SUCCESS in 4m 22s\n- networking-sfc-functional http://logs.openstack.org/48/668848/2/gate/networking-sfc-functional/080f6c8/ : SUCCESS in 14m 32s\n- networking-sfc-tempest http://logs.openstack.org/48/668848/2/gate/networking-sfc-tempest/7ec9b69/ : SUCCESS in 1h 44m 24s","accounts_in_message":[],"_revision_number":2},{"id":"58587c19937e0fe9c5e398972f19755a681fcab7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 17:46:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7fb02079435f3963c95da3514e3f19d04fdea2e4","revisions":{"22b0e17ccd4409eee87544bb524b1de42a951351":{"kind":"REWORK","_number":1,"created":"2019-07-03 09:40:33.000000000","uploader":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"ref":"refs/changes/48/668848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/48/668848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/48/668848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/48/668848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/48/668848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/48/668848/1"}}},"commit":{"parents":[{"commit":"6d62a74e483d64c5d7b22c00010dbad24a1f6696","subject":"Use opendev repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/6d62a74e483d64c5d7b22c00010dbad24a1f6696"}]}],"author":{"name":"dharmendra","email":"dharmendra.kushwaha@india.nec.com","date":"2019-07-03 09:38:54.000000000","tz":0},"committer":{"name":"dharmendra","email":"dharmendra.kushwaha@india.nec.com","date":"2019-07-03 09:40:31.000000000","tz":0},"subject":"Update the constraints url","message":"Update the constraints url\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: I799b976d1e8764de344e7dc8efe27767800611e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/22b0e17ccd4409eee87544bb524b1de42a951351"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/22b0e17ccd4409eee87544bb524b1de42a951351"}]},"branch":"refs/heads/master"},"7fb02079435f3963c95da3514e3f19d04fdea2e4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-05 13:25:30.000000000","uploader":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"ref":"refs/changes/48/668848/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/48/668848/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/48/668848/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/48/668848/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/48/668848/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/48/668848/2"}}},"commit":{"parents":[{"commit":"315eefc9b55df58060e4760fe90f7f545ef0753d","subject":"Add libpq-dev to bindep.txt to fix gate error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/315eefc9b55df58060e4760fe90f7f545ef0753d"}]}],"author":{"name":"dharmendra","email":"dharmendra.kushwaha@india.nec.com","date":"2019-07-03 09:38:54.000000000","tz":0},"committer":{"name":"dharmendra kushwaha","email":"dharmendra.kushwaha@india.nec.com","date":"2019-07-05 13:25:30.000000000","tz":0},"subject":"Update the constraints url","message":"Update the constraints url\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: I799b976d1e8764de344e7dc8efe27767800611e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/7fb02079435f3963c95da3514e3f19d04fdea2e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/7fb02079435f3963c95da3514e3f19d04fdea2e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
