)]}'
{"id":"openstack%2Fnetworking-sfc~669423","triplet_id":"openstack%2Fnetworking-sfc~master~I22cf36bae9d46c99d16e92e32d45df8a460b7c84","project":"openstack/networking-sfc","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I22cf36bae9d46c99d16e92e32d45df8a460b7c84","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 17:49:52.000000000","updated":"2019-08-14 23:47:26.000000000","submitted":"2019-08-14 23:47:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669423-1565826446208-d21f4b70","meta_rev_id":"c5afc38e97f552ec0286d3d0af506de534858cf6","_number":669423,"virtual_id_number":669423,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-08-14 23:47:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"recommended":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"all":[{"value":1,"date":"2019-08-06 12:16:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-08-14 00:56:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":2,"date":"2019-08-10 12:46:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2019-08-13 14:55:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-08-12 03:33:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2019-08-14 13:22:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-08-14 14:52:58.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":1,"date":"2019-08-14 14:52:58.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 12:16:37.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-08-10 12:46:35.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-08-12 03:33:16.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-08-13 14:55:52.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-08-14 00:56:26.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-08-14 13:22:53.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-08-14 14:52:58.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-08-14 23:47:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2660596e47a27da136f2f8952551844c3117ecc9","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 17:49:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef6a378cf6cf155db4d285628cffe93b9bf70bac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 19:59:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/23/669423/1/check/test-release-openstack/8d94907/ : SUCCESS in 2m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/669423/1/check/openstack-tox-lower-constraints/fbe2b0c/ : SUCCESS in 8m 51s\n- openstack-tox-pep8 http://logs.openstack.org/23/669423/1/check/openstack-tox-pep8/f4953db/ : SUCCESS in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/23/669423/1/check/openstack-tox-py27/f207e6d/ : SUCCESS in 7m 58s\n- openstack-tox-docs http://logs.openstack.org/23/669423/1/check/openstack-tox-docs/8662882/html/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/23/669423/1/check/openstack-tox-py36/8b0341d/ : SUCCESS in 8m 06s\n- openstack-tox-py37 http://logs.openstack.org/23/669423/1/check/openstack-tox-py37/fb31a69/ : FAILURE in 2m 47s\n- build-openstack-releasenotes http://logs.openstack.org/23/669423/1/check/build-openstack-releasenotes/b5da047/html/ : SUCCESS in 3m 25s\n- networking-sfc-functional http://logs.openstack.org/23/669423/1/check/networking-sfc-functional/3227688/ : SUCCESS in 14m 26s\n- networking-sfc-tempest http://logs.openstack.org/23/669423/1/check/networking-sfc-tempest/c0f46bd/ : FAILURE in 1h 38m 54s\n- networking-sfc-tempest-multinode http://logs.openstack.org/23/669423/1/check/networking-sfc-tempest-multinode/ca697a6/ : SUCCESS in 2h 00m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"21721066eaec726ddf94dc583d7b5bb7d66f0bcb","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-07-13 09:52:13.000000000","message":"Patch Set 1:\n\nhttp://logs.openstack.org/16/669416/1/check/openstack-tox-py37/b8234ea/testr_results.html.gz\nError: pg_config executable not found for psycopg2\n\nDo we need psycopg2 for networking-sfc tests?","accounts_in_message":[],"_revision_number":1},{"id":"1a84e04458a3b4a864fb4375d792e13cd14fa37a","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-07-26 20:24:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"04e81358d05f03da60efe44d56b6e3792480d3d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 22:25:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/23/669423/1/check/test-release-openstack/3c0cf90/ : SUCCESS in 3m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/669423/1/check/openstack-tox-lower-constraints/2964291/ : SUCCESS in 9m 35s\n- openstack-tox-pep8 http://logs.openstack.org/23/669423/1/check/openstack-tox-pep8/1e58e85/ : SUCCESS in 8m 16s\n- openstack-tox-py27 http://logs.openstack.org/23/669423/1/check/openstack-tox-py27/e5ee9dd/ : SUCCESS in 6m 47s\n- openstack-tox-docs http://logs.openstack.org/23/669423/1/check/openstack-tox-docs/28b0c15/html/ : SUCCESS in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/23/669423/1/check/openstack-tox-py36/1a3a009/ : SUCCESS in 8m 18s\n- openstack-tox-py37 http://logs.openstack.org/23/669423/1/check/openstack-tox-py37/48f7aeb/ : FAILURE in 3m 00s\n- build-openstack-releasenotes http://logs.openstack.org/23/669423/1/check/build-openstack-releasenotes/a01456d/html/ : SUCCESS in 3m 42s\n- networking-sfc-functional http://logs.openstack.org/23/669423/1/check/networking-sfc-functional/1b3eb66/ : SUCCESS in 14m 51s\n- networking-sfc-tempest http://logs.openstack.org/23/669423/1/check/networking-sfc-tempest/a0344dc/ : FAILURE in 1h 36m 37s\n- networking-sfc-tempest-multinode http://logs.openstack.org/23/669423/1/check/networking-sfc-tempest-multinode/af3b1de/ : SUCCESS in 1h 54m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee169abb17351d2b62f7b117044366cec7bc4743","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-08-01 09:04:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"aa716d16cb1cef92abc65501c5014dcae0e1986a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 10:55:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://logs.opendev.org/23/669423/2/check/test-release-openstack/0beddd0/ : SUCCESS in 1m 58s\n- openstack-tox-lower-constraints https://logs.opendev.org/23/669423/2/check/openstack-tox-lower-constraints/d5840bb/ : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://logs.opendev.org/23/669423/2/check/openstack-tox-pep8/548d27b/ : SUCCESS in 5m 17s\n- openstack-tox-py27 https://logs.opendev.org/23/669423/2/check/openstack-tox-py27/0b80e4f/ : SUCCESS in 8m 39s\n- openstack-tox-docs https://logs.opendev.org/23/669423/2/check/openstack-tox-docs/ad78050/html/ : SUCCESS in 4m 01s\n- openstack-tox-py36 https://logs.opendev.org/23/669423/2/check/openstack-tox-py36/65fa3c5/ : SUCCESS in 7m 30s\n- openstack-tox-py37 https://logs.opendev.org/23/669423/2/check/openstack-tox-py37/6b1bfde/ : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://logs.opendev.org/23/669423/2/check/build-openstack-releasenotes/b6da6f5/html/ : SUCCESS in 3m 47s\n- networking-sfc-functional https://logs.opendev.org/23/669423/2/check/networking-sfc-functional/93a8837/ : SUCCESS in 16m 53s\n- networking-sfc-tempest https://logs.opendev.org/23/669423/2/check/networking-sfc-tempest/1edce58/ : FAILURE in 1h 36m 23s\n- networking-sfc-tempest-multinode https://logs.opendev.org/23/669423/2/check/networking-sfc-tempest-multinode/33688e6/ : SUCCESS in 1h 48m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"76f800a9ec31ad0b93771f6b46a57f093fa243f0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-06 10:00:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d0546f32b7c66afe3be61ef4ab482c8ff2f89e54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 12:11:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://logs.opendev.org/23/669423/2/check/test-release-openstack/656a6ae/ : SUCCESS in 2m 00s\n- openstack-tox-lower-constraints https://logs.opendev.org/23/669423/2/check/openstack-tox-lower-constraints/23d790f/ : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://logs.opendev.org/23/669423/2/check/openstack-tox-pep8/d97f0fe/ : SUCCESS in 5m 48s\n- openstack-tox-py27 https://logs.opendev.org/23/669423/2/check/openstack-tox-py27/a30ccb5/ : SUCCESS in 8m 34s\n- openstack-tox-docs https://logs.opendev.org/23/669423/2/check/openstack-tox-docs/3e326db/html/ : SUCCESS in 4m 20s\n- openstack-tox-py36 https://logs.opendev.org/23/669423/2/check/openstack-tox-py36/0579bce/ : SUCCESS in 8m 26s\n- openstack-tox-py37 https://logs.opendev.org/23/669423/2/check/openstack-tox-py37/40bfdbb/ : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://logs.opendev.org/23/669423/2/check/build-openstack-releasenotes/9fe5a4b/html/ : SUCCESS in 3m 49s\n- networking-sfc-functional https://logs.opendev.org/23/669423/2/check/networking-sfc-functional/f32f173/ : SUCCESS in 16m 01s\n- networking-sfc-tempest https://logs.opendev.org/23/669423/2/check/networking-sfc-tempest/65b07be/ : SUCCESS in 1h 38m 22s\n- networking-sfc-tempest-multinode https://logs.opendev.org/23/669423/2/check/networking-sfc-tempest-multinode/ba8b501/ : SUCCESS in 2h 00m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d2601eb03d8d12a635a0b11abec3d75e1c4e32d5","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-06 12:16:37.000000000","message":"Patch Set 2: Code-Review+1\n\ntempest failure was indeed unrelated (these were UT changes)","accounts_in_message":[],"_revision_number":2},{"id":"fd9cd967c4230229241b88689bf9291bafadd941","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-10 12:46:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4438f2648f190f3a7b12b6f87672bdcca868f6e0","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-08-12 03:33:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"22f14d262ff677c5ded26e24a7d943f9cd951856","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-08-13 14:55:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b32d18b408aa74929b8f75f8c6b562f2c571f266","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-08-14 00:56:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2bfa091338504b2ab62a5a54804172b1cc732098","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-08-14 13:22:53.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d21047e69dc3c37c3b18c51b0e412264b38f34b5","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-08-14 14:52:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6c6acf6b0e309bd5a4ade05c8a40578dd65daaa3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 14:53:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"93ea961b26b664ff4db2ce4b8b682298482e6974","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 16:46:55.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7d04bdb9253b4456a639865be04ba5cb : SUCCESS in 2m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/873c61bf4f7e4ea18ceb8ae4cd8fa3e9 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83c5efc8eb34456fbbadfbbffea9be06 : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8611c65edadf43c49d0dccee41c99818 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0487d05c5572456ead01896106f89d38 : SUCCESS in 3m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/70741ef77e064651b1e628133b707273 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da1437ba3d4c4f6bae8176c9cecc67fb : SUCCESS in 11m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a65ef6ad5a0340eab517cb14ceeb4745 : SUCCESS in 11m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bd71f166f1a423a91c5d7200d052a36 : SUCCESS in 4m 06s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/4792eb712a824110b4621e0d11cd0ee2 : SUCCESS in 19m 21s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/ae32c97628a6458dbd244902d5655063 : FAILURE in 1h 47m 45s","accounts_in_message":[],"_revision_number":2},{"id":"897bcf4d9446bc7b7c957bf1a4ad40728457aa82","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-08-14 19:52:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"217cafdbe750011420418e2bdf37da6d8defae82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 21:59:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cb0764cf38694bcfafb736fe08b33a1f : SUCCESS in 3m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0bc5eafc6ad4235a70a6b91833571a7 : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f24bcab36d04e438222151dffe5625f : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0534d6451bc040d4a4d0aae8130b7234 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e305ecafa944c14b453edc36407bc93 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53f9e542b214436da5ca21a62ca77174 : SUCCESS in 8m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3935fdfe38944e72bb52cc2e8021c0f4 : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bffb303a3a0649bd8dc090346b0f48bd : SUCCESS in 3m 20s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/20ea0328a59e4288a3866a3af2f3296a : SUCCESS in 22m 53s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/86000218dc914fee9f5fcf46c8521eb6 : SUCCESS in 1h 43m 56s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/f710d9c821424ab988b1b68d969eee28 : SUCCESS in 1h 57m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6e065076208d592b60e082f8a0560640e4a483a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 21:59:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7c140bea305f8851ab7d63921d85cbddb2d208f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 23:47:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/10329409f85a4f0cb13315fdf7b85c0a : SUCCESS in 2m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe1b1aeb4f54420b8ec8a110f6d4980d : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d34e76ea4f0e4a3b9b54a0fc8281ba84 : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f10c36e9c472474e932164a5afc733a2 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa3a722d567f4293a2e0f09400057b66 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3bdd46111d54f1dba7f989233f5b7f7 : SUCCESS in 7m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7769cfecf5db47299ad9a8a45c667660 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ec255b62f9043929aca89bcd55d3d98 : SUCCESS in 3m 46s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/4c2aab4068a74a8a8d04f7f3e07246b7 : SUCCESS in 18m 40s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/47af8c909260411b895e4655ab8d96ad : SUCCESS in 1h 41m 36s","accounts_in_message":[],"_revision_number":2},{"id":"c2bd960965d96e11286d939391e33f92683d9090","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 23:47:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"32702b57bd43357b5f6754e6b9d2adff040a47e5","revisions":{"17daf931dd1751556d15e1f70057c7027b0d654b":{"kind":"REWORK","_number":1,"created":"2019-07-05 17:49:52.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/23/669423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/23/669423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/23/669423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/23/669423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/23/669423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/23/669423/1"}}},"commit":{"parents":[{"commit":"6d62a74e483d64c5d7b22c00010dbad24a1f6696","subject":"Use opendev repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/6d62a74e483d64c5d7b22c00010dbad24a1f6696"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:45:57.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 17:49:51.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I22cf36bae9d46c99d16e92e32d45df8a460b7c84\nStory: #2005924\nTask: #34225\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/17daf931dd1751556d15e1f70057c7027b0d654b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/17daf931dd1751556d15e1f70057c7027b0d654b"}]},"branch":"refs/heads/master"},"32702b57bd43357b5f6754e6b9d2adff040a47e5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-01 09:04:59.000000000","uploader":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"ref":"refs/changes/23/669423/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/23/669423/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/23/669423/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/23/669423/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/23/669423/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/23/669423/2"}}},"commit":{"parents":[{"commit":"7fb02079435f3963c95da3514e3f19d04fdea2e4","subject":"Update the constraints url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/7fb02079435f3963c95da3514e3f19d04fdea2e4"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:45:57.000000000","tz":-240},"committer":{"name":"dharmendra kushwaha","email":"dharmendra.kushwaha@india.nec.com","date":"2019-08-01 09:04:59.000000000","tz":0},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I22cf36bae9d46c99d16e92e32d45df8a460b7c84\nStory: #2005924\nTask: #34225\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/32702b57bd43357b5f6754e6b9d2adff040a47e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/32702b57bd43357b5f6754e6b9d2adff040a47e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
