)]}'
{"id":"openstack%2Fnetworking-sfc~673962","triplet_id":"openstack%2Fnetworking-sfc~master~Ib886ffae0317f7d719ead76292f758a732e86521","project":"openstack/networking-sfc","branch":"master","topic":"sphinx","hashtags":[],"change_id":"Ib886ffae0317f7d719ead76292f758a732e86521","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","status":"ABANDONED","created":"2019-08-01 08:59:56.000000000","updated":"2019-12-04 08:36:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1e453d4c2a4cd2dedf4cc760e6bbfee8680d9581","_number":673962,"virtual_id_number":673962,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-09 09:58:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2019-10-18 07:17:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 11:08:29.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-09-14 15:46:31.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-10-09 09:58:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-18 07:17:14.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"44ebb4a62e2c26a286def41b77f3b75393b8e76b","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-08-01 08:59:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32e54c4e97e1b5a6b43160d9d78f032a6de1573e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 09:19:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/62/673962/1/check/openstack-tox-lower-constraints/a5d75c6/ : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://logs.opendev.org/62/673962/1/check/openstack-tox-pep8/d7ac05f/ : SUCCESS in 5m 59s\n- openstack-tox-py27 https://logs.opendev.org/62/673962/1/check/openstack-tox-py27/685c951/ : SUCCESS in 10m 52s\n- openstack-tox-docs https://logs.opendev.org/62/673962/1/check/openstack-tox-docs/95132d8/html/ : SUCCESS in 5m 12s\n- requirements-check https://logs.opendev.org/62/673962/1/check/requirements-check/62ea9d4/ : SUCCESS in 2m 30s\n- openstack-tox-py36 https://logs.opendev.org/62/673962/1/check/openstack-tox-py36/c592d65/ : SUCCESS in 8m 36s\n- openstack-tox-py37 https://logs.opendev.org/62/673962/1/check/openstack-tox-py37/59eea5e/ : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://logs.opendev.org/62/673962/1/check/build-openstack-releasenotes/23d59b8/html/ : SUCCESS in 3m 39s\n- networking-sfc-functional https://logs.opendev.org/62/673962/1/check/networking-sfc-functional/04e5955/ : SUCCESS in 14m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a33e800be76ed7d812587954fca09f67c13e5dfb","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-08-22 11:08:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d72ee3017f3de40a60ec92257ee5cc8045d30e3a","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-09-14 15:46:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1a7e458c6f58dd486c15624df1075efea068a464","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-10-08 06:57:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c033954cb92c4742d7cd0afe102a4054c953f7e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 07:25:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83053e872b234d07b39c2a6e88952e26 : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4d52aea037045c9ae6176ce71baeab3 : SUCCESS in 6m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0b09e382a5f4694bfa923eced3051f5 : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57de5537145b4505be30d2c069594940 : SUCCESS in 7m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2cabd79b25b34b248186f37dffd995dd : FAILURE in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0c03e54add7436fb0b62cfd0b28cce4 : SUCCESS in 9m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/699e170779aa4670a0a5c7a95c2fed38 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d71fbf06b5484858833c9426197306ef : SUCCESS in 3m 35s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/01f60298431740e8badfab4942711a36 : SUCCESS in 25m 31s","accounts_in_message":[],"_revision_number":2},{"id":"5f91b963e88e3ae7769a4d6193649a586ef20762","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-10-09 08:04:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"602cd29d854a547a7f3b0e885d9f8a59d04dd910","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 09:58:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/791ba76b92a349d0a9c86144fb275c6e : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3c943e0536a44109ead185eba2f36e3 : SUCCESS in 5m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aa3ad1b4f99941878e90f02fccca932a : SUCCESS in 2m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f866668534a84de6861289bd67857558 : SUCCESS in 2m 41s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/29757b854a70435a8ff5078361d5d9b6 : FAILURE in 1h 34m 02s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/69ef0616dfc64931a33a254e09437b2b : SUCCESS in 1h 47m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e22f2f2739a431427d396d1c4fb6054f5a104a1d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-18 07:17:14.000000000","message":"Patch Set 3: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8f4e0ac322a9854fa9e43f6068872fb9eebbb598","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 09:19:45.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c5f0ea775d84faeac58cf87c93e5193 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d19fd36dbf9456e91e3f756193f94d6 : SUCCESS in 6m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5e9facd734b9439ea361325e4144412e : SUCCESS in 2m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/125a33002e624a30859dfce5b9671399 : SUCCESS in 2m 47s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/8affed654cd34baabdf27912eaa7a543 : FAILURE in 1h 40m 08s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/76b0ce7de0e44754b42ec59799e77bc0 : SUCCESS in 1h 54m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"22cac3021935ad4023e6e24b0e8570e905cdaa04","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-04 08:36:26.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Jenkins the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6fbc57589379d6cfb4acc9dc449adc27a90da1ea","revisions":{"afb66c796127990641d5d4adc579577c02c96ea7":{"kind":"REWORK","_number":1,"created":"2019-08-01 08:59:56.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/62/673962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/62/673962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/62/673962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/62/673962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/62/673962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/62/673962/1"}}},"commit":{"parents":[{"commit":"7fb02079435f3963c95da3514e3f19d04fdea2e4","subject":"Update the constraints url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/7fb02079435f3963c95da3514e3f19d04fdea2e4"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-01 08:59:26.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-01 09:00:12.000000000","tz":480},"subject":"Blacklist sphinx 2.1.0 (autodoc bug)","message":"Blacklist sphinx 2.1.0 (autodoc bug)\n\nSee https://github.com/sphinx-doc/sphinx/issues/6440 for upstream details\n\nChange-Id: Ib886ffae0317f7d719ead76292f758a732e86521\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/afb66c796127990641d5d4adc579577c02c96ea7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/afb66c796127990641d5d4adc579577c02c96ea7"}]},"branch":"refs/heads/master"},"ddf7da55d833ff0e3b58d8f381ab1704b90a4fe1":{"kind":"REWORK","_number":2,"created":"2019-10-08 06:57:09.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/62/673962/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/62/673962/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/62/673962/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/62/673962/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/62/673962/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/62/673962/2"}}},"commit":{"parents":[{"commit":"4daff0549902a2bb470a5ddbc95e89cf27019b8d","subject":"PDF documentation build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/4daff0549902a2bb470a5ddbc95e89cf27019b8d"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-01 08:59:26.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-10-08 06:57:31.000000000","tz":480},"subject":"Blacklist sphinx 2.1.0 (autodoc bug)","message":"Blacklist sphinx 2.1.0 (autodoc bug)\n\nSee https://github.com/sphinx-doc/sphinx/issues/6440 for upstream details\n\nChange-Id: Ib886ffae0317f7d719ead76292f758a732e86521\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/ddf7da55d833ff0e3b58d8f381ab1704b90a4fe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/ddf7da55d833ff0e3b58d8f381ab1704b90a4fe1"}]},"branch":"refs/heads/master"},"6fbc57589379d6cfb4acc9dc449adc27a90da1ea":{"kind":"REWORK","_number":3,"created":"2019-10-09 08:04:56.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/62/673962/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/62/673962/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/62/673962/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/62/673962/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/62/673962/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/62/673962/3"}}},"commit":{"parents":[{"commit":"4daff0549902a2bb470a5ddbc95e89cf27019b8d","subject":"PDF documentation build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/4daff0549902a2bb470a5ddbc95e89cf27019b8d"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-01 08:59:26.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-10-09 08:02:56.000000000","tz":480},"subject":"Blacklist sphinx 2.1.0 (autodoc bug)","message":"Blacklist sphinx 2.1.0 (autodoc bug)\n\nSee https://github.com/sphinx-doc/sphinx/issues/6440 for upstream details\n\nChange-Id: Ib886ffae0317f7d719ead76292f758a732e86521\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/6fbc57589379d6cfb4acc9dc449adc27a90da1ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/6fbc57589379d6cfb4acc9dc449adc27a90da1ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
