)]}'
{"id":"openstack%2Fnetworking-sfc~696128","triplet_id":"openstack%2Fnetworking-sfc~master~I69c8dfd113269f7092963536498e313c4d7afca1","project":"openstack/networking-sfc","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I69c8dfd113269f7092963536498e313c4d7afca1","subject":"Drop python 2 support and testing","status":"MERGED","created":"2019-11-26 16:38:17.000000000","updated":"2019-12-19 17:47:16.000000000","submitted":"2019-12-19 17:43:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696128-1576777397450-46a14a57","meta_rev_id":"483402e3d39d31e061948876461172a08aa4f8ed","_number":696128,"virtual_id_number":696128,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-19 17:43:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-12-19 16:02:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-12-18 16:31:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-12-19 16:02:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 16:11:18.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-12-18 16:31:33.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-12-19 16:02:31.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-12-19 17:43:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6aad0d85953d3e5c9bf1d5832e916910f4c43a09","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-11-26 16:38:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e28ace699c0c553eeb3c2f463253d4a04c6402a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 18:38:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7bc99c8e06404f99b4f5a772abb3afda : SUCCESS in 2m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5263821b035b492bb47015c1dcec546f : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d228213f9ba34fa2a36800adf65fa32e : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5369c3456927441ba00e05dd7e17d254 : FAILURE in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5b175a7c0884e519db4825ba548ccaf : SUCCESS in 5m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8e81a435950c4a06a5a7a834ad2bb32e : FAILURE in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37b7129bca8e4780843942981a6e4d81 : SUCCESS in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b5d6457e7c10411fba8c96ab17782e46 : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14ea3cb8676642379391e944dfb1528a : SUCCESS in 4m 16s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/630a3eb3f5044e55aca68c1422d054fa : SUCCESS in 15m 38s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/49ce9750e1cd44639c38df8adabf79a2 : SUCCESS in 1h 53m 47s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/0d713f45e332492fa43dd7f7e3807a2c : SUCCESS in 1h 39m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e340a72b0be8576c837dabda240abc88260d3e7b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-11-27 13:58:16.000000000","message":"Patch Set 1:\n\n(1 comment)\n\npy27 test comes from openstack-python-jobs-neutron templates, these should probably be updated first (or this one dependson)","accounts_in_message":[],"_revision_number":1},{"id":"76d391f91d419704cd63cada9c0f891a9e58f6da","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-11-27 13:58:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c6781c507ef7cf8279857e5a201bdd6673e8aa41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 16:18:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/06bde8660d094eee812604bb921ed7fe : SUCCESS in 2m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5c5483350154fc497ffd7c23d9b31c5 : SUCCESS in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dee23b8f94eb4187b5087100a64cf65b : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/395c8c567bf6449b9165d89ba1df2c65 : FAILURE in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31d1212e47124293bb74c83288e16e05 : SUCCESS in 5m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/699a3b6d5adf42d2b253196ff027d9b4 : SUCCESS in 2m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b7079184b5c4a8eaa3efdae05854e46 : SUCCESS in 7m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/340bd9e8c0594f00bfe2a11a4872d8de : SUCCESS in 9m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0b2f4cebcc8406a8d09f0cc2d8ac38d : SUCCESS in 5m 30s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/9c7b486bc0c7499185459782665c9eba : SUCCESS in 22m 15s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/2ee559f920e2451c8e27949a4bbb8c45 : SUCCESS in 2h 04m 41s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/ad31e100bfa34571a55b01b9ca756b59 : SUCCESS in 2h 11m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5ff1ee7626b3a7b8fa1fc4d2bc6a9368619f8dd3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-09 21:46:43.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"bb0eddc18b61ceb4c8a72407ff48cfe36d1409e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 00:02:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/67740dece022493eb8fa78e6363cf180 : SUCCESS in 3m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d135f6e709b142bf8f76c48bc9e27617 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5df75a018223408eb3ce74c3e0924575 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5bbb1b6fc4a450299ac8ca7e9e26bfd : SUCCESS in 7m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/692e5e0fb6c34502851d1a8ab3dda64a : SUCCESS in 8m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cf4c0fc4f314334ab96a2853491b704 : SUCCESS in 9m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/018d64123889456da8748bd348ee17cb : SUCCESS in 2m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/567b0f3761ca4c56acd8f47270e04fb0 : SUCCESS in 3m 23s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/86df574aa6ad41fa889560260108fcc3 : SUCCESS in 17m 39s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/8260a3e52f6c487cbe0d719143efb9b5 : SUCCESS in 1h 41m 42s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/fd3114fa2b8c4a07b6253d4092dfc940 : SUCCESS in 2h 09m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"456ce775a43e6d59b497491d95746f397fcc9211","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-10 09:54:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d0d0b0bd425912c232b29f55d55660e5e81cdf08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 11:56:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a09c6158d8de4e10b80487ebaf055fdc : SUCCESS in 2m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f46a76bcef1428ca8656fc69618a9dd : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7ccdff612c4420fbae2a07eed2e7fd0 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2279c244b512492bbe9c1252bf31165d : SUCCESS in 7m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b00d7de7c5446ed8947727748a963c2 : SUCCESS in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbcf0b27aa524c86ae1901f4415632bc : SUCCESS in 5m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b2cec56b9ce14dd78b6d0394a3f21e44 : SUCCESS in 2m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c766ec83a1464a889813605d4b6fb479 : SUCCESS in 4m 29s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/6f93a4d971d34e7fbcbe0e70eba73f32 : SUCCESS in 15m 31s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/407ac476eac44337bdf5867025ea5a5b : SUCCESS in 1h 48m 04s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/b5e74b2be91548189e95313e01fb32b4 : SUCCESS in 1h 56m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9de4f048072aa3b566ed72fc98c641023679e484","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-16 20:16:59.000000000","message":"Patch Set 4:\n\nAccording to the community-wide goal page [1], we need a release note.\n\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html#completion-criteria","accounts_in_message":[],"_revision_number":4},{"id":"c763495e09e7f62749dd63814f6125281f9a05f6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-17 13:53:20.000000000","message":"Patch Set 4:\n\nThanks Akihiro! Adding the release note (and to other repos I have patches in progress too)","accounts_in_message":[],"_revision_number":4},{"id":"123ba8f2857da9091db49fe5655d775dc45ae038","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-17 13:57:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"25368cd288868ab27c7e6b993eed6ea669364293","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-17 13:58:45.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5c707e9bfaa32f422a34521e6f83733a232ec181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 15:57:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/97055e902dd64ecca577f8e7ccc0df24 : SUCCESS in 3m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5075fd7b2c4243398b03c3f934b2e307 : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7aa8515929940b1bedda79fddb3e20d : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c2f4d35a4c04f628fbd6493732780a3 : SUCCESS in 10m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/99b5fbf19e5b44ee920ad163177e2df9 : SUCCESS in 12m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40da24fa5a53491d81186ed30222b859 : SUCCESS in 6m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a9ffa9ca2bd44b2da434cedd393ea6a3 : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d87653388a9e4f49af96391183df8975 : SUCCESS in 4m 48s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/75e180a31d4a4dc78cbc5359bc7955f1 : SUCCESS in 17m 14s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/4ed0541ccf844ee99d0fca49645e6bfa : SUCCESS in 1h 41m 10s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/cc05d051a7934f27bcc498e5ee208428 : SUCCESS in 1h 54m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9f60ec33ede37eb58e9ceb023ea3df340fe944ea","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-12-17 16:11:18.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"5cb9d5ebc33934240f5ea40f9a807b62e9901ac5","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-18 09:52:38.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d10e90dd778d58f4992fd7a3fe7f4f76ff34b5ba","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-18 09:55:45.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fba18505b131a50938ae79acad79d31b6abf69bf","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-18 14:03:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4b0b6511a3fd1a7f0f9d738ab0f37752a1ccac65","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-18 14:04:21.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"67969da7c4cb202c3dfaec9e243a523c397fe139","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 16:05:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/60eee1256c524fbb8ad323d6de5bacf3 : SUCCESS in 2m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78d0fc4d8d0d4ef8bd7606a44e840985 : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56b10fc65bc24a90b44b6a6f69fbc985 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14c680ff7e81467087b69b971b00a0dd : SUCCESS in 8m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/28ad0e42d6e643caba81eadea53e945d : SUCCESS in 9m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf1c9d7e847d42a28681f6bb4394c428 : SUCCESS in 7m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d2971a6d3ef94918b0924d03f26029b6 : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/313f57907eb7477fb97fb195454d6405 : SUCCESS in 3m 39s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/f7274ada12204a4bba1dc22690e240e7 : SUCCESS in 22m 47s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/1c1e0f7291e04076a76de48fff33f812 : SUCCESS in 1h 48m 02s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/77950f5b73dd491285929cb240ee6304 : SUCCESS in 1h 54m 16s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c71b8814546daf13a3d08f1025f94243d450439f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-18 16:31:33.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"d7c5ba8eccf71f1fd8492cf13009ed3319efea6f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-19 16:02:31.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"1d84e56dadadab306e9856915decb8ad79bf8c26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 16:02:45.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"562c816fd0998452e3630c2e9ce023c1b4514636","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 17:43:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"f64d242359726d97902171bd557c6676a7dea2ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 17:43:17.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ee3edd4cff9b4a3ab08e66c4c32747de : SUCCESS in 2m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e36eed88f6e44f9d959f365e7757de1b : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94d16837fcc94abfaf23b0b1e8cd0704 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5a1152c329349938ebad09a993b8795 : SUCCESS in 8m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a91c4e3488b47f6a26833fc55fdc55c : SUCCESS in 9m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3684cd63aae0483faa99ab2696f8f7b2 : SUCCESS in 6m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a1f5affdbc194bf9b97f14d438ca9841 : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cef3fd9bc43d43b09c433f78666ba9f0 : SUCCESS in 3m 47s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/6749d139f4354d3987a892649a090599 : SUCCESS in 18m 24s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/01dba4c374d94537afd291f392cda7d0 : SUCCESS in 1h 38m 34s","accounts_in_message":[],"_revision_number":7},{"id":"44ca201a8c0cb386a086e16de52b600b5167567b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 17:47:16.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4187b3678e884268adaead82c92a1136 : SUCCESS in 1m 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d45b49d436d2454cbdc2e3565c35af90 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"84175687fb5c23fdf76322b18392c783ca8d6d73","revisions":{"c49ebb19aaf7d03af39cde0dcdbb95e7843406ef":{"kind":"REWORK","_number":1,"created":"2019-11-26 16:38:17.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/28/696128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/28/696128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/1"}}},"commit":{"parents":[{"commit":"5417e6cf8d25ba7ea760710560a9ccb6507f2dbb","subject":"Switch to Ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/5417e6cf8d25ba7ea760710560a9ccb6507f2dbb"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-11-26 16:32:14.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-11-26 16:32:14.000000000","tz":60},"subject":"Drop python 2 support and testing","message":"Drop python 2 support and testing\n\nThis is part of Ussuri goal tracked for Neutron stadium at:\nhttps://etherpad.openstack.org/p/neutron-train-zuulv3-py27drop\n\nChange-Id: I69c8dfd113269f7092963536498e313c4d7afca1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/c49ebb19aaf7d03af39cde0dcdbb95e7843406ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/c49ebb19aaf7d03af39cde0dcdbb95e7843406ef"}]},"branch":"refs/heads/master"},"16969c24f3815a896fc20f415fd573f65a553b47":{"kind":"REWORK","_number":2,"created":"2019-11-27 13:58:36.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/28/696128/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/28/696128/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/2"}}},"commit":{"parents":[{"commit":"5417e6cf8d25ba7ea760710560a9ccb6507f2dbb","subject":"Switch to Ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/5417e6cf8d25ba7ea760710560a9ccb6507f2dbb"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-11-26 16:32:14.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-11-27 13:58:26.000000000","tz":60},"subject":"Drop python 2 support and testing","message":"Drop python 2 support and testing\n\nThis is part of Ussuri goal tracked for Neutron stadium at:\nhttps://etherpad.openstack.org/p/neutron-train-zuulv3-py27drop\n\nChange-Id: I69c8dfd113269f7092963536498e313c4d7afca1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/16969c24f3815a896fc20f415fd573f65a553b47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/16969c24f3815a896fc20f415fd573f65a553b47"}]},"branch":"refs/heads/master"},"8c481d687887ca770af8cddb22dc783b1fc86e9b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-12-09 21:46:43.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/28/696128/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/28/696128/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/3"}}},"commit":{"parents":[{"commit":"5417e6cf8d25ba7ea760710560a9ccb6507f2dbb","subject":"Switch to Ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/5417e6cf8d25ba7ea760710560a9ccb6507f2dbb"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-11-26 16:32:14.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-12-09 21:46:11.000000000","tz":60},"subject":"Drop python 2 support and testing","message":"Drop python 2 support and testing\n\nThis is part of Ussuri goal tracked for Neutron stadium at:\nhttps://etherpad.openstack.org/p/neutron-train-zuulv3-py27drop\n\nChange-Id: I69c8dfd113269f7092963536498e313c4d7afca1\nDepends-On: https://review.opendev.org/698095\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/8c481d687887ca770af8cddb22dc783b1fc86e9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/8c481d687887ca770af8cddb22dc783b1fc86e9b"}]},"branch":"refs/heads/master"},"87b91ee618ae8e774f3477e9c47ed39d92e12d9a":{"kind":"REWORK","_number":4,"created":"2019-12-10 09:54:01.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/28/696128/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/28/696128/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/4"}}},"commit":{"parents":[{"commit":"5417e6cf8d25ba7ea760710560a9ccb6507f2dbb","subject":"Switch to Ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/5417e6cf8d25ba7ea760710560a9ccb6507f2dbb"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-11-26 16:32:14.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-12-10 09:53:38.000000000","tz":60},"subject":"Drop python 2 support and testing","message":"Drop python 2 support and testing\n\nThis is part of Ussuri goal tracked for Neutron stadium at:\nhttps://etherpad.openstack.org/p/neutron-train-zuulv3-py27drop\n\nChange-Id: I69c8dfd113269f7092963536498e313c4d7afca1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/87b91ee618ae8e774f3477e9c47ed39d92e12d9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/87b91ee618ae8e774f3477e9c47ed39d92e12d9a"}]},"branch":"refs/heads/master"},"7df82864df5d904d73404b598f5fcfb3f4fd1d17":{"kind":"REWORK","_number":5,"created":"2019-12-17 13:57:12.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/28/696128/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/28/696128/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/5"}}},"commit":{"parents":[{"commit":"5417e6cf8d25ba7ea760710560a9ccb6507f2dbb","subject":"Switch to Ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/5417e6cf8d25ba7ea760710560a9ccb6507f2dbb"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-11-26 16:32:14.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-12-17 13:51:24.000000000","tz":60},"subject":"Drop python 2 support and testing","message":"Drop python 2 support and testing\n\nThis is part of Ussuri goal tracked for Neutron stadium at:\nhttps://etherpad.openstack.org/p/neutron-train-zuulv3-py27drop\n\nChange-Id: I69c8dfd113269f7092963536498e313c4d7afca1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/7df82864df5d904d73404b598f5fcfb3f4fd1d17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/7df82864df5d904d73404b598f5fcfb3f4fd1d17"}]},"branch":"refs/heads/master"},"8e1e17a7b33c36b44d330c329c78c09fecd0aa4e":{"kind":"REWORK","_number":6,"created":"2019-12-17 13:58:45.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/28/696128/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/28/696128/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/6"}}},"commit":{"parents":[{"commit":"5417e6cf8d25ba7ea760710560a9ccb6507f2dbb","subject":"Switch to Ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/5417e6cf8d25ba7ea760710560a9ccb6507f2dbb"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-11-26 16:32:14.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-12-17 13:58:37.000000000","tz":60},"subject":"Drop python 2 support and testing","message":"Drop python 2 support and testing\n\nThis is part of Ussuri goal tracked for Neutron stadium at:\nhttps://etherpad.openstack.org/p/neutron-train-zuulv3-py27drop\n\nChange-Id: I69c8dfd113269f7092963536498e313c4d7afca1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/8e1e17a7b33c36b44d330c329c78c09fecd0aa4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/8e1e17a7b33c36b44d330c329c78c09fecd0aa4e"}]},"branch":"refs/heads/master"},"84175687fb5c23fdf76322b18392c783ca8d6d73":{"kind":"REWORK","_number":7,"created":"2019-12-18 14:04:21.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/28/696128/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/28/696128/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/28/696128/7"}}},"commit":{"parents":[{"commit":"5417e6cf8d25ba7ea760710560a9ccb6507f2dbb","subject":"Switch to Ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/5417e6cf8d25ba7ea760710560a9ccb6507f2dbb"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-11-26 16:32:14.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-12-18 14:04:12.000000000","tz":60},"subject":"Drop python 2 support and testing","message":"Drop python 2 support and testing\n\nThis is part of Ussuri goal tracked for Neutron stadium at:\nhttps://etherpad.openstack.org/p/neutron-train-zuulv3-py27drop\n\nChange-Id: I69c8dfd113269f7092963536498e313c4d7afca1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/84175687fb5c23fdf76322b18392c783ca8d6d73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/84175687fb5c23fdf76322b18392c783ca8d6d73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
