)]}'
{"id":"openstack%2Fnetworking-sfc~697629","triplet_id":"openstack%2Fnetworking-sfc~master~Id86828d5ea6f43a7ea72074f78e9d85663be67bc","project":"openstack/networking-sfc","branch":"master","topic":"bug/1855410","hashtags":[],"change_id":"Id86828d5ea6f43a7ea72074f78e9d85663be67bc","subject":"port pair leave some parameter no default value set","status":"ABANDONED","created":"2019-12-06 08:15:30.000000000","updated":"2023-02-28 10:57:22.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3294ef64b992267d9b694df792f455aeb1ac8373","_number":697629,"virtual_id_number":697629,"owner":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-04-03 00:57:30.000000000","_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-04-02 14:34:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},{"value":1,"date":"2020-09-13 17:40:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":1,"date":"2020-04-03 22:09:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-04-06 09:59:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-04-06 10:44:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 14:34:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-03 22:09:21.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-04-06 09:59:46.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-04-06 10:44:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-13 17:40:02.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"}],"messages":[{"id":"efb3a200299b4f562ef92f04adf7ddad128a068f","author":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"date":"2019-12-06 08:15:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a18a2aabcc0c9b7c14f834a1a19ad8249d22d390","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 10:03:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac78e2ccca6a43fa81591d54f8226801 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e53408906046410eb23c2e97ae765c84 : FAILURE in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b01903081194c6c82778b19417767ae : SUCCESS in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b26cd2d44fa482ca0d6f0db6afb11ed : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c513d854b55a492db31a7c940936549d : SUCCESS in 7m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c10aa15e1d6942e9978f2b39ef1cce47 : SUCCESS in 7m 11s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/6dff690ec1c94ba59d8b4ffd4f2119a7 : SUCCESS in 14m 52s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/eda3181dcd3e4ec5a88cca9c2487265a : SUCCESS in 1h 35m 08s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/6274bc7b2d634bd4ac0bf6bc455fb818 : SUCCESS in 1h 46m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a29408a2309b95f77908796177fdfaf22d431bfa","author":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"date":"2019-12-06 10:24:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bbfd75a3298a872a4ed2f9b14b511959c13dcce5","author":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"date":"2019-12-06 10:25:31.000000000","message":"Topic set to bug/1855410","accounts_in_message":[],"_revision_number":2},{"id":"140a4d1d41ad5b7bb9bd8022ed969fbdfaec77c2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-06 10:38:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b37a59bc5f4fbea587a55c87b75d8b2fc1ca614b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 12:33:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b8542dabca14250962c6e04fd66e6be : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f5d4458f5f342b1818da217c6af937e : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b3b5a47858fc4f0db227955dd9ffe97a : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4be5c63dc79b43e0aca592b3ce0cdef9 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a5e86e74a474cee9c5539095acf9ae6 : SUCCESS in 7m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b7c180103e84df1a3d70cb2851b5019 : SUCCESS in 7m 49s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/772151298f424689bb6b11656dc6b0cc : SUCCESS in 12m 10s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/764d833b9db44516b655da95c7eaefe1 : SUCCESS in 1h 44m 06s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/d6b9ec265b404f7bb088ef1fd8cd59d0 : SUCCESS in 1h 59m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"38dc1ae0ab8cbb2cb5dffe86e28428c6fd28c936","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-07 11:44:48.000000000","message":"Patch Set 2: Code-Review-1\n\nIMO You should add tempest api test to cover such case. It can be added in https://github.com/openstack/neutron-tempest-plugin/blob/master/neutron_tempest_plugin/sfc/tests/api/test_sfc_extensions.py\nCan You add such test and make this new neutron_tempest_plugin patch to be \"Depends-On\" on this one? Than we will be able to check if this change actually works as expected.","accounts_in_message":[],"_revision_number":2},{"id":"2c7db7aa46fceb070cfc77fcf861578e51a64275","author":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"date":"2020-03-22 11:08:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4c2ca7ecfff8e32fedb547b920863d34df6140e7","author":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"date":"2020-03-22 11:09:32.000000000","message":"Patch Set 3:\n\n\u003e IMO You should add tempest api test to cover such case. It can be\n \u003e added in https://github.com/openstack/neutron-tempest-plugin/blob/master/neutron_tempest_plugin/sfc/tests/api/test_sfc_extensions.py\n \u003e Can You add such test and make this new neutron_tempest_plugin\n \u003e patch to be \"Depends-On\" on this one? Than we will be able to check\n \u003e if this change actually works as expected.\n\nDone, pls review again，thanks.","accounts_in_message":[],"_revision_number":3},{"id":"2685abf2c26a3c12ffae55cf206d869acff5ca70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 13:06:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f00704ca3d3242a8ad2d9edc641cf4b0 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca7d957eeafb4c85a63c4346ea558d72 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a87ca2861c34a6ab4734ff8dd3ac51b : SUCCESS in 6m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d307f0c63e844a068b2b2dee003d3c94 : SUCCESS in 9m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57b9118ddbe34e63a18910c10ad0feaa : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dded715512704865a80cbb1584998cef : SUCCESS in 4m 57s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/edf017d400b84f0da91446b94db9de47 : SUCCESS in 17m 30s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/f8542217cc0e4b3f9b2b6161f1ceed26 : SUCCESS in 1h 39m 11s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/4117dd592a0e4cefa2925328391cac02 : SUCCESS in 1h 53m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c098242cd12d3b4a94668f94aa3a3677696684dc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-30 09:17:14.000000000","message":"Patch Set 3: Code-Review+1\n\nCan you add related tests [1]?\n\nElse LGTM in the current state.\n\n[1] https://github.com/openstack/networking-sfc/blob/master/networking_sfc/tests/unit/extensions/test_sfc.py","accounts_in_message":[],"_revision_number":3},{"id":"50aff05b48dd87925af8602924e8d7dca03fff6a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-01 08:11:35.000000000","message":"Patch Set 3: Code-Review-1\n\n\u003e Can you add related tests [1]?\n \u003e \n \u003e Else LGTM in the current state.\n \u003e \n \u003e [1] https://github.com/openstack/networking-sfc/blob/master/networking_sfc/tests/unit/extensions/test_sfc.py\n\nI agree, simple UT would be good to have","accounts_in_message":[],"_revision_number":3},{"id":"fde2ce826927cfc5b2d37fce5bebb17650cc9e69","author":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"date":"2020-04-02 12:38:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"07c14d791852485cbffb44aefd32732f10bf853d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 14:34:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09457f71e5864a3fb2baec3f9b3fee37 : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/943afd8682e843f5b35a0d2c2340cc59 : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bca6d783ffe2486fb0943a805eeddb1d : SUCCESS in 9m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eebb8868e17f455fb5e1ddd7ea87df93 : SUCCESS in 11m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac03ba834be340e6bdf409a69a80f86d : SUCCESS in 10m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80c9bec96b0d41e8a2feda68f76e0385 : SUCCESS in 10m 33s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/dcd9d1bf814443418d3a9a2c367ea089 : SUCCESS in 24m 57s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/bd2e6ade5ffc409c88dd71e1f48076cc : SUCCESS in 1h 31m 57s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/17d086283c2245e38369c118b3fa6781 : SUCCESS in 1h 50m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"13f1272f28f1a4bdbb64fc350a09d73c3c090274","author":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"date":"2020-04-03 00:57:30.000000000","message":"Patch Set 4:\n\n\u003e \u003e Can you add related tests [1]?\n \u003e \u003e\n \u003e \u003e Else LGTM in the current state.\n \u003e \u003e\n \u003e \u003e [1] https://github.com/openstack/networking-sfc/blob/master/networking_sfc/tests/unit/extensions/test_sfc.py\n \u003e \n \u003e I agree, simple UT would be good to have\n\nDone","accounts_in_message":[],"_revision_number":4},{"id":"7826c8bf12cd0e5e14890d613dc16a3ff6e21362","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-03 22:09:21.000000000","message":"Patch Set 4: Code-Review+1\n\nChange looks good to me indeed for \"partial\" parameters","accounts_in_message":[],"_revision_number":4},{"id":"ade9d28ee1b45689df3e67150fc15c628c33bb5a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-06 09:59:46.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9b58dca205b99e2951a82f5644b0fdbd2bae78cc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-06 10:44:30.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"ed255252a9a8ff9e8c536bbeca81f9ceec921a86","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-13 17:40:02.000000000","message":"Patch Set 4: Code-Review+1\n\nLGTM, but please revise it as Slawek suggested.","accounts_in_message":[],"_revision_number":4},{"id":"3294ef64b992267d9b694df792f455aeb1ac8373","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-28 10:57:22.000000000","message":"Abandoned\n\nIf needed, feel free to restore the patch","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2e59dde33b3035965d74b71a23dd4fb7ef106286","revisions":{"f17cd4ed55dad563863f5ae0c25b9ad9d64da519":{"kind":"REWORK","_number":1,"created":"2019-12-06 08:15:30.000000000","uploader":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"ref":"refs/changes/29/697629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/29/697629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/29/697629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/29/697629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/29/697629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/29/697629/1"}}},"commit":{"parents":[{"commit":"5417e6cf8d25ba7ea760710560a9ccb6507f2dbb","subject":"Switch to Ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/5417e6cf8d25ba7ea760710560a9ccb6507f2dbb"}]}],"author":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2019-12-06 08:07:13.000000000","tz":480},"committer":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2019-12-06 08:15:21.000000000","tz":480},"subject":"port pair leave some parameter no default value set","message":"port pair leave some parameter no default value set\n\nCloses-Bug: #1855410\n\ncreate port pair parameter --service-function-parameters is a dict,\nif it short of some keys, function normalize_sf_parameters will not\nset default value for those keys. for example, if only set weight key,\nport pair will has no correlation which will furter lead to port pair\ngroup create failed.\n\nChange-Id: Id86828d5ea6f43a7ea72074f78e9d85663be67bc\nSigned-off-by: Wei Hui \u003chuiweics@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f17cd4ed55dad563863f5ae0c25b9ad9d64da519"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f17cd4ed55dad563863f5ae0c25b9ad9d64da519"}]},"branch":"refs/heads/master"},"8b7f5030ee07c2de31ac4d357a818c80f17e49ab":{"kind":"REWORK","_number":2,"created":"2019-12-06 10:24:08.000000000","uploader":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"ref":"refs/changes/29/697629/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/29/697629/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/29/697629/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/29/697629/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/29/697629/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/29/697629/2"}}},"commit":{"parents":[{"commit":"5417e6cf8d25ba7ea760710560a9ccb6507f2dbb","subject":"Switch to Ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/5417e6cf8d25ba7ea760710560a9ccb6507f2dbb"}]}],"author":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2019-12-06 08:07:13.000000000","tz":480},"committer":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2019-12-06 10:23:51.000000000","tz":480},"subject":"port pair leave some parameter no default value set","message":"port pair leave some parameter no default value set\n\nCloses-Bug: #1855410\n\ncreate port pair parameter --service-function-parameters is a dict,\nif it short of some keys, function normalize_sf_parameters will not\nset default value for those keys. for example, if only set weight key,\nport pair will has no correlation which will furter lead to port pair\ngroup create failed.\n\nChange-Id: Id86828d5ea6f43a7ea72074f78e9d85663be67bc\nSigned-off-by: Wei Hui \u003chuiweics@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/8b7f5030ee07c2de31ac4d357a818c80f17e49ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/8b7f5030ee07c2de31ac4d357a818c80f17e49ab"}]},"branch":"refs/heads/master"},"e2f7206f5c3c2d24c9723b50130e64a369f6dcee":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2020-03-22 11:08:13.000000000","uploader":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"ref":"refs/changes/29/697629/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/29/697629/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/29/697629/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/29/697629/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/29/697629/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/29/697629/3"}}},"commit":{"parents":[{"commit":"84175687fb5c23fdf76322b18392c783ca8d6d73","subject":"Drop python 2 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/84175687fb5c23fdf76322b18392c783ca8d6d73"}]}],"author":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2020-03-22 11:05:43.000000000","tz":480},"committer":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2020-03-22 11:05:47.000000000","tz":480},"subject":"port pair leave some parameter no default value set","message":"port pair leave some parameter no default value set\n\nCloses-Bug: #1855410\nTempest tests: https://review.opendev.org/#/c/714286/\n\ncreate port pair parameter --service-function-parameters is a dict,\nif it short of some keys, function normalize_sf_parameters will not\nset default value for those keys. for example, if only set weight key,\nport pair will has no correlation which will furter lead to port pair\ngroup create failed.\n\nChange-Id: Id86828d5ea6f43a7ea72074f78e9d85663be67bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/e2f7206f5c3c2d24c9723b50130e64a369f6dcee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/e2f7206f5c3c2d24c9723b50130e64a369f6dcee"}]},"branch":"refs/heads/master"},"2e59dde33b3035965d74b71a23dd4fb7ef106286":{"kind":"REWORK","_number":4,"created":"2020-04-02 12:38:46.000000000","uploader":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"ref":"refs/changes/29/697629/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/29/697629/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/29/697629/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/29/697629/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/29/697629/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/29/697629/4"}}},"commit":{"parents":[{"commit":"84175687fb5c23fdf76322b18392c783ca8d6d73","subject":"Drop python 2 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/84175687fb5c23fdf76322b18392c783ca8d6d73"}]}],"author":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2020-03-22 11:05:43.000000000","tz":480},"committer":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2020-04-02 12:38:30.000000000","tz":480},"subject":"port pair leave some parameter no default value set","message":"port pair leave some parameter no default value set\n\nCloses-Bug: #1855410\nTempest tests: https://review.opendev.org/#/c/714286/\n\ncreate port pair parameter --service-function-parameters is a dict,\nif it short of some keys, function normalize_sf_parameters will not\nset default value for those keys. for example, if only set weight key,\nport pair will has no correlation which will furter lead to port pair\ngroup create failed.\n\nChange-Id: Id86828d5ea6f43a7ea72074f78e9d85663be67bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/2e59dde33b3035965d74b71a23dd4fb7ef106286"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/2e59dde33b3035965d74b71a23dd4fb7ef106286"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
