)]}'
{"id":"openstack%2Fnetworking-sfc~700800","triplet_id":"openstack%2Fnetworking-sfc~master~I4c2b85d07ee181f92c6e766fea9b5e9a727bd513","project":"openstack/networking-sfc","branch":"master","topic":"bug/1857954","hashtags":[],"change_id":"I4c2b85d07ee181f92c6e766fea9b5e9a727bd513","subject":"encap(nsh) encap(ethternet) does not set outter mac","status":"ABANDONED","created":"2019-12-31 06:44:55.000000000","updated":"2023-02-28 10:57:30.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d3389665c2dc40867c37bf11d74925e5cddd41c5","_number":700800,"virtual_id_number":700800,"owner":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-04-03 11:06:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},{"value":1,"date":"2020-09-07 19:23:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":2,"date":"2020-08-14 06:51:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-02 01:31:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-03 11:06:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-14 06:51:34.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-09-07 19:23:21.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"}],"messages":[{"id":"07712bec6c20d5981b970b5cc6e445bdab53c380","author":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"date":"2019-12-31 06:44:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9eb6cfeb456a26438d56e81f60c6663cea1350d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 08:48:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/030bdabd62654df7b9d6e32d46ce11db : FAILURE in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a2fc88ee8684b5989ff174ee72def2e : FAILURE in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a720defc4cc24893b5474f316d23bb95 : FAILURE in 7m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60b66e662a664975b2e00a9b7ec27abe : FAILURE in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b43c5af4656d4203947ed0329537301a : SUCCESS in 5m 56s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/4f68d0e7f69e4eca85c182b6aa068170 : SUCCESS in 17m 00s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/5173e22c96c04d8686edae24a05dc67d : SUCCESS in 1h 43m 30s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/c00439cb083e4041b7f1e8aafd2361f4 : SUCCESS in 1h 51m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f606f563a57704bad023a7b6e3b9ba9c1b953d79","author":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"date":"2019-12-31 13:06:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5f3888972ca3b41dc03632f76ecb74d9e5429254","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 14:56:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1066ab0fdc424f1398776d04ff111fcb : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab1170dad951497dab83e471e1bcdc7b : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f537fd766bc4a689fd735613b534f4e : SUCCESS in 7m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2af5efd7067e48e8a8727150f84165d9 : SUCCESS in 8m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86e6ff9749634c5d918b074b5c4300ff : SUCCESS in 5m 54s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/4b79eba98859458c8a672e6ec64930b8 : SUCCESS in 14m 46s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/ab7ea402f6b74d4fbf4f489c3904fad7 : FAILURE in 38m 59s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/f643b58fafd846e0898d61ed9097348b : SUCCESS in 1h 48m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1a49ac546e15fa07389a2f6c17fa2186c02bb699","author":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"date":"2020-01-01 01:37:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a0a9deb3f0b822af61815bf4a01e12af024d26c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-01 03:38:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bb822d3b462438b85da654da1ad1a86 : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a36c58afa4d9429693adc619de34c32d : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05de2611773d4e4491da24e655d7ad70 : SUCCESS in 6m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c8e833940f9c4cb9b2e9a3cb8e6530ff : SUCCESS in 9m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae00fad847ce494d950a6bff610de9bd : SUCCESS in 6m 57s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/7f6a5707c50d418797728a177157be7e : SUCCESS in 17m 45s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/7bda1337f7b14afcb67a1b34fe874076 : SUCCESS in 1h 45m 21s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/d2a8e30c197e4b229af72fc5a55c081d : SUCCESS in 2h 00m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"42b6f5f4614919e44796480720c38f894543f51b","author":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"date":"2020-04-03 08:42:19.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d8539510728c9f872d081bdd20d9dcc9e8c2510e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 11:06:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f703042aca0940459507ecd10ffa63ab : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24558f9f1bde47a79b5b2c094e623d90 : SUCCESS in 9m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6da24c8c71b14659913381b8630b1f12 : SUCCESS in 11m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb492e7480a44c3eacf21ef7c4b475a6 : SUCCESS in 9m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/672dd0112a254dd3bc1a7314ba7b7167 : SUCCESS in 11m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/242c6ffba1fe443db369105e299f7650 : SUCCESS in 7m 41s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/49b0b0d6a7f24d1ba13c440ca10f76d5 : SUCCESS in 20m 31s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/850b1d92355c4635ad8cc3fe9fbbe723 : SUCCESS in 1h 52m 25s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/ab1b26127f9e43719719dce36f691e65 : SUCCESS in 1h 57m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c33bef92c80909e07433786ea9fe7f4445841260","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-14 06:51:34.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"feb19bfb61235df2c77c070f86567f1b03dd92ce","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-07 19:23:21.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d3389665c2dc40867c37bf11d74925e5cddd41c5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-28 10:57:30.000000000","message":"Abandoned\n\nIf needed, feel free to restore the patch","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"130a0d2934b8a340c04e75761547e021720822ee","revisions":{"7140813cc80f89eea1eee143832fb5ecbb80833c":{"kind":"REWORK","_number":1,"created":"2019-12-31 06:44:55.000000000","uploader":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"ref":"refs/changes/00/700800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/00/700800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/00/700800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/00/700800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/00/700800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/00/700800/1"}}},"commit":{"parents":[{"commit":"84175687fb5c23fdf76322b18392c783ca8d6d73","subject":"Drop python 2 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/84175687fb5c23fdf76322b18392c783ca8d6d73"}]}],"author":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2019-12-31 06:40:21.000000000","tz":480},"committer":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2019-12-31 06:44:28.000000000","tz":480},"subject":"encap(nsh) encap(ethternet) does not set outter mac","message":"encap(nsh) encap(ethternet) does not set outter mac\n\nCloses-Bug: #1857954\n\nChange-Id: I4c2b85d07ee181f92c6e766fea9b5e9a727bd513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/7140813cc80f89eea1eee143832fb5ecbb80833c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/7140813cc80f89eea1eee143832fb5ecbb80833c"}]},"branch":"refs/heads/master"},"19336711494c7507b32624b6df6783245753f652":{"kind":"REWORK","_number":2,"created":"2019-12-31 13:06:17.000000000","uploader":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"ref":"refs/changes/00/700800/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/00/700800/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/00/700800/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/00/700800/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/00/700800/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/00/700800/2"}}},"commit":{"parents":[{"commit":"84175687fb5c23fdf76322b18392c783ca8d6d73","subject":"Drop python 2 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/84175687fb5c23fdf76322b18392c783ca8d6d73"}]}],"author":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2019-12-31 06:40:21.000000000","tz":480},"committer":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2019-12-31 13:06:00.000000000","tz":480},"subject":"encap(nsh) encap(ethternet) does not set outter mac","message":"encap(nsh) encap(ethternet) does not set outter mac\n\nCloses-Bug: #1857954\n\nChange-Id: I4c2b85d07ee181f92c6e766fea9b5e9a727bd513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/19336711494c7507b32624b6df6783245753f652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/19336711494c7507b32624b6df6783245753f652"}]},"branch":"refs/heads/master"},"130a0d2934b8a340c04e75761547e021720822ee":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-04-03 08:42:19.000000000","uploader":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"ref":"refs/changes/00/700800/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/00/700800/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/00/700800/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/00/700800/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/00/700800/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/00/700800/3"}}},"commit":{"parents":[{"commit":"84175687fb5c23fdf76322b18392c783ca8d6d73","subject":"Drop python 2 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/84175687fb5c23fdf76322b18392c783ca8d6d73"}]}],"author":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2019-12-31 06:40:21.000000000","tz":480},"committer":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2020-04-03 08:32:13.000000000","tz":480},"subject":"encap(nsh) encap(ethternet) does not set outter mac","message":"encap(nsh) encap(ethternet) does not set outter mac\n\nCloses-Bug: #1857954\nhttps://www.openvswitch.org/support/dist-docs/ovs-actions.7.html\nencap(ethernet) leave ethernet source and destination are initially\nzeroed. while decap() don\u0027t know whether a ethernet header inside or\nnot, can not set dst mac unless check nsh next protocol filed,\nbut here leave it be.\n\nChange-Id: I4c2b85d07ee181f92c6e766fea9b5e9a727bd513\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/130a0d2934b8a340c04e75761547e021720822ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/130a0d2934b8a340c04e75761547e021720822ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
