)]}'
{"id":"openstack%2Fnetworking-sfc~712611","triplet_id":"openstack%2Fnetworking-sfc~master~I8b8962925d24deaa4bbbbc7eea5888b58745f8c5","project":"openstack/networking-sfc","branch":"master","topic":"restart_vlan_none","hashtags":[],"change_id":"I8b8962925d24deaa4bbbbc7eea5888b58745f8c5","subject":"restart neutron-openvswitch-agent dl_vlan is None","status":"ABANDONED","created":"2020-03-12 05:06:52.000000000","updated":"2023-02-28 10:57:07.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ae3dee4f0a1a43796265809150e648e669c8d406","_number":712611,"virtual_id_number":712611,"owner":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-22 08:18:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":-1,"date":"2020-09-13 17:46:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":-1,"date":"2020-08-14 06:40:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-22 08:18:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-14 06:40:48.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-09-13 17:46:50.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"}],"messages":[{"id":"3e218a04fe3b179f47a3b8607ff0fe7d33e5b680","author":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"date":"2020-03-12 05:06:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"405da6675ac283271f8c44c668c22c1e7dc55ad8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 07:09:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5addd3354b67456db633b4051df298bb : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c0bf5049f4f4181833ebe407db6600f : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5694860ebf63451a89060571907420cd : SUCCESS in 7m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a5c8de42dcd482183548911fe35d3d0 : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/165d5caabd8f45c6ab53acfc7a6122b0 : SUCCESS in 10m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce0f01dcc5b84c24a4799d21bfa5e57d : SUCCESS in 8m 08s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/7b18968f0ae4457bb56362ec8f7cdaaf : SUCCESS in 29m 50s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/a8dbba084a1f4a9b9023ea194d4620f6 : FAILURE in 1h 50m 17s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/8e9b4e4f043040b58770f2feb5104c10 : FAILURE in 1h 59m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05e962a298a4610a6dd85f4eae1c0c2fa1928d23","author":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"date":"2020-03-22 03:28:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b0e9ab7a02a16f54bebf6718597aa23974db6da2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 05:13:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26a1ab5d37fe470ea94e246e3bb5d6dd : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/535d395de14d478d916c429ba3e6c147 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6c367956eae49dcafebf66e43ca4bde : SUCCESS in 6m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/82098edb78aa4c4e9ba63adf2bf06e6c : SUCCESS in 8m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8cea3d0257c464fb466644ca854ce76 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6951a6eeb9294500bef63cb60beefd32 : SUCCESS in 5m 54s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/27b4c2c6eb374472b058b02c83b0a0a7 : SUCCESS in 15m 24s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/a55c55b4c0be4f9c885285c3198dcfa4 : FAILURE in 1h 40m 14s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/10584e2259b9409ba6a562551b09541c : FAILURE in 1h 44m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"983702dbc1d860db75e2233c6c7b92865a299d7f","author":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"date":"2020-03-22 06:15:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a9638dde4b97d6c1d7ecd70b8cb452a47ec3f30e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 08:18:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/306e6b97c7e749f4abec4a446993d0e8 : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c230709840234c09b023191361bf8620 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/903f1976f4434549a1c2ca44b47d45aa : SUCCESS in 8m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd48c0b776e14e0da3f53992b2e8b4a2 : SUCCESS in 9m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72f67b517872452d8ce76b41f24bd968 : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf27cca3c3114b3ba1f5e1d39496b898 : SUCCESS in 8m 12s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/04b2da3b82964bffb1fe66e7b7d0c406 : SUCCESS in 20m 50s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/aa5beb1477784b958c355e64ef5e154b : SUCCESS in 1h 47m 49s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/273fc3acdd184b59bb6d85296b51756a : FAILURE in 1h 54m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"43cdf998d13513f81c1c67e4ff3a174b5dbaae71","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-14 06:40:48.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8b41290aca250bbbe7b1e6f40395ac8130d48b3d","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-13 17:46:50.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ae3dee4f0a1a43796265809150e648e669c8d406","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-28 10:57:07.000000000","message":"Abandoned\n\nIf needed, feel free to restore the patch and address the comments","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8abc5db53578387c59e2fa4c092a7fcd5daa89a7","revisions":{"b3e4c5938d123e789028cd490571d5c291e8052b":{"kind":"REWORK","_number":1,"created":"2020-03-12 05:06:52.000000000","uploader":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"ref":"refs/changes/11/712611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/11/712611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/11/712611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/11/712611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/11/712611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/11/712611/1"}}},"commit":{"parents":[{"commit":"84175687fb5c23fdf76322b18392c783ca8d6d73","subject":"Drop python 2 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/84175687fb5c23fdf76322b18392c783ca8d6d73"}]}],"author":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2020-03-12 04:24:45.000000000","tz":480},"committer":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2020-03-12 05:05:56.000000000","tz":480},"subject":"restart neutron-openvswitch-agent dl_vlan is None","message":"restart neutron-openvswitch-agent dl_vlan is None\n\nrestart neutron-openvswitch-agent dl_vlan is None,\nbecause handle_port exec before openvswitch-agent\nport vlan manager, better get vlan from ovs db.\n\nChange-Id: I8b8962925d24deaa4bbbbc7eea5888b58745f8c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/b3e4c5938d123e789028cd490571d5c291e8052b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/b3e4c5938d123e789028cd490571d5c291e8052b"}]},"branch":"refs/heads/master"},"fdbb2370d62cca80f52e8a6bfef5991c67cce452":{"kind":"REWORK","_number":2,"created":"2020-03-22 03:28:34.000000000","uploader":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"ref":"refs/changes/11/712611/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/11/712611/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/11/712611/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/11/712611/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/11/712611/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/11/712611/2"}}},"commit":{"parents":[{"commit":"84175687fb5c23fdf76322b18392c783ca8d6d73","subject":"Drop python 2 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/84175687fb5c23fdf76322b18392c783ca8d6d73"}]}],"author":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2020-03-12 04:24:45.000000000","tz":480},"committer":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2020-03-22 03:28:15.000000000","tz":480},"subject":"restart neutron-openvswitch-agent dl_vlan is None","message":"restart neutron-openvswitch-agent dl_vlan is None\n\nrestart neutron-openvswitch-agent dl_vlan is None,\nbecause handle_port exec before openvswitch-agent\nport vlan manager, better get vlan from ovs db.\n\nChange-Id: I8b8962925d24deaa4bbbbc7eea5888b58745f8c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/fdbb2370d62cca80f52e8a6bfef5991c67cce452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/fdbb2370d62cca80f52e8a6bfef5991c67cce452"}]},"branch":"refs/heads/master"},"8abc5db53578387c59e2fa4c092a7fcd5daa89a7":{"kind":"REWORK","_number":3,"created":"2020-03-22 06:15:48.000000000","uploader":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"ref":"refs/changes/11/712611/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/11/712611/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/11/712611/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/11/712611/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/11/712611/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/11/712611/3"}}},"commit":{"parents":[{"commit":"84175687fb5c23fdf76322b18392c783ca8d6d73","subject":"Drop python 2 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/84175687fb5c23fdf76322b18392c783ca8d6d73"}]}],"author":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2020-03-12 04:24:45.000000000","tz":480},"committer":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2020-03-22 06:13:10.000000000","tz":480},"subject":"restart neutron-openvswitch-agent dl_vlan is None","message":"restart neutron-openvswitch-agent dl_vlan is None\n\nrestart neutron-openvswitch-agent dl_vlan is None,\nbecause handle_port exec before openvswitch-agent\nport vlan manager, better get vlan from ovs db.\n\nChange-Id: I8b8962925d24deaa4bbbbc7eea5888b58745f8c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/8abc5db53578387c59e2fa4c092a7fcd5daa89a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/8abc5db53578387c59e2fa4c092a7fcd5daa89a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
