)]}'
{"id":"openstack%2Fnetworking-sfc~714242","triplet_id":"openstack%2Fnetworking-sfc~master~I3d6334e2dd0c9a59824dff74cd57aff18ef33cc4","project":"openstack/networking-sfc","branch":"master","topic":"rpc_host","hashtags":[],"change_id":"I3d6334e2dd0c9a59824dff74cd57aff18ef33cc4","subject":"service graph flow rule no host key","status":"ABANDONED","created":"2020-03-21 02:18:49.000000000","updated":"2020-10-12 09:59:29.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"284ba7c9e36aca966b7b2c30f16a77c0feb44955","_number":714242,"virtual_id_number":714242,"owner":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-21 04:08:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":-1,"date":"2020-08-14 06:36:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"date":"2020-09-13 17:49:02.000000000","_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-21 04:08:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-14 06:36:25.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-09-13 17:49:02.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"}],"messages":[{"id":"159f84bd68f970058abf9b0e9b08972360a0e79a","author":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"date":"2020-03-21 02:18:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f89542076dd3d3de85f4607a046fdac3e8d2c998","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 04:08:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5294dcb180004d8d8c038095dd579116 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35e1fa68ea6b49de9d7506f60834fbe3 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08703687c34d4ef4b0309bc5282c6c81 : SUCCESS in 7m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c90480fb901408db955ab8e748ccf38 : SUCCESS in 9m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67206f6bd6864d2fbd4385066f82186b : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c68d569c7e8645119a80df26f33dc077 : SUCCESS in 5m 44s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/76a873f7194e40b88b419552930c05a6 : SUCCESS in 17m 08s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/15d0535d326d4a21bd3e374444c2f16c : SUCCESS in 1h 44m 21s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/7d1a92c9a3914142883fc25533469e39 : SUCCESS in 1h 44m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ecc54a797eb7d3e57c313a9181cfc90bb0eb29df","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-14 06:36:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nMaybe You should add some UT to cover such cases?","accounts_in_message":[],"_revision_number":1},{"id":"e79f77592a306c2603ee374f33fbfe0b89a936d7","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-13 17:49:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3876f4301fbde049503434aa1a69f8478985851e","author":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"date":"2020-10-12 09:59:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"13ed2a24c6e21368c42687cee56ae2674a46dd17","revisions":{"13ed2a24c6e21368c42687cee56ae2674a46dd17":{"kind":"REWORK","_number":1,"created":"2020-03-21 02:18:49.000000000","uploader":{"_account_id":15907,"name":"Wei Hui","email":"huiweics@gmail.com","username":"huiweics"},"ref":"refs/changes/42/714242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/42/714242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/42/714242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/42/714242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/42/714242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/42/714242/1"}}},"commit":{"parents":[{"commit":"84175687fb5c23fdf76322b18392c783ca8d6d73","subject":"Drop python 2 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/84175687fb5c23fdf76322b18392c783ca8d6d73"}]}],"author":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2020-03-21 02:18:09.000000000","tz":480},"committer":{"name":"Wei Hui","email":"huiweics@gmail.com","date":"2020-03-21 02:18:09.000000000","tz":480},"subject":"service graph flow rule no host key","message":"service graph flow rule no host key\n\nChange-Id: I3d6334e2dd0c9a59824dff74cd57aff18ef33cc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/13ed2a24c6e21368c42687cee56ae2674a46dd17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/13ed2a24c6e21368c42687cee56ae2674a46dd17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
