)]}'
{"id":"openstack%2Fnetworking-sfc~723726","triplet_id":"openstack%2Fnetworking-sfc~stable%2Fussuri~I5defd6765be94ed9af4981116d34a3d6fad13be6","project":"openstack/networking-sfc","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I5defd6765be94ed9af4981116d34a3d6fad13be6","subject":"Update .gitreview for stable/ussuri","status":"MERGED","created":"2020-04-27 22:02:11.000000000","updated":"2020-04-28 19:30:45.000000000","submitted":"2020-04-28 19:27:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723726-1588102078545-39316be2","meta_rev_id":"26faa0c705f3e73104f5bdcc224de75842cec0f8","_number":723726,"virtual_id_number":723726,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-28 19:27:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2020-04-27 23:20:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-04-28 10:00:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-04-28 10:00:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-27 23:20:36.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-04-28 10:00:31.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-04-28 19:27:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ae3f028b26d10fa38b6addbc9c76bb27a833c3a","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-27 22:02:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"860fae2b60f0eb73700fe115b0062c0348b5a40d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-27 23:20:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"51eddce89c1fc90bdf60de8ae044d7c897458620","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 00:10:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fafd2018fc56409489900c74e05416bd : SUCCESS in 14m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d923fbe30c74760bc588bee4612ed3b : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/253a233070f141ff8036251ad1fe9691 : SUCCESS in 11m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/894aa3c3c313437d8d74ad0ed5025c08 : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709005c75de14efaa16eaec62650c38b : SUCCESS in 6m 09s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/aa3a3e220ecf464ba798f0344363868e : SUCCESS in 15m 23s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/007693b5e34946d38378f1451986af97 : SUCCESS in 1h 37m 08s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/50dbdead9d7b48768b97cd0a19309fdb : SUCCESS in 1h 49m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9175255c2221744741cf8b7fe11ed20fc7adebb1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-28 10:00:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db156ef148eea5bc5553c531e5efd447f719faf0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-28 16:00:31.000000000","message":"Patch Set 1:\n\nrecheck zuul restart lost events","accounts_in_message":[],"_revision_number":1},{"id":"a11157f95454c356a6ef30c6f5965e01334ee5e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 17:55:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d897d0e1b8e951dfa53635d6905a5aa496a6950","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 18:39:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44cc517eb195460f996997b7c4d21101 : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/377697c7c0684fd9b35aa7311c4be34a : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12c768d8c9844266af354e3352e30217 : SUCCESS in 10m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a59e75a093b14a8a8102573c62d9bdfa : SUCCESS in 12m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e9d2a6e1231462fab9798773afbbd28 : SUCCESS in 7m 22s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/f547af29b13a4899b58600e75f7c5b04 : SUCCESS in 21m 38s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/e79a10dccfd24550b4eaae1b83366653 : SUCCESS in 1h 38m 55s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/a35b4e6c81f54db2929c75825515c400 : SUCCESS in 1h 56m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3bf47c08c242bcde85d3092097da8ef5f5113a95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 19:27:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82c39ef574644e8292320be8feb6b72c : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7a3ebb0d1054f6d89629857ed23acfe : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0a7217f59864b8996509eb99a3677bc : SUCCESS in 8m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8370a7af76e1478aae63ae797bd465be : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a2aca44b61a4cdeac61add07bb8c074 : SUCCESS in 7m 01s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/1f7a1b80a0724fbd9d3f5b85b62ffb99 : SUCCESS in 27m 47s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/85d3eacf18bd40e8b64c2f78eacd73b6 : SUCCESS in 1h 31m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d45258dad0cdd09f5dd35b1706803ba4af399eab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 19:27:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e06a0461f459f7d7ed18533c0911c8241f41578c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 19:30:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a865c9a731854bd39aa778f7a2d1bee6 : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8f44ad9e11da6728fff7e507f41d4e0471652634","revisions":{"8f44ad9e11da6728fff7e507f41d4e0471652634":{"kind":"REWORK","_number":1,"created":"2020-04-27 22:02:11.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/26/723726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/26/723726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/26/723726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/26/723726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/26/723726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/26/723726/1"}}},"commit":{"parents":[{"commit":"07f17590febf20a1f97794c4a2bc29fce415fe64","subject":"Merge \"Cleanup py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/07f17590febf20a1f97794c4a2bc29fce415fe64"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-27 22:02:09.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-27 22:02:09.000000000","tz":0},"subject":"Update .gitreview for stable/ussuri","message":"Update .gitreview for stable/ussuri\n\nChange-Id: I5defd6765be94ed9af4981116d34a3d6fad13be6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/8f44ad9e11da6728fff7e507f41d4e0471652634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/8f44ad9e11da6728fff7e507f41d4e0471652634"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
