)]}'
{"id":"openstack%2Fnetworking-sfc~726077","triplet_id":"openstack%2Fnetworking-sfc~stable%2Fussuri~I5b00ee328f83cec8375ad1538be3a16059af08a3","project":"openstack/networking-sfc","branch":"stable/ussuri","topic":"bug/1863021-stable/ussuri","hashtags":[],"change_id":"I5b00ee328f83cec8375ad1538be3a16059af08a3","subject":"Monkey patch original current_thread _active","status":"MERGED","created":"2020-05-07 09:59:12.000000000","updated":"2021-02-16 10:27:23.000000000","submitted":"2021-02-16 10:14:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726077-bug/1863021-stable/ussuri","meta_rev_id":"9b85afbcddc1dff296522d4db4678ecba831bc2a","_number":726077,"virtual_id_number":726077,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32324,"name":"zhangjunhui","email":"zhangjunhui@yovole.com","username":"zhangjunhui"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-16 10:14:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"all":[{"value":1,"date":"2021-01-18 09:51:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32324,"name":"zhangjunhui","email":"zhangjunhui@yovole.com","username":"zhangjunhui"},{"value":1,"date":"2021-02-16 10:27:23.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-02-16 10:27:23.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":2,"date":"2021-02-16 10:27:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-16 10:27:23.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":32324,"name":"zhangjunhui","email":"zhangjunhui@yovole.com","username":"zhangjunhui"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":1,"date":"2021-01-13 10:54:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32324,"name":"zhangjunhui","email":"zhangjunhui@yovole.com","username":"zhangjunhui"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"_account_id":32324,"name":"zhangjunhui","email":"zhangjunhui@yovole.com","username":"zhangjunhui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 09:59:12.000000000","updated_by":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"reviewer":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"state":"REVIEWER"},{"updated":"2020-05-07 10:25:30.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-09-01 17:58:32.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2020-11-03 13:56:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-10 03:50:07.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"},{"updated":"2020-12-15 08:54:39.000000000","updated_by":{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},"reviewer":{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},"state":"REVIEWER"},{"updated":"2021-01-18 09:51:44.000000000","updated_by":{"_account_id":32324,"name":"zhangjunhui","email":"zhangjunhui@yovole.com","username":"zhangjunhui"},"reviewer":{"_account_id":32324,"name":"zhangjunhui","email":"zhangjunhui@yovole.com","username":"zhangjunhui"},"state":"REVIEWER"}],"messages":[{"id":"6da2793896783b2840667506ffc8bdb889f57ff5","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-07 09:59:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bcc05f6e86fcc379bba0da3cc368636f54868f66","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-07 10:25:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5d3da8eff75d91f9295824550739c1de7b426e20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 11:52:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/230fa06163af42f693faf5c8a64f18f3 : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7255fbd629da4437826e5771af3a25f5 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c7646669c0f4598b124c03eced5e0a2 : SUCCESS in 9m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b623cc399324403a91e96cb7256acc4 : SUCCESS in 9m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4abf900b867548eeb04da64a2634cb8f : SUCCESS in 7m 08s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/9a8fe8e7bfb04ab1987ac7198261387d : SUCCESS in 16m 10s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/d1f3e3766b6b4429a8d6aa4fd194d24d : SUCCESS in 1h 52m 33s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/3e48585d143b41848d8894226d2287a2 : SUCCESS in 1h 45m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6012c028a3dde021757d7c6117d2f5580993f745","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-19 13:59:44.000000000","message":"Patch Set 1: Workflow+1\n\nSimple fix applied in many repositories (and backports), approving","accounts_in_message":[],"_revision_number":1},{"id":"cbb1531ee75327492fce9b71f603f16db324a399","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 13:59:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"375cabb895868f3fff19b40c94f55bc3db29fdc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 15:37:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2453f9cc9cd94e06a180584b71941c50 : SUCCESS in 9m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/796bf6cffa44403d9838b2c423c49f6d : FAILURE in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0122b9c7ad184015a10e10b6d84aa2e2 : SUCCESS in 9m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f680572552d9448da6c5b7fbd0cb99a0 : SUCCESS in 9m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bfe3be5e1eb4ff9a4e69eb3093f9e2a : SUCCESS in 7m 39s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/dce1c22ac1ec458ca5e4659e71482122 : SUCCESS in 32m 40s\n- networking-sfc-tempest-ussuri https://zuul.opendev.org/t/openstack/build/de1360394ec3437bb9cd3476bd25e178 : SUCCESS in 1h 29m 28s","accounts_in_message":[],"_revision_number":1},{"id":"e03dd2649c6abf1e0e2095e5479f5f0d950798ac","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-26 12:40:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"34178639c7c78ad2c99a13a01dff9d97376face3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 14:34:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c099cff1ae944fc9852506480acbe928 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f91d093ada844d1b0267a15cc1602bd : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/151a6aea049240dcb55fb9b167136a14 : SUCCESS in 9m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2fb95aa448e448b87381ad1cadc254f : SUCCESS in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1999039666ce48c7b8b5837e170b4d12 : SUCCESS in 7m 50s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/7848b365c832443cbd2d04f63bc6aaba : SUCCESS in 19m 56s\n- networking-sfc-tempest-ussuri https://zuul.opendev.org/t/openstack/build/3d065037c8614f1882dddef8c596a64a : SUCCESS in 1h 27m 41s\n- networking-sfc-tempest-multinode-ussuri https://zuul.opendev.org/t/openstack/build/f3457029f3f5442991bd9f9f7bc640d7 : SUCCESS in 1h 35m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ae4a8f7aaea9c44b0118af23faa30d92a7505e0c","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-01 17:58:32.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"c69ee3c687d5965ea80cb7bbe32e01584d71a968","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-03 12:15:16.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"42c73138a964bb377d8a06ab6bca5781195091d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 12:15:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2533b4b577e861a42c52f4e2f92f3216e2b65520","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 13:56:53.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"2b18059df9ceb94334aba2c997ca4cdd90afe2d0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-01 08:52:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c0155daba0c2ca88f3cbf4771956a49d5601116f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 09:13:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ceb915659607416594891c4831e1f7d8 : SUCCESS in 9m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed4e870ebc344097ae0f4ff422d10294 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8166738b6d040e4afceb05aa61489ef : SUCCESS in 7m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c846a8e2ca3420e81a9c6b264ed72c8 : SUCCESS in 9m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0526020dd8bb4c218339225220dc2f9f : SUCCESS in 8m 44s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/028f911f6d44413685f3544b5113e699 : SUCCESS in 16m 25s\n- networking-sfc-tempest-ussuri https://zuul.opendev.org/t/openstack/build/46b0228a040545f3adab1f3d84d16f01 : FAILURE in 12m 11s\n- networking-sfc-tempest-multinode-ussuri https://zuul.opendev.org/t/openstack/build/bdef4eca7d0e402086c3c71cf3518b9c : FAILURE in 17m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4c7543ccce0a3189744430dc77bb2ef7a076ebbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-12-05 23:47:35.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2d4a98493c5f582cd1de57c41a5220dbd1af34e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-06 01:03:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ae0cd7a0fb5442fa5959b03f4af1362 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ae6aba27c2741c0a986f81689d175b3 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/457f8a2d827940d7b095c96f2963fff1 : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56445e836e5b4622a5df9e8bdd04e314 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/043276e185cf4ebfa506ed478b4389e1 : SUCCESS in 5m 03s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/5cc9bc63e02d4bfe940194ad07d0dea5 : SUCCESS in 14m 56s\n- networking-sfc-tempest-ussuri https://zuul.opendev.org/t/openstack/build/38fc28c88b49460bbad90abba30a5b07 : SUCCESS in 1h 15m 10s\n- networking-sfc-tempest-multinode-ussuri https://zuul.opendev.org/t/openstack/build/f4a63794f30444a8acc9013c6f06fd97 : FAILURE in 35m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c65d6aefe1fe9031ac8a053de922202eb94d8fc7","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2020-12-10 03:50:07.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6e23c637fb8a54c6979458f20dfbefe59fa43634","author":{"_account_id":31821,"name":"LiangLu","email":"lu.liang@jp.fujitsu.com","username":"LiangLu"},"date":"2020-12-15 08:54:39.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6759a84c01624820b5dbbd79e1b8042faa7d05b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 19:50:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c2cb4559d0335675b95cee4c0194226fe2c29bde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 21:25:21.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"fe8a4f55a2cf3ae8b3cb11543fa1c13a16c1b967","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-13 10:54:13.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"03662bb637c1f4dccfb78634b85e316ee304dc9a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-13 10:54:26.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b177d307bd50d0c051cfcd719edb815957267579","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 11:40:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb75675141fb4315b3aa8d3a58c74bfe : SUCCESS in 11m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f65f795bbb7d4c21b92a5afe29d87bf1 : SUCCESS in 13m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/592dc25aaca642e584da57462af34b82 : SUCCESS in 9m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c02987c48ce4562ab602ab908729124 : SUCCESS in 17m 42s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/46cb28d398354902b944083da6637f2f : SUCCESS in 17m 34s\n- networking-sfc-tempest-ussuri https://zuul.opendev.org/t/openstack/build/7d7e72d5d2124a7fb2f1ea3ded6a6d81 : FAILURE in 4m 18s\n- networking-sfc-tempest-multinode-ussuri https://zuul.opendev.org/t/openstack/build/15fcd01e2b124eecb010c1c55093080a : RETRY_LIMIT in 3m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b768d284b04a9f56c9098578c8c215c87185db70","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-18 09:47:04.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"196f50fac4fd45be350d5cad1bb34c7aeb6aca15","author":{"_account_id":32324,"name":"zhangjunhui","email":"zhangjunhui@yovole.com","username":"zhangjunhui"},"date":"2021-01-18 09:51:44.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c11fde803176c2740040e361ab6859eaf0e7c6ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 11:25:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af4c2f46665f4ec9b9778123d2249058 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25f7b96197a94ffd83aa5775cf3165b3 : SUCCESS in 7m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f68b4bb40954fcbb774392f931084b6 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6270e056c828495387587bd66251b3de : SUCCESS in 8m 20s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/7303acef3bd24180a62242ce4edda1dd : SUCCESS in 18m 12s\n- networking-sfc-tempest-ussuri https://zuul.opendev.org/t/openstack/build/b634741eef5f4a688cce1efbee8eb424 : SUCCESS in 1h 30m 55s\n- networking-sfc-tempest-multinode-ussuri https://zuul.opendev.org/t/openstack/build/f155b14dcd9247a2b35e090247411138 : FAILURE in 34m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b5908654b9ec56a065fa9b6973748fe9a0c980ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 08:48:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"64a1577f5fe4550fe9b0ad5cd55adcfe2cb891a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 10:14:03.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8128336ac70d49c182055e2528ccf181 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96811ff5e61c498d9b929f306bf4f00d : SUCCESS in 8m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73869fbd4daf46e991169954ab870461 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03682d52e34742a39593eb2c9cb39ff2 : SUCCESS in 6m 35s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/1e80354a5cdd4879ac99d17f98ea7dba : SUCCESS in 18m 23s\n- networking-sfc-tempest-ussuri https://zuul.opendev.org/t/openstack/build/7e959f73e4c144a7889638caeda5886f : SUCCESS in 1h 21m 46s","accounts_in_message":[],"_revision_number":4},{"id":"8809e9ece1d5ea856c6ab657a3b33466017abcb6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 10:14:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"edcd6f5230c22310759a5d577677faebbb8af718","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 10:27:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3b02ec909fc488291fba054ccc02b92 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ba57ca7d1f2d8f5986103cf114089f5ca159dd77","revisions":{"845fd9ecb8ee78d6d1e1e6f4e30cd05c8c10df6b":{"kind":"REWORK","_number":1,"created":"2020-05-07 09:59:12.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/77/726077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/77/726077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/77/726077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/77/726077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/77/726077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/77/726077/1"}}},"commit":{"parents":[{"commit":"0d4cb4f1471b3bcd81016776b3f8decf563d835a","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/0d4cb4f1471b3bcd81016776b3f8decf563d835a"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-04 18:30:35.000000000","tz":-240},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-05-07 09:59:12.000000000","tz":0},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nMonkey patch the original current_thread to use the up-to-date _active\nglobal variable. This solution is based on that documented at:\nhttps://github.com/eventlet/eventlet/issues/592\n\nChange-Id: I5b00ee328f83cec8375ad1538be3a16059af08a3\nCloses-Bug: #1863021\n(cherry picked from commit 0b78dad62e72caa4d348cb3087f97f138f26a0cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/845fd9ecb8ee78d6d1e1e6f4e30cd05c8c10df6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/845fd9ecb8ee78d6d1e1e6f4e30cd05c8c10df6b"}]},"branch":"refs/heads/stable/ussuri"},"d4244555659c47d09bffde1794c82f4c9998db17":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-26 12:40:54.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/77/726077/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/77/726077/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/77/726077/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/77/726077/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/77/726077/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/77/726077/2"}}},"commit":{"parents":[{"commit":"0b4761f76b688332379f91c436b5535b73c58957","subject":"Fix pep8 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/0b4761f76b688332379f91c436b5535b73c58957"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-04 18:30:35.000000000","tz":-240},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-08-26 12:40:54.000000000","tz":0},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nMonkey patch the original current_thread to use the up-to-date _active\nglobal variable. This solution is based on that documented at:\nhttps://github.com/eventlet/eventlet/issues/592\n\nChange-Id: I5b00ee328f83cec8375ad1538be3a16059af08a3\nCloses-Bug: #1863021\n(cherry picked from commit 0b78dad62e72caa4d348cb3087f97f138f26a0cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/d4244555659c47d09bffde1794c82f4c9998db17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/d4244555659c47d09bffde1794c82f4c9998db17"}]},"branch":"refs/heads/stable/ussuri"},"5adefdfa9e2dba3c146b738b6f08bff1d22ff4e7":{"kind":"NO_CHANGE","_number":3,"created":"2020-12-05 23:47:35.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/77/726077/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/77/726077/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/77/726077/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/77/726077/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/77/726077/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/77/726077/3"}}},"commit":{"parents":[{"commit":"3265e82a94fcc8d6b0b297fab2b4d40421a294c4","subject":"Fix pep8 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/3265e82a94fcc8d6b0b297fab2b4d40421a294c4"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-04 18:30:35.000000000","tz":-240},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-12-05 23:47:35.000000000","tz":0},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nMonkey patch the original current_thread to use the up-to-date _active\nglobal variable. This solution is based on that documented at:\nhttps://github.com/eventlet/eventlet/issues/592\n\nChange-Id: I5b00ee328f83cec8375ad1538be3a16059af08a3\nCloses-Bug: #1863021\n(cherry picked from commit 0b78dad62e72caa4d348cb3087f97f138f26a0cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/5adefdfa9e2dba3c146b738b6f08bff1d22ff4e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/5adefdfa9e2dba3c146b738b6f08bff1d22ff4e7"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"},"ba57ca7d1f2d8f5986103cf114089f5ca159dd77":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-01-13 10:54:13.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/77/726077/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/77/726077/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/77/726077/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/77/726077/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/77/726077/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/77/726077/4"}}},"commit":{"parents":[{"commit":"90305b4f48341cfb3ee310e695ac57a8d50f4db8","subject":"[stable] Drop lower-constraints job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/90305b4f48341cfb3ee310e695ac57a8d50f4db8"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2020-05-04 18:30:35.000000000","tz":-240},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-01-13 10:54:13.000000000","tz":0},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nMonkey patch the original current_thread to use the up-to-date _active\nglobal variable. This solution is based on that documented at:\nhttps://github.com/eventlet/eventlet/issues/592\n\nChange-Id: I5b00ee328f83cec8375ad1538be3a16059af08a3\nCloses-Bug: #1863021\n(cherry picked from commit 0b78dad62e72caa4d348cb3087f97f138f26a0cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/ba57ca7d1f2d8f5986103cf114089f5ca159dd77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/ba57ca7d1f2d8f5986103cf114089f5ca159dd77"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
