)]}'
{"id":"openstack%2Fnetworking-sfc~726781","triplet_id":"openstack%2Fnetworking-sfc~stable%2Fussuri~Ic3fb1e046f2897bf356f573f63c3d48d24a131ee","project":"openstack/networking-sfc","branch":"stable/ussuri","topic":"ussuri-release","hashtags":[],"change_id":"Ic3fb1e046f2897bf356f573f63c3d48d24a131ee","subject":"Switch to stable/ussuri neutron-tempest-plugin jobs","status":"MERGED","created":"2020-05-11 11:26:43.000000000","updated":"2020-05-18 13:12:31.000000000","submitted":"2020-05-18 13:08:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726781-1589807338935-34ba9a06","meta_rev_id":"aa3d7ffd5a577eccc0266161e267101bafc4a553","_number":726781,"virtual_id_number":726781,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-18 13:08:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2020-05-14 16:38:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2020-05-15 15:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2020-05-15 15:39:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"date":"2020-05-13 15:18:41.000000000","_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 15:18:41.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-05-14 16:38:33.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-05-15 15:39:16.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-05-18 13:08:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7db5a8987666a87a63a94dc253572838db95622d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-11 11:26:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"027763ff512a9c0bf9941f3bd6a6cad01a165dc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 13:42:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6802f27483d2447e9a3cbdf24b0dc211 : SUCCESS in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82899a33e8fa45aa963a0cb5975270e0 : SUCCESS in 9m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/395b0c2d375c420cb3cff68f62ea017f : SUCCESS in 11m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/00f6c13e3fe84084b96340ae3014ca4e : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dbe7b424d5a4b7cbfed34f2ca56d031 : SUCCESS in 8m 26s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/d30ed0a5a1ac4b6db8993e6b4960c72d : SUCCESS in 20m 00s\n- networking-sfc-tempest-ussuri https://zuul.opendev.org/t/openstack/build/fb60b2a85857456caa95a7fb26054bef : SUCCESS in 1h 43m 11s\n- networking-sfc-tempest-multinode-ussuri https://zuul.opendev.org/t/openstack/build/9cef7b83f801477b9d0c6e423434513b : SUCCESS in 2h 06m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"74ba2b02791886eadad6c42d50b213ab732eade8","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-14 16:38:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6e3e184d695a12df133a62cd623dd2736b89a6ef","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-05-15 15:39:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f120c61908038c3de490a49175612a07a2d42357","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-05-18 09:26:47.000000000","message":"Patch Set 1:\n\nrecheck Depends-on merged did not trigger zuul","accounts_in_message":[],"_revision_number":1},{"id":"9ded8c5e0311a2d20c6509203d9daa04154f677d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 11:35:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e817bb8e87a44329c28d7f9654c3bc8 : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e507a2b59b848f093a0dc4cc15e8d83 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f922b9fea3294cd79d9c1e5d04473ea9 : SUCCESS in 8m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6378fb188f349aa83c8d78c61678f4b : SUCCESS in 10m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e33104fc13cc47e59e6a31d9003cbfa1 : SUCCESS in 9m 44s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/2dd94d95b3c94fe794d7f95b18b4da85 : SUCCESS in 28m 41s\n- networking-sfc-tempest-ussuri https://zuul.opendev.org/t/openstack/build/c4644af3439a46e692513c15572aab61 : SUCCESS in 1h 47m 09s\n- networking-sfc-tempest-multinode-ussuri https://zuul.opendev.org/t/openstack/build/6620f262bda64138bef221e982aae991 : SUCCESS in 2h 01m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1db4fb6b159b18fa54e2c7fc5f7017f4e8b4c633","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 11:36:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c07bcd3bf5b015d00b004e4a79bb84f2b403bae7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 13:08:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ad5ef3ae398035cacc3f8c92e76f42bbaf100f27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 13:08:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0126b5d74db743d0b9bb6ec4618ee37d : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8956bee9f904e7490870b096abcc541 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0d4bb73e61d43e9be8820a1fe9b10c9 : SUCCESS in 7m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/02fe784436a3479b9ba6e19488918488 : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/596765c2035d4e1fa7f21513bcb84674 : SUCCESS in 7m 07s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/f4164fb7f84f491b99ec568a714689e4 : SUCCESS in 15m 23s\n- networking-sfc-tempest-ussuri https://zuul.opendev.org/t/openstack/build/b7e0dd85f1ae4aedaed51c120562814a : SUCCESS in 1h 31m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ff8fe7c0263836ba16f3864d7224d24aa94cc8a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 13:12:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d99874405844973b3f52c6f3339ae82 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d4f69841d4cb20880a2cbfaa263dee8d0f545b3","revisions":{"3d4f69841d4cb20880a2cbfaa263dee8d0f545b3":{"kind":"REWORK","_number":1,"created":"2020-05-11 11:26:43.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/81/726781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/81/726781/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/81/726781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/81/726781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/81/726781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/81/726781/1"}}},"commit":{"parents":[{"commit":"0d4cb4f1471b3bcd81016776b3f8decf563d835a","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/0d4cb4f1471b3bcd81016776b3f8decf563d835a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-05-11 11:26:10.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-05-11 11:26:10.000000000","tz":120},"subject":"Switch to stable/ussuri neutron-tempest-plugin jobs","message":"Switch to stable/ussuri neutron-tempest-plugin jobs\n\nDepends-On: https://review.opendev.org/726776\nChange-Id: Ic3fb1e046f2897bf356f573f63c3d48d24a131ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/3d4f69841d4cb20880a2cbfaa263dee8d0f545b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/3d4f69841d4cb20880a2cbfaa263dee8d0f545b3"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
