)]}'
{"id":"openstack%2Fnetworking-sfc~747118","triplet_id":"openstack%2Fnetworking-sfc~master~I9eb5fb14c67171c8ba1fa1ea12d3154d6ea19aa7","project":"openstack/networking-sfc","branch":"master","hashtags":[],"change_id":"I9eb5fb14c67171c8ba1fa1ea12d3154d6ea19aa7","subject":"Fix pep8 job","status":"MERGED","created":"2020-08-20 07:55:19.000000000","updated":"2020-08-26 08:45:14.000000000","submitted":"2020-08-25 13:30:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747118-1598362212191-1545f216","meta_rev_id":"817ef92e98e41ea457d471176be85d28744372a6","_number":747118,"virtual_id_number":747118,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-25 13:30:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2020-08-25 09:40:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-08-25 11:53:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-08-25 11:53:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2020-08-25 09:40:12.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-25 09:40:12.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-08-25 11:53:27.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-08-25 13:30:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7880b85517c8fc12041517a07172d9bea91ec6d7","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-20 07:55:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ead08f5e4f424af6c8da1f7cbd6986380d75a613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 09:42:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5439e24911d47d1a9d36112a904aa3c : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1b67b79f6c846cfba4fb0b546608496 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/602c52ee78b142bb8b6a0ed474c2b77c : SUCCESS in 8m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0adea818cc1c41d185d1ec83d90f8226 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/632d76a70d36424f9c54143db728a5ef : SUCCESS in 5m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/86e69203c7d848ceb112210f016632d5 : SUCCESS in 2m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/349909393c8848858522dc568e4dcd15 : SUCCESS in 3m 31s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/f64f244280cb49bf8168e834cafcc29b : SUCCESS in 18m 00s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/55a619a792404cddb4a3e0dff8f5e272 : SUCCESS in 1h 33m 07s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/e5a62472e74644548658ee3644a317c1 : SUCCESS in 1h 40m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd27c4327f3c3591cb9a875b7005e71c56500fb7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-25 09:40:12.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"dd001f3997fae713730da8f3c7c1053fb6d8b17a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-25 11:53:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5958aaa5897e00e155c907989c683c6d8dd4ebdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 11:53:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab98b701f0b3f58bd57745f5ce5ee4b976b80f15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 13:30:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00f280062b1a421ea1fcdbd6ea4ab574 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d141d7fb021c4c1481f910a515d04ccf : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3dadbd3cd5ff4d6ca193cb8fe4113ea0 : SUCCESS in 9m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d13831b0899343f4ad86fe42e18a1eed : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61330c7e55f54161b90faef2c59a72b3 : SUCCESS in 9m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9055f48a5cc947b28247c66a906443ac : SUCCESS in 2m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc77669e8f5b4c0f89ab19c8b75e3a34 : SUCCESS in 3m 37s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/b3b958bd65724f209be332433ec459d3 : SUCCESS in 22m 25s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/c96814b5bf2947f8b4b812290979d8ba : SUCCESS in 1h 32m 10s","accounts_in_message":[],"_revision_number":1},{"id":"02566be4b728337643823f45fabd09e968cd1541","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 13:30:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b1a53c5f291bc2d7668bc5147aae1eb3c92e5c7e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 13:35:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/067314df74d445c59e9f056502f26d47 : SUCCESS in 4m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b41877c0cf3433888e9f99118b76833 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1},{"id":"24f1bdbb0958cbe0c92f0bfdc2d3a45eaa394e92","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-26 08:45:07.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 0b4761f76b688332379f91c436b5535b73c58957","accounts_in_message":[],"_revision_number":1},{"id":"817ef92e98e41ea457d471176be85d28744372a6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-26 08:45:14.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 9a3e457e4e3fb019d8ad645d577bd74a9627b90c","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79f95d9c3c6ab4feb9eccc739f58cb5ff2621fe2","revisions":{"79f95d9c3c6ab4feb9eccc739f58cb5ff2621fe2":{"kind":"REWORK","_number":1,"created":"2020-08-20 07:55:19.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/18/747118/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/18/747118/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/18/747118/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/18/747118/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/18/747118/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/18/747118/1"}}},"commit":{"parents":[{"commit":"211e515630d2695c3a48a7ba6212a22f312bb3bc","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/211e515630d2695c3a48a7ba6212a22f312bb3bc"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-08-20 07:54:26.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-08-20 07:54:26.000000000","tz":120},"subject":"Fix pep8 job","message":"Fix pep8 job\n\nNew versions of isort broke pylint. So force version 4.3.21, similarly\nto other Neutron projects fixes\n\nChange-Id: I9eb5fb14c67171c8ba1fa1ea12d3154d6ea19aa7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/79f95d9c3c6ab4feb9eccc739f58cb5ff2621fe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/79f95d9c3c6ab4feb9eccc739f58cb5ff2621fe2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
