)]}'
{"id":"openstack%2Fnetworking-sfc~748135","triplet_id":"openstack%2Fnetworking-sfc~stable%2Fussuri~I9eb5fb14c67171c8ba1fa1ea12d3154d6ea19aa7","project":"openstack/networking-sfc","branch":"stable/ussuri","hashtags":[],"change_id":"I9eb5fb14c67171c8ba1fa1ea12d3154d6ea19aa7","subject":"Fix pep8 job","status":"ABANDONED","created":"2020-08-26 08:45:07.000000000","updated":"2021-01-13 10:53:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"78d0c60a23fa33d9a8b5b0c421e76753d5fe4298","_number":748135,"virtual_id_number":748135,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":-2,"date":"2021-01-08 21:25:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"all":[{"value":1,"date":"2020-11-24 23:03:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-12-01 08:51:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":1,"date":"2021-01-08 19:50:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-17 04:18:16.000000000","updated_by":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"reviewer":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"state":"REVIEWER"},{"updated":"2020-09-17 08:06:13.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2020-10-03 13:32:09.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-11-03 16:40:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62f3fbccdc4024b5580f1f127f7d8cc641e1b73a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-26 08:45:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a8049cb20bca48eaef2e78925e6e1bf23a6c9f9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 10:45:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07bb3025cfd14fdb8723b462650803c4 : SUCCESS in 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bd6615171184ee9820ad4ef10b1fc84 : SUCCESS in 9m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2d78ce9271f45e0afca64a7e3d530d5 : SUCCESS in 8m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/087d4433db1c45819bcbe986a0686f28 : SUCCESS in 11m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc89b522266544d3bd6482dccc531363 : SUCCESS in 9m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/19ae56c450584b0480c145c7784668b1 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/910c9818c9e148caabfe4749cbba9cb2 : SUCCESS in 4m 40s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/723dc2f1a89b4dd7b7edbe20245d137c : SUCCESS in 19m 14s\n- networking-sfc-tempest-ussuri https://zuul.opendev.org/t/openstack/build/49ea42b574f849d088e4af1835daaa46 : SUCCESS in 1h 26m 52s\n- networking-sfc-tempest-multinode-ussuri https://zuul.opendev.org/t/openstack/build/f78329947ff14e84ae138e8b9b39d23e : SUCCESS in 1h 52m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34946d482363cdd7cdbd8ea3f595ba37a3e96340","author":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"date":"2020-09-17 04:18:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"be679b2db3515b5a8380f3acfe9733bc5beb1a7c","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2020-09-17 08:06:13.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"8ca8391657092f1fe0f27ec7b5f76478e2827a56","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-03 13:32:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"61abf8a859549e79ce488621bd0a0e530543c088","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-03 12:14:52.000000000","message":"Patch Set 1: Workflow+1\n\nW+1 as it is stable gate fix with other +2","accounts_in_message":[],"_revision_number":1},{"id":"3fe899507caf40fd40d75ff063ef387c6646bbaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 12:15:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"063a1ae7f86c7795cf29468a18c351fe8b935822","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 13:56:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f247334aec444eb9fa69465dead4a02 : SUCCESS in 11m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d254a9d0cf5345dca08842fa24ca22dd : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e367b1ac2904b139663f58a75baa9fa : SUCCESS in 8m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53c16bd9dd4548b7830bfdce8dead3e6 : SUCCESS in 8m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6c6d4994be54693a0768332205f77eb : SUCCESS in 6m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/74d0d55252b941e7a0d55f75a5c9edba : SUCCESS in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31736da0000f4068bdad7edb39f5eff9 : SUCCESS in 3m 25s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/1d0c91d522c54433a3fbb03090c54c8c : FAILURE in 20m 35s\n- networking-sfc-tempest-ussuri https://zuul.opendev.org/t/openstack/build/0ba39e81119f47619b5f7b76b7922109 : SUCCESS in 1h 32m 37s","accounts_in_message":[],"_revision_number":1},{"id":"fe6a3e23fd9092e5a7d4317a38f4e5c53b874f0a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-03 14:45:18.000000000","message":"Patch Set 1:\n\nrecheck functional","accounts_in_message":[],"_revision_number":1},{"id":"1fba52e29d3bba88a2de3e9ad28a3545da7ddc9b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-03 15:26:06.000000000","message":"Patch Set 1: -Workflow\n\nLooks like functional is failing picking up a too new neutron-lib :/","accounts_in_message":[],"_revision_number":1},{"id":"06fd59a94134021d885655e560ae4d32967e8929","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 16:40:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d633086d16c47c18b603bc34c43fa9f : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11d879e436954a529ad457bf67794805 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a8e8e1318614220b7ad5a1233cc329b : SUCCESS in 7m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd5669b1eba34f47a9daff1282618010 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd4807c13e014e77b899c466dd195f8b : SUCCESS in 6m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4678a10bb7f34d3fa4d09e6a57cf5ff0 : SUCCESS in 2m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7592891cda342fc81d3ccab352d02c0 : SUCCESS in 5m 00s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/cc5ece62f011441cad5c82926eba1466 : FAILURE in 17m 55s\n- networking-sfc-tempest-ussuri https://zuul.opendev.org/t/openstack/build/d5b9662e436741a2a5098d71a14de1de : SUCCESS in 1h 22m 34s\n- networking-sfc-tempest-multinode-ussuri https://zuul.opendev.org/t/openstack/build/76a79640793d4d4cadb807226ce39740 : SUCCESS in 1h 43m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4fe005e7e4454e675043758f814f161b908595df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-24 16:43:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8f72996a052519a070741f4ebdce57edeaa802e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 18:12:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddd3829c105d4f2e8f1567a8ac3fe927 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6b34cb5ac074cef928913eabe60a399 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1740288f9ae94c4388966f487de626d0 : SUCCESS in 10m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d8c56f464e7b44aeaeda023b2d30fdf2 : SUCCESS in 11m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5bafd1b781d476a9630ee28bd69effd : SUCCESS in 6m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/82da50c49a90453b829bab3bc25277d0 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a71d5a8590a430c95de5e5a8ec96e2a : SUCCESS in 3m 33s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/6c44641de40443c39b5f6327205f16cf : SUCCESS in 14m 38s\n- networking-sfc-tempest-ussuri https://zuul.opendev.org/t/openstack/build/67239bb13c844eb99bc55642e71e1fe9 : SUCCESS in 1h 22m 20s\n- networking-sfc-tempest-multinode-ussuri https://zuul.opendev.org/t/openstack/build/8d25b6f2d98242b39f676cbb657626ff : FAILURE in 36m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0f53ea3052489c572955976d5f168664b7046725","author":{"_account_id":4149,"name":"Itsuro Oda","email":"oda@valinux.co.jp","username":"oda-g"},"date":"2020-11-24 23:03:49.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5d3c74769b5f3d34e403819a3de515224ba304c3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-01 08:51:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0c9f8e4a731e0a03f759d30186dfa00484da4b4d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-08 19:50:09.000000000","message":"Patch Set 2: Workflow+1\n\nrecheck https://review.opendev.org/c/openstack/requirements/+/764021 merged","accounts_in_message":[],"_revision_number":2},{"id":"3478d93fb5a84188bd3079f68eed466ba63df8e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 19:50:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9532b7e5124f52b0f9ddb09ca4f3b8b8b802dce5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 21:22:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7396e369d52456797ec488a3059f8c2 : FAILURE in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a02123dc7c142a895e8d56a8e719321 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/538d69b26dfe4b46ab5f8bcd71a9f300 : SUCCESS in 7m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b63eef186c1a4906aad4bc2fdef5ae5c : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6bdf7c5b21c4472a65c221d6a12c0b9 : SUCCESS in 6m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1fd368af0ceb444b818c428314e3ecde : FAILURE in 2m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ca3e1a2680a442fa5811b5dfebc0407 : SUCCESS in 3m 27s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/b080fd204eea45b5830dc99600e04339 : SUCCESS in 18m 07s\n- networking-sfc-tempest-ussuri https://zuul.opendev.org/t/openstack/build/9ae3a72e1dfb4cf089cface1cf7caf47 : SUCCESS in 1h 21m 14s\n- networking-sfc-tempest-multinode-ussuri https://zuul.opendev.org/t/openstack/build/d76d97dce6124cedb2d552fb26ec8ef5 : FAILURE in 36m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"88381674c98f5c5c75addb139ada1c422591c91f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 21:25:20.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6cf0018fe8348bda2cf068bb1a5e687 : FAILURE in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b5b784416f448dcbecc476bfff1fd50 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d4e4cfb0bad417eb023508f9a761b09 : SUCCESS in 8m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85428bb181ea47d48ab5ce29c6bdbda4 : SUCCESS in 9m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb650872be7b4dc6ac6551a4ce3bc255 : SUCCESS in 7m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1cdf74443b144d2a8710e7211d71c617 : FAILURE in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/573193d7310945c1ab61d1725525e398 : SUCCESS in 4m 17s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/87b79f586a13417c80d20e3bd404a3af : SUCCESS in 16m 42s\n- networking-sfc-tempest-ussuri https://zuul.opendev.org/t/openstack/build/57267e36c43c460daf994197bc0540ed : SUCCESS in 1h 28m 25s","accounts_in_message":[],"_revision_number":2},{"id":"78d0c60a23fa33d9a8b5b0c421e76753d5fe4298","tag":"autogenerated:gerrit:abandon","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-13 10:53:49.000000000","message":"Abandoned\n\nFix merged in change https://review.opendev.org/c/openstack/networking-sfc/+/770190","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3265e82a94fcc8d6b0b297fab2b4d40421a294c4","revisions":{"0b4761f76b688332379f91c436b5535b73c58957":{"kind":"REWORK","_number":1,"created":"2020-08-26 08:45:07.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/35/748135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/35/748135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/35/748135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/35/748135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/35/748135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/35/748135/1"}}},"commit":{"parents":[{"commit":"3d4f69841d4cb20880a2cbfaa263dee8d0f545b3","subject":"Switch to stable/ussuri neutron-tempest-plugin jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/3d4f69841d4cb20880a2cbfaa263dee8d0f545b3"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-08-20 07:54:26.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-08-26 08:45:07.000000000","tz":0},"subject":"Fix pep8 job","message":"Fix pep8 job\n\nNew versions of isort broke pylint. So force version 4.3.21, similarly\nto other Neutron projects fixes\n\nChange-Id: I9eb5fb14c67171c8ba1fa1ea12d3154d6ea19aa7\n(cherry picked from commit 79f95d9c3c6ab4feb9eccc739f58cb5ff2621fe2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/0b4761f76b688332379f91c436b5535b73c58957"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/0b4761f76b688332379f91c436b5535b73c58957"}]},"branch":"refs/heads/stable/ussuri"},"3265e82a94fcc8d6b0b297fab2b4d40421a294c4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-24 16:43:31.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/35/748135/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/35/748135/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/35/748135/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/35/748135/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/35/748135/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/35/748135/2"}}},"commit":{"parents":[{"commit":"3d4f69841d4cb20880a2cbfaa263dee8d0f545b3","subject":"Switch to stable/ussuri neutron-tempest-plugin jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/3d4f69841d4cb20880a2cbfaa263dee8d0f545b3"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-08-20 07:54:26.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-11-24 16:43:16.000000000","tz":0},"subject":"Fix pep8 job","message":"Fix pep8 job\n\nNew versions of isort broke pylint. So force version 4.3.21, similarly\nto other Neutron projects fixes\n\nChange-Id: I9eb5fb14c67171c8ba1fa1ea12d3154d6ea19aa7\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/764021\n(cherry picked from commit 79f95d9c3c6ab4feb9eccc739f58cb5ff2621fe2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/3265e82a94fcc8d6b0b297fab2b4d40421a294c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/3265e82a94fcc8d6b0b297fab2b4d40421a294c4"}]},"branch":"refs/heads/stable/ussuri","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
