)]}'
{"id":"openstack%2Fnetworking-sfc~753614","triplet_id":"openstack%2Fnetworking-sfc~stable%2Fvictoria~I8979f59e5e4edc3a9972abdccdbf18dc83034f4c","project":"openstack/networking-sfc","branch":"stable/victoria","topic":"create-victoria","hashtags":[],"change_id":"I8979f59e5e4edc3a9972abdccdbf18dc83034f4c","subject":"Update .gitreview for stable/victoria","status":"MERGED","created":"2020-09-23 12:43:44.000000000","updated":"2020-10-28 12:27:51.000000000","submitted":"2020-10-28 12:27:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753614-1603888071967-5004bf0a","meta_rev_id":"c5b4be70052547f4763a9b52d5c074e9e92aea65","_number":753614,"virtual_id_number":753614,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-28 12:27:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2020-10-28 09:49:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-10-07 21:02:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2020-10-28 09:49:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 15:52:01.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-10-03 13:31:41.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-10-28 09:49:26.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-10-28 12:27:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"751016d2f2a8f9bffd94fd767e4114931bb87b22","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-23 12:43:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"baaccc9d67cbd40fcb5d15e08f141a363e2150ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 13:04:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/945cbf52dea645d086fce57d54d4da6d : FAILURE in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4e8c3c708ae42248a6e523bc196fc9c : FAILURE in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b5283c22f1844ebb1d5c74d21050eb0 : SUCCESS in 9m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eaeeec9d4a2c4f3ea6313c3b734c8cc3 : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c7874f7c51945b8920685a4eb9b7a5f : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":1},{"id":"24e5df553b66dac8648441a939adb25aadacfc00","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-24 15:52:01.000000000","message":"Patch Set 1:\n\nThis will need https://review.opendev.org/#/c/754068/ backport","accounts_in_message":[],"_revision_number":1},{"id":"94637dbf4ba54075a3dd3f766e89f8806101923e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-03 13:31:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"666467d6bec37230197feb6417e1453af5c0f22b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-10-07 21:02:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"36fd04a8b172e495d3a97c08f128240a55984f5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 22:58:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a449e809d22420b8ada62f80515663c : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca240f19d81f406f97a4b0da25e145de : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e0e94f08a4143148fec80ad0de93ea0 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b533f3ade1ed47b597ab68d90fdaf459 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34e596ba631c49828b6536be34385196 : SUCCESS in 6m 41s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/03b117520ad54a2dbdd72fad3cddf982 : SUCCESS in 20m 25s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/7aa95d9d6e544e75b8400adff3d11bcb : SUCCESS in 1h 24m 18s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/7b720b380b354c56829026d506dacab3 : SUCCESS in 1h 40m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"34fe43dc5a7480cf45fbbd7435158b2f2426cfe0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-10-28 09:49:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3f78b918724e3e9215116b4fd7edd62f3bafc168","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 09:49:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"63199ef42503431c7ba595c87065400365c6a410","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 12:27:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6e8da15e96ea92eba32196c459eb8a4e4ec3fdfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 12:27:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fe006e3c7174f46b62d48a4f1fdbb90 : SUCCESS in 22m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4de29682542434db66170806d05e625 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfe88776f121412f98dd24a697261869 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac28a96a1e9449f1b2589fb50eda1e58 : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3369135c22bf4cf098b0d87c42175243 : SUCCESS in 5m 26s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/b6cf1424afdd449f9dadc103a15f98c9 : SUCCESS in 1h 04m 09s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/5f6da83c48754a2897b6751ef6c425d8 : SUCCESS in 1h 28m 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"19924285deb7eef7caf7fa4139ca0fd8cbb27a96","revisions":{"e389f4a2cc61c954c0bb256d53b3c38a78ccba50":{"kind":"REWORK","_number":1,"created":"2020-09-23 12:43:44.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/14/753614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/14/753614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/14/753614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/14/753614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/14/753614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/14/753614/1"}}},"commit":{"parents":[{"commit":"79f95d9c3c6ab4feb9eccc739f58cb5ff2621fe2","subject":"Fix pep8 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/79f95d9c3c6ab4feb9eccc739f58cb5ff2621fe2"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-23 12:43:42.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-23 12:43:42.000000000","tz":0},"subject":"Update .gitreview for stable/victoria","message":"Update .gitreview for stable/victoria\n\nChange-Id: I8979f59e5e4edc3a9972abdccdbf18dc83034f4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/e389f4a2cc61c954c0bb256d53b3c38a78ccba50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/e389f4a2cc61c954c0bb256d53b3c38a78ccba50"}]},"branch":"refs/heads/stable/victoria"},"19924285deb7eef7caf7fa4139ca0fd8cbb27a96":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-10-07 21:02:59.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/14/753614/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/14/753614/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/14/753614/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/14/753614/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/14/753614/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/14/753614/2"}}},"commit":{"parents":[{"commit":"465dd90e181b5f4f31d74d2a2b61a197e02c0187","subject":"Fix lower-constraints and pep8 for Focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/465dd90e181b5f4f31d74d2a2b61a197e02c0187"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-23 12:43:42.000000000","tz":0},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-10-07 21:02:59.000000000","tz":0},"subject":"Update .gitreview for stable/victoria","message":"Update .gitreview for stable/victoria\n\nChange-Id: I8979f59e5e4edc3a9972abdccdbf18dc83034f4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/19924285deb7eef7caf7fa4139ca0fd8cbb27a96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/19924285deb7eef7caf7fa4139ca0fd8cbb27a96"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
