)]}'
{"id":"openstack%2Fnetworking-sfc~753617","triplet_id":"openstack%2Fnetworking-sfc~master~Ie34dda6a1cf2a844a9c3e92162c2eea066824750","project":"openstack/networking-sfc","branch":"master","topic":"add-wallaby-python-jobtemplates","hashtags":[],"change_id":"Ie34dda6a1cf2a844a9c3e92162c2eea066824750","subject":"Add Python3 wallaby unit tests","status":"MERGED","created":"2020-09-23 12:43:52.000000000","updated":"2020-10-07 18:10:55.000000000","submitted":"2020-10-07 18:05:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753617-1602093914935-2bbe566a","meta_rev_id":"6c7163e1288c6550166df226bc2d8997bb98b272","_number":753617,"virtual_id_number":753617,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-07 18:05:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2020-10-07 16:23:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-10-03 13:31:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-10-03 13:31:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-03 13:31:27.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-10-07 16:23:24.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-10-07 18:05:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"362a89e6e02db967e5abb0a7dcd5a8fe88899349","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-23 12:43:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aca5ac0016d162522cc538d56a99dd0704f16d4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 14:47:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6cc3d569d5f4b6ab1330b778f82d7d0 : FAILURE in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0aed159cc8c34d86a7ed82fb7952e44f : FAILURE in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a89e1f1cdf3e46cc85964a5a9e8028f8 : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f69e1cb1a134f6f944af1ccb2f28131 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b8f89bad5bd4fed8eeee5ae8d6411c4 : SUCCESS in 7m 04s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/0ceb88b96ee346b78f955c57311b8641 : SUCCESS in 18m 20s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/a16272670b9b4021ab522be028e3be88 : SUCCESS in 1h 19m 48s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/6e49681500224c648c2aeba2686d63f7 : SUCCESS in 1h 25m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4452d68bebe77a2debd8454342c50915bc80ee14","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-24 15:53:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"565fdfe6fada5c3561514fc8a0510fa4275c4603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 17:39:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e28d860b7c140d48e9bbc8967074a09 : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be7fdbda021a47bfbc69e2bec3d7c112 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b8866ba62a042b4b00e4bc2fae0f815 : SUCCESS in 9m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0795fd110ca2445a9bff2389c265f90d : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e63384c6b57740bb92a82f74de4ffc87 : SUCCESS in 10m 11s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/8f496cacb7174347b33e7bee91aef230 : SUCCESS in 25m 56s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/49e9470c043343dfb689076d49056402 : SUCCESS in 1h 30m 14s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/551679f268e74ddd825cef910c80ba02 : SUCCESS in 1h 36m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"59d87b2d3f32e64def9f3caf0aee70204af3bf69","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-03 13:31:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f99457e7247dd89f4bf8f95f98abdd0adcf7a362","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 15:13:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2d85f0394dbb92832650da956107ea6b8350784b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-10-07 16:23:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3a6b72c8115a4381808c2915720755bc294a5ff7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 18:05:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5c80a9373633560566c95c111228a9e921c3ee87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 18:05:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc64e4bd62dd43eab99e1851e7410d04 : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d92738d72ec433e8aa8002062f5894d : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c48e9d7aede84e47a51ecede3a7023cc : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/904532b471274d77bca2e6874852ddf2 : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b97959d248a40f0a26d613a8ea00876 : SUCCESS in 5m 57s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/a590144f51c04ddfa030a451c4cdb349 : SUCCESS in 21m 46s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/dc4ff57bf2c747a4b26b3920ef78554c : SUCCESS in 2h 40m 21s","accounts_in_message":[],"_revision_number":2},{"id":"55590738d2712dec73a2011e68375b82150f0d8c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 18:10:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5fba74205ab4c129fe8ded0b0de7f27 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c1fd7280d4ddac10b2c436bdbde0bf7f88cee1db","revisions":{"bfb8e0a4f7d6b4e1bcc4a676ccbe52a3dedb9202":{"kind":"REWORK","_number":1,"created":"2020-09-23 12:43:52.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/17/753617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/17/753617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/17/753617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/17/753617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/17/753617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/17/753617/1"}}},"commit":{"parents":[{"commit":"e66ccced79d94b8accf5f98dc549fca4bc5c03cb","subject":"Update master for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/e66ccced79d94b8accf5f98dc549fca4bc5c03cb"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-23 12:43:49.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-23 12:43:49.000000000","tz":0},"subject":"Add Python3 wallaby unit tests","message":"Add Python3 wallaby unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for wallaby.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: Ie34dda6a1cf2a844a9c3e92162c2eea066824750\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/bfb8e0a4f7d6b4e1bcc4a676ccbe52a3dedb9202"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/bfb8e0a4f7d6b4e1bcc4a676ccbe52a3dedb9202"}]},"branch":"refs/heads/master"},"c1fd7280d4ddac10b2c436bdbde0bf7f88cee1db":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-24 15:53:36.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/17/753617/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/17/753617/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/17/753617/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/17/753617/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/17/753617/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/17/753617/2"}}},"commit":{"parents":[{"commit":"c396e96e716cbab189c7f37b5fc0142e91554a0e","subject":"Update master for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/c396e96e716cbab189c7f37b5fc0142e91554a0e"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-23 12:43:49.000000000","tz":0},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-09-24 15:53:36.000000000","tz":0},"subject":"Add Python3 wallaby unit tests","message":"Add Python3 wallaby unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for wallaby.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: Ie34dda6a1cf2a844a9c3e92162c2eea066824750\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/c1fd7280d4ddac10b2c436bdbde0bf7f88cee1db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/c1fd7280d4ddac10b2c436bdbde0bf7f88cee1db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
