)]}'
{"id":"openstack%2Fnetworking-sfc~756588","triplet_id":"openstack%2Fnetworking-sfc~stable%2Fvictoria~I743acf9f2b7d7000e3340fee4887c8a1f3647133","project":"openstack/networking-sfc","branch":"stable/victoria","topic":"migrate-to-focal-stable/victoria","hashtags":[],"change_id":"I743acf9f2b7d7000e3340fee4887c8a1f3647133","subject":"Fix lower-constraints and pep8 for Focal","status":"MERGED","created":"2020-10-07 21:02:45.000000000","updated":"2020-10-28 12:33:04.000000000","submitted":"2020-10-28 12:27:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756588-1603888071270-eaa21e4c","meta_rev_id":"a4d0ce46793090fc4a7473ad4e3a70877f800b21","_number":756588,"virtual_id_number":756588,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-28 12:27:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2020-10-28 09:49:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2020-10-11 16:10:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2020-10-28 09:49:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-11 16:10:55.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-10-28 09:49:06.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-10-28 12:27:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f44fe80e8d71213d9aa2cb0b1f2d7dfc08177a7","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-10-07 21:02:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"55a4cc3e68885ee5c543111f7d2ea29112793183","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 22:55:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2a418b40b2b4e469d3b04d30d4b72f7 : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c84d7ff09ad3437db3901fded2c6a2cb : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eaa1707efe45426d9e55b6e48dbc6cb7 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b88f5fbbd41d41f6bb67885d17991cae : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a5675194d0d4748b02f3da0580d17ca : SUCCESS in 5m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ddde7fc227604ff582dcbeb66ac6f524 : SUCCESS in 2m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34c636fb4b0b4286b0958c03a6a8e32c : SUCCESS in 3m 27s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/a074d632d27d4edc8e43cbad952c42a7 : SUCCESS in 18m 45s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/41617648d75545b2b0653ac2e5a9e551 : SUCCESS in 1h 36m 39s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/6dc9aecbb8d44f649a74a687d2b90dd4 : SUCCESS in 1h 45m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d70f536d3e7116d592a1a722142cc8c56b7adbf5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-11 16:10:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b92fc9d586d8a2a2a1584be423413e871c88d56a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-10-28 09:49:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean cherry pick of a constraints fixing patch, needed to unblock the gate. Merged in master. LGTM.","accounts_in_message":[],"_revision_number":1},{"id":"737c79e84b691ab010e1e47346bb6528b4a04f6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 09:49:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d79c9414ebd4cda8830086496f472922337d4cf1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 12:27:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"249537d81e89b4d2e9cae8599f0cad30bd6a7a85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 12:27:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c96033fa344343dfb7b96ee4c53c57cf : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ace3fb5cffbe497c9242a91be07dfc66 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/117b1dbb2db54a499610da06678028d7 : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f97195a127a642a0808ed717e8ffdac9 : SUCCESS in 16m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91d158216f004193a7311e3f05d2c52f : SUCCESS in 5m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ece07dae1904488d9b91a28e92c4218c : SUCCESS in 2m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d291769fba9748cab891818b4f09a419 : SUCCESS in 3m 32s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/089bf1714e264f98a10351bd318632da : SUCCESS in 1h 02m 21s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/6685b8d427e546a98c686ff8b11abbc8 : SUCCESS in 2h 19m 40s","accounts_in_message":[],"_revision_number":1},{"id":"3417b96bd921653926d7273b47c9a0044c797aa8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 12:33:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d22191e0f484f388f92af099172c17e : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9faf80ea3dba496997af5ec040238147 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"465dd90e181b5f4f31d74d2a2b61a197e02c0187","revisions":{"465dd90e181b5f4f31d74d2a2b61a197e02c0187":{"kind":"REWORK","_number":1,"created":"2020-10-07 21:02:45.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/88/756588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/88/756588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/88/756588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/88/756588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/88/756588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/88/756588/1"}}},"commit":{"parents":[{"commit":"79f95d9c3c6ab4feb9eccc739f58cb5ff2621fe2","subject":"Fix pep8 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/79f95d9c3c6ab4feb9eccc739f58cb5ff2621fe2"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-09-24 11:12:20.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-10-07 21:02:45.000000000","tz":0},"subject":"Fix lower-constraints and pep8 for Focal","message":"Fix lower-constraints and pep8 for Focal\n\nBump lower-constraints to versions supporting python 3.8 and with wheels\navailable and sync requirements\n\nBump hacking to 3.2.0 in order to work with Focal, disabling W504\nto align with neutron\n\nClean pep8 requirements from lower-constraints\n\nChange-Id: I743acf9f2b7d7000e3340fee4887c8a1f3647133\n(cherry picked from commit 5ea0ed67c45e6f76cae40c28cbe361a62b66142c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/465dd90e181b5f4f31d74d2a2b61a197e02c0187"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/465dd90e181b5f4f31d74d2a2b61a197e02c0187"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
