)]}'
{"id":"openstack%2Fnetworking-sfc~769637","triplet_id":"openstack%2Fnetworking-sfc~master~I439535f0ac5df735bdf0caf538edad97e1b0ac8d","project":"openstack/networking-sfc","branch":"master","hashtags":[],"change_id":"I439535f0ac5df735bdf0caf538edad97e1b0ac8d","subject":"Use TOX_CONSTRAINTS_FILE","status":"MERGED","created":"2021-01-07 02:10:08.000000000","updated":"2021-03-10 12:16:19.000000000","submitted":"2021-03-10 12:14:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769637","meta_rev_id":"56cda8a7414ee04449192d89aa68637ec3371b06","_number":769637,"virtual_id_number":769637,"owner":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-10 12:14:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"recommended":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"all":[{"value":1,"date":"2021-03-01 20:49:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":2,"date":"2021-03-02 22:47:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":1,"date":"2021-03-10 10:48:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-07 03:30:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-29 00:36:03.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"CC"},{"updated":"2021-02-23 03:20:21.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"},{"updated":"2021-03-02 22:47:28.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"4d99884867c69e0b55635607c606f334ce94bbc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-07 02:10:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9388fb9998aafd80d3fee0b6d3ac8c0fa6f9cbf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 03:30:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd22893b36154cb6b4a480bf89ea11c1 : FAILURE in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88519771092b42faa179dfb3410f72bb : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e271d5ac80c4e17a7775c73daf5ee78 : FAILURE in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef19e0d528dd48b79e4cfe7e02036853 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9232134be26d4b8c88a3c2f679cb6d01 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5616063b3707498d87ce1d1663910393 : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fcd8586acc74ee08bf99977b514cf3d : SUCCESS in 3m 14s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/6fa95add4554480c9807abafdb64387b : SUCCESS in 16m 06s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/754ccec048084137a86bbcd46abe06a6 : SUCCESS in 1h 18m 33s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/1c0560b1127a444d8a2d31845e0bb72b : FAILURE in 41m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99ccfe09c93cb3a750b8f83f1b4026df5c988f8a","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-01-29 00:36:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ce82f95d00a23ae87db86c8be1a2ea491e45c828","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 01:56:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8c0caa74dd04f4e9121edf28996073f : FAILURE in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a3208726074415f8d958ab90e1eef0c : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e06767630ea43d7875440aefa4e68fd : FAILURE in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18e2199660044b8d9dcf71d534c5fb8a : SUCCESS in 7m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca798e0143af46f3b9175708f690bff8 : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cd1c3f850af42c588dd2810be5d862e : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb00276b10ea4b84bd62cf08098aabf4 : SUCCESS in 3m 40s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/4e1aaec591f74c3ea81357e6bc252d1f : SUCCESS in 18m 03s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/54680bfefe9945c1b0d398918e3317f1 : POST_FAILURE in 1h 17m 20s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/1e31f309a5ce464da4be616a981f925a : FAILURE in 48m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"478f160d5c8078a0f1a25bd4112165408b287ac5","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-02-23 03:20:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c021ca8218bb5b0aaae4e107d71ce6cafa8afb7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-03-01 18:02:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"32011d636a3feb6efaa91bd3174127f7f0ec6974","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-03-01 18:02:55.000000000","message":"Removed Code-Review+1 by Manpreet Kaur \u003ckaurmanpreet2620@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"a976942acdb861467aa790fc4c3ea2098618f180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 19:57:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/009d74037e9e432e88006a555cb98636 : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63718c9956274ae3ba5d04ef4fea2b2a : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c00142b96e504b38a746c5c1559cdf00 : SUCCESS in 9m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf6b58936bd947c19c23de3926261150 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36f4cc4c3af84269b19bec4112e777d4 : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709a88a1c8b7472d802eb651ee162ea4 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5848283f7e494ed288918212a9a372e9 : SUCCESS in 4m 10s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/efe31d1949484518b44ed12de2f56047 : SUCCESS in 20m 57s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/e54b6b2064bb449d927508af22aac39d : SUCCESS in 1h 28m 50s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/a226154c471742b6aa7ddb3304bbdaea : FAILURE in 34m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e32226a65f391665788799eb1d00a9e54d26a99e","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-03-01 20:49:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"edc5b10150bf2600e1f157fe9c3a4c5e986d6c96","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-03-02 22:47:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f789fa33837b866bdc923e580b68e67fa7c20bd5","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-03-10 10:48:02.000000000","message":"Patch Set 2: Workflow+1\n\nstraight-forward change across OpenStack projects. Let\u0027s land it while we have a single +2.","accounts_in_message":[],"_revision_number":2},{"id":"7f7ddc7c43d5326feb4311d5e6d53fe5ae440327","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 10:48:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0cafb3d3fba0265ba9b11c488d3bfe705cf671ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 12:14:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1fd2713892634d1b8da8526c5ec22a6d : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1ed42edae224c4d8221526d69bd45a6 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08cd378cb49847f79f0d1190118cf4ba : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/522eac01556c4b91b64d07b2b6f2eb6e : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3778149abeb84d72ba4742269067b9fb : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac44f6a778364c2fba87ae7f8b1757ef : SUCCESS in 5m 36s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/70fb600123a847019e7aa47df152c3c0 : SUCCESS in 19m 50s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/f57954b2abf8445e8048a0a1e9babce9 : SUCCESS in 1h 20m 29s","accounts_in_message":[],"_revision_number":2},{"id":"b1624faea69eb03c8e92615203c94a8ce282767a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 12:14:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"56cda8a7414ee04449192d89aa68637ec3371b06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 12:16:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7505e49cf4e445098a92677cb5e2bc2 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/068cfe9bc5a141529a2c301cc9276fa0 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"42509e238cdb184708f62453887f11aac545169c","revisions":{"6b0b4a240cba09538da32f4ccab5a68718d63394":{"kind":"REWORK","_number":1,"created":"2021-01-07 02:10:08.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/37/769637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/37/769637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/37/769637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/37/769637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/37/769637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/37/769637/1"}}},"commit":{"parents":[{"commit":"c1fd7280d4ddac10b2c436bdbde0bf7f88cee1db","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/c1fd7280d4ddac10b2c436bdbde0bf7f88cee1db"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 02:09:16.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 02:10:01.000000000","tz":480},"subject":"Use TOX_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\nThis allows to use lower-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\nChange-Id: I439535f0ac5df735bdf0caf538edad97e1b0ac8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/6b0b4a240cba09538da32f4ccab5a68718d63394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/6b0b4a240cba09538da32f4ccab5a68718d63394"}]},"branch":"refs/heads/master"},"42509e238cdb184708f62453887f11aac545169c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-01 18:02:07.000000000","uploader":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"ref":"refs/changes/37/769637/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/37/769637/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/37/769637/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/37/769637/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/37/769637/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/37/769637/2"}}},"commit":{"parents":[{"commit":"0eb06a29299264c7a5ca3c5ff0691715335d4385","subject":"Fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/0eb06a29299264c7a5ca3c5ff0691715335d4385"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 02:09:16.000000000","tz":480},"committer":{"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","date":"2021-03-01 18:02:07.000000000","tz":0},"subject":"Use TOX_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\nThis allows to use lower-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\nChange-Id: I439535f0ac5df735bdf0caf538edad97e1b0ac8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/42509e238cdb184708f62453887f11aac545169c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/42509e238cdb184708f62453887f11aac545169c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
