)]}'
{"id":"openstack%2Fnetworking-sfc~770363","triplet_id":"openstack%2Fnetworking-sfc~stable%2Ftrain~Ic62b63fd2e1f314f6c465a259df75d197180e1f9","project":"openstack/networking-sfc","branch":"stable/train","topic":"oslo_lc_drop","hashtags":[],"change_id":"Ic62b63fd2e1f314f6c465a259df75d197180e1f9","subject":"[stable] Drop lower-constraints job","status":"MERGED","created":"2021-01-12 15:21:38.000000000","updated":"2021-02-16 10:21:43.000000000","submitted":"2021-02-16 10:14:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770363-oslo_lc_drop","meta_rev_id":"ece626995d67eca2beece7232e53b5a37784ceb5","_number":770363,"virtual_id_number":770363,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-16 10:14:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-02-16 08:47:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-01-20 11:17:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-02-15 13:01:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-15 02:27:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-02-16 08:47:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-12 15:42:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-01-12 18:21:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-15 02:27:56.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2021-01-20 11:17:10.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-02-15 13:01:28.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"38ed37d0ded6b5f733b2e7f3cc6f6ecc67cb86cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-12 15:21:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7019ba1e456de2e3ea5cf8030f1ad08dac5a696","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-12 15:29:43.000000000","message":"Topic set to oslo_lc_drop","accounts_in_message":[],"_revision_number":1},{"id":"a52e58fb85e26dae85e7432a457bd4406f903923","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-12 15:37:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c7fc690cf5d396ad23ce45576e25d281d3c013d0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-12 15:42:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"df292cb763e7dcda0736accc63dc636364701e50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 18:21:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/617f687e64df4dc8a7d36db8a71c1b01 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1f4fc38e6cb4bcab4ec072b220560a2 : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1655ebdb1e504419a2cbeefac335b764 : SUCCESS in 4m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b2c957df890447deaf63f4753087aa4d : FAILURE in 2m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/076bb199e8f3414689e5484148ce928b : SUCCESS in 8m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb6da9cba78d4e02998777e0f766a4ca : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4da3613341ab4697a0a6083156ce8d00 : SUCCESS in 3m 41s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/685b86cad0b14a8bb790fee17ca16dfe : SUCCESS in 33m 42s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/fe2b44a014cb4d0195663af6d372b36a : SUCCESS in 1h 57m 25s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/f7b38fb8025e443886adfc62e2ac8d49 : SUCCESS in 1h 59m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"248db7471861ed85e620820e6d98485b0b0cb21a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-01-13 10:46:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"587853691bcfdb5b4172f259f5f9823f1a579fbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 12:41:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5061641fdf394b4baa49919e31fa0393 : SUCCESS in 7m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1e7c92bc4f8e4fae9629ccbbe23599c2 : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d3e33eb8fa24b57855dbf1c71f4f550 : SUCCESS in 6m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f7d99d2be19b4ac0b587ff7c2d7909f3 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/034d6e332c1f4d638c9eee7573e467e8 : SUCCESS in 8m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71ae0bde7a504f80bfa4fd9bbfbc58d0 : SUCCESS in 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5039109c75b24c9382bfa651c8ec90e7 : SUCCESS in 3m 35s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/80375e183d9d46d4a8f939d44a51241f : SUCCESS in 28m 56s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/2b9ff12b3e984dbba8336f7bf7b9b6b1 : SUCCESS in 1h 31m 43s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/6bfdeb1f9686406bb7c41a3dbb4188dd : SUCCESS in 1h 37m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7aab6d3743539f30da9b0af3b6554151bd99feef","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-15 02:27:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"eebff6c77c037a82ed045e4ad75e98ec51748d17","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-20 11:17:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2c1999bb73362810606e59b625c57f1d30ec0a06","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-02-15 13:01:28.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"39a947d1a23e7ef1b0bd7b9de12970eb1d7674eb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-02-16 08:47:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"198a46f8994b77b052b10eecbcd9b97c63d1b0f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 08:48:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c7fdd27939d8ec62baf66a3b6fb0995e36836d0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 10:14:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4dc15bf2e6e4078846b1e86de6d6280 : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/972af2be04bd42849aa66cd9793d2b0e : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaec3c78bc274ee8a1e26a322ee61100 : SUCCESS in 4m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3566180a6a4e4497baa1e3b76250500a : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/984d1e3c332e4669880f3da999d1881a : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b787b8540594c43abcfa430d74a8b98 : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da459483f0854dfe89932eac4a578aa7 : SUCCESS in 3m 19s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/264107d09b9f4895899ebb11db9e9982 : SUCCESS in 14m 21s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/f2aa060f34f34709aca7ef40dc55ed73 : SUCCESS in 1h 09m 19s","accounts_in_message":[],"_revision_number":3},{"id":"c7c16f296b357be73a70aee1fca2390d354ffb46","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 10:14:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ece626995d67eca2beece7232e53b5a37784ceb5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 10:21:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f8284c5dc834b519d47d487e4223c18 : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98c91514967f4a059091854866dd23d4 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2f75b3083bd8f04f469efc9393d8d3695e24b413","revisions":{"ac51189adcfbf39394b41e5dc6c8ba2bceb94fce":{"kind":"REWORK","_number":1,"created":"2021-01-12 15:21:38.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/63/770363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/63/770363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/63/770363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/63/770363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/63/770363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/63/770363/1"}}},"commit":{"parents":[{"commit":"31b28ea8355af8c64cfddc48d4045de4613ba199","subject":"Monkey patch original current_thread _active","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/31b28ea8355af8c64cfddc48d4045de4613ba199"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-01-12 15:15:28.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-01-12 15:21:21.000000000","tz":60},"subject":"[stable] Drop lower-constraints job","message":"[stable] Drop lower-constraints job\n\nAs discussed in neutron team meeting [1], it is not worth fixing issues\nidentified by new pip resolver in stable branches. So let\u0027s drop the job\n\n[1] http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html#l-164\n\nConflicts:\n\tzuul.d/project.yaml\n\nChange-Id: Ic62b63fd2e1f314f6c465a259df75d197180e1f9\n(cherry picked from commit 3e179bfb4d82bcc81d9056cc5322e33f4f8d390e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/ac51189adcfbf39394b41e5dc6c8ba2bceb94fce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/ac51189adcfbf39394b41e5dc6c8ba2bceb94fce"}]},"branch":"refs/heads/stable/train"},"0b4702b601f76e7e58ef14d7b229207871d99ae2":{"kind":"REWORK","_number":2,"created":"2021-01-12 15:37:52.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/63/770363/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/63/770363/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/63/770363/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/63/770363/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/63/770363/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/63/770363/2"}}},"commit":{"parents":[{"commit":"31b28ea8355af8c64cfddc48d4045de4613ba199","subject":"Monkey patch original current_thread _active","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/31b28ea8355af8c64cfddc48d4045de4613ba199"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-01-12 15:15:28.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-01-12 15:37:32.000000000","tz":60},"subject":"[stable] Drop lower-constraints job","message":"[stable] Drop lower-constraints job\n\nAs discussed in neutron team meeting [1], it is not worth fixing issues\nidentified by new pip resolver in stable branches. So let\u0027s drop the job\n\n[1] http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html#l-164\n\nConflicts:\n\tlower-constraints.txt\n\ttox.ini\n\tzuul.d/project.yaml\n\nChange-Id: Ic62b63fd2e1f314f6c465a259df75d197180e1f9\n(cherry picked from commit 133f6f2961e42ec0626896efc3087bfafc68217b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/0b4702b601f76e7e58ef14d7b229207871d99ae2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/0b4702b601f76e7e58ef14d7b229207871d99ae2"}]},"branch":"refs/heads/stable/train"},"2f75b3083bd8f04f469efc9393d8d3695e24b413":{"kind":"REWORK","_number":3,"created":"2021-01-13 10:46:43.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/63/770363/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/63/770363/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/63/770363/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/63/770363/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/63/770363/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/63/770363/3"}}},"commit":{"parents":[{"commit":"31b28ea8355af8c64cfddc48d4045de4613ba199","subject":"Monkey patch original current_thread _active","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/31b28ea8355af8c64cfddc48d4045de4613ba199"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-01-12 15:15:28.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-01-13 10:45:50.000000000","tz":60},"subject":"[stable] Drop lower-constraints job","message":"[stable] Drop lower-constraints job\n\nAs discussed in neutron team meeting [1], it is not worth fixing issues\nidentified by new pip resolver in stable branches. So let\u0027s drop the job\n\n[1] http://eavesdrop.openstack.org/meetings/networking/2021/networking.2021-01-12-14.00.log.html#l-164\n\nConflicts:\n\tlower-constraints.txt\n\ttox.ini\n\tzuul.d/project.yaml\n\nChange-Id: Ic62b63fd2e1f314f6c465a259df75d197180e1f9\n(cherry picked from commit 1a51c0f52e4fa8275333904d617f3b24a7458f4f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/2f75b3083bd8f04f469efc9393d8d3695e24b413"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/2f75b3083bd8f04f469efc9393d8d3695e24b413"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
