)]}'
{"id":"openstack%2Fnetworking-sfc~789629","triplet_id":"openstack%2Fnetworking-sfc~master~If9c66f225fb2fcd299b05c3f29652768a99a040b","project":"openstack/networking-sfc","branch":"master","topic":"update-bionic-jobs","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-12-20 10:03:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If9c66f225fb2fcd299b05c3f29652768a99a040b","subject":"Update networking-sfc-tempest-multinode to be run on Focal","status":"MERGED","created":"2021-05-04 20:09:39.000000000","updated":"2021-12-20 10:09:37.000000000","submitted":"2021-12-20 10:03:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"789629-update-bionic-jobs","meta_rev_id":"82cbf61a55d5de11082038c646ba1fe56cb69d77","_number":789629,"virtual_id_number":789629,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-20 10:03:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-12-20 08:11:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-12-20 10:09:37.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-12-20 10:09:37.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":2,"date":"2021-12-20 10:09:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-20 10:09:37.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-12-20 08:11:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-04 21:24:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-05 06:31:17.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-05-06 14:25:19.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-05-10 10:49:51.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"},{"updated":"2021-05-13 03:13:52.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2021-10-07 11:07:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2021-12-20 08:11:07.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"3c423d74aef1566954fcf8b8f55a6b940bf64d92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-04 20:09:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8315e7ed69be87d2543e46e2ecbb856f487422e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 21:24:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/796ac2e18fb54869a43b4ee6e501f02a : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daeef0734f4e4ab7ac2ad2f3c5e31844 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d61ae2a3921c449f8c8cf3cefb0202a2 : SUCCESS in 8m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c848db6e3b44ad583e30c969d915321 : SUCCESS in 8m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08c4d7d84fa44448ae119de08a4355fa : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3ac20ca7ee74178bd7dbe84e1f012d2 : SUCCESS in 7m 16s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/11a865a9ea854fd1b2369fe7f97fbcc3 : SUCCESS in 16m 57s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/8e62ff190c804b73bbd410cad4bbccef : SUCCESS in 1h 13m 36s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/844f23e142164c2cbc2908686ac2a1a1 : FAILURE in 36m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ea4e88bbdc67c7eae16df76d20485a882ea0967","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-05 06:31:17.000000000","message":"Patch Set 1: Code-Review+2\n\nIn sync with the mail thread:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-May/022248.html","accounts_in_message":[],"_revision_number":1},{"id":"f3df0e8ebd29444fb503721829531e8cd22aa7d8","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-05-06 14:25:19.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nGood catch!","accounts_in_message":[],"_revision_number":1},{"id":"1db316f0bdf336646a5485cbb11e5bb8976e0cb3","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-05-10 10:49:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1edecbd0381ac8367d817e10c34b84e39e037249","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-13 03:13:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"239f76e5dd4ba1c4767cf9a905ce51781d6073ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-07 09:51:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c4aef1ce1bc7e7856bc2b732132834446882fdcd","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-10-07 10:26:26.000000000","message":"Patch Set 2:\n\nnetworking-sfc-tempest-multinode has other issues if I remember correctly but at least it picks the correct system image with this fix","accounts_in_message":[],"_revision_number":2},{"id":"4660d56a25b4ebf5e6dad5c47e7a0778173382c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 10:46:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe95de61bffb41df9416026c5e14e24d : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed2d1cd1f939446e9e144723d4c6400d : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1110812564d14189962fc32cd9b6c714 : SUCCESS in 8m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8eac4fea997b467d9f9c36e83c06b87e : SUCCESS in 8m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4a6858c9423498e86e6a4cc4cd04bc9 : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/def1602e0e294b09a9c47eb357463859 : SUCCESS in 8m 17s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/0569599d6c0b4793977c17f99823455e : SUCCESS in 22m 43s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/bf01ab94d664400fae7e5be4dc55644f : FAILURE in 37m 59s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/efd4d1f9d7bc45f0b9a802e67db38b6f : FAILURE in 32m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"83f94a429ed04f187101ca41081b1d4fc7ffdc56","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-10-07 11:07:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0332b9595f0da132e98b845aaff03b583ea3431e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 12:24:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3c764e1c5804b349ebbf181fa2c1649 : SUCCESS in 9m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03ae210a3b0e4ebb829efab5ade42183 : SUCCESS in 9m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee9b0f74489347439ac749bd38b9e06d : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c07befee73e43eaad70156a690a6029 : SUCCESS in 8m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17f3fec9d1bb47f48dcc93fa9bce3782 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3560c430ca7f4910b91c8f32159ce99a : SUCCESS in 8m 40s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/a5fb81c4c9e54fbca49cc5ea6d097b37 : SUCCESS in 21m 21s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/81b69c5904ea47878c684bf79ff39b68 : SUCCESS in 1h 15m 07s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/c22ff6ecc17248c0950cdbe900672aec : FAILURE in 30m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a858a8fa984aa778e01a1ad928401af4e905638b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-20 08:11:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b24235a3ff63597654c7c6585dd6b8e603ff2817","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 08:11:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e6ae5cc09910edcf11d78e915467a3802ba0a75c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 10:03:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/283733b7ee514d0684b89f4f025389e1 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da1f9e7e43734ff1a5ca6a683eadc85d : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9859d6c2c75c466b841c0722c6a9f9db : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c4b4a694d444fdb81f7130310586562 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c3da8717eb844919c44451ea591525f : SUCCESS in 5m 46s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/410c85e5dc78453299d5e29923e66b07 : SUCCESS in 12m 45s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/a6b88e987323494fa8fea7e6ec8580d1 : SUCCESS in 1h 47m 17s","accounts_in_message":[],"_revision_number":2},{"id":"3deb25c0e6d8a4d01ae41b737f5acb9a45c2bb91","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 10:03:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ecd010d05a87418362a8b8211c91564d57d290c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 10:09:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef4509bd37d64b3ab5951ce59aa5dfc8 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fc91fa4a081a07dbd1f90e38257e61fe01b4917f","revisions":{"18f8623748d02c23d3461f134ff9ee0c93bc7478":{"kind":"REWORK","_number":1,"created":"2021-05-04 20:09:39.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/29/789629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/29/789629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/29/789629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/29/789629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/29/789629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/29/789629/1"}}},"commit":{"parents":[{"commit":"c8befc7c60177027a4184e848788fe50eced965b","subject":"Merge \"Remove six\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/c8befc7c60177027a4184e848788fe50eced965b"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-04 20:08:05.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-04 20:08:05.000000000","tz":120},"subject":"Update networking-sfc-tempest-multinode to be run on Focal","message":"Update networking-sfc-tempest-multinode to be run on Focal\n\nChange-Id: If9c66f225fb2fcd299b05c3f29652768a99a040b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/18f8623748d02c23d3461f134ff9ee0c93bc7478"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/18f8623748d02c23d3461f134ff9ee0c93bc7478"}]},"branch":"refs/heads/master"},"fc91fa4a081a07dbd1f90e38257e61fe01b4917f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-07 09:51:45.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/29/789629/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/29/789629/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/29/789629/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/29/789629/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/29/789629/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/29/789629/2"}}},"commit":{"parents":[{"commit":"5cabef6a858c46a502bd9d9f706b349543ea8a0e","subject":"Add Python3 yoga unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/5cabef6a858c46a502bd9d9f706b349543ea8a0e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-05-04 20:08:05.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-10-07 09:51:45.000000000","tz":0},"subject":"Update networking-sfc-tempest-multinode to be run on Focal","message":"Update networking-sfc-tempest-multinode to be run on Focal\n\nChange-Id: If9c66f225fb2fcd299b05c3f29652768a99a040b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/fc91fa4a081a07dbd1f90e38257e61fe01b4917f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/fc91fa4a081a07dbd1f90e38257e61fe01b4917f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
