)]}'
{"id":"openstack%2Fnetworking-sfc~791359","triplet_id":"openstack%2Fnetworking-sfc~master~If5e544792e7b3bb8c62bb9d6bb38c189bf923dff","project":"openstack/networking-sfc","branch":"master","topic":"assertItemsEqual","hashtags":[],"change_id":"If5e544792e7b3bb8c62bb9d6bb38c189bf923dff","subject":"Use assertCountEqual instead of assertItemsEqual","status":"MERGED","created":"2021-05-14 05:52:15.000000000","updated":"2021-05-20 10:44:50.000000000","submitted":"2021-05-20 10:42:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791359-assertItemsEqual","meta_rev_id":"fbaada08a31a93da6a3ff416ba4429c6b2581588","_number":791359,"virtual_id_number":791359,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-20 10:42:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-05-20 10:44:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-05-20 10:44:50.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2021-05-20 10:44:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-05-20 09:22:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}],"reviewers":{"CC":[{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-14 05:58:55.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-05-14 05:58:55.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-05-14 05:58:55.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-05-14 05:58:55.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2021-05-14 07:19:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-17 03:36:22.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"CC"}],"messages":[{"id":"5aca01e82076e49e7cd41de5e59f7476d5d8fc55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-05-14 05:52:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73338ba6745672031c3ff5d792f6267571380cba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 07:19:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef36676144614827ad2989ff650edc37 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bddd5e27de44bc8b0a4201a4f5cd149 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efeadb240ab74a54b637d81e890cc9cc : SUCCESS in 8m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d0ba35a8bc2457b994ae9a8b775a6fe : SUCCESS in 9m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b927d31e914458bbc4e951d5cf5c319 : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8080ebbceae647079dec990f9eb13173 : FAILURE in 5m 59s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/d71d38e8031f424f9375d29849ca774a : SUCCESS in 17m 39s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/7a1ce33a1cb5432096d7bcbd9c16df73 : SUCCESS in 1h 22m 50s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/0357315f884545c69caeeccba5bcd04c : FAILURE in 31m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9428cd37b43e8be5d0858c748fa4c2268805032","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-14 14:36:07.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d996bee366ca478ba7cc0d261a937fb4322ce9bf","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-05-14 15:38:40.000000000","message":"Patch Set 1:\n\nThe docs build failure looks real and unrelated to change, i see similar failure in other projects, like:-\n\nhttps://099834f07e3210a9a951-86b8f16365265aa457f1b55e12deb361.ssl.cf1.rackcdn.com/791137/1/check/openstack-tox-docs/aa5c29b/sphinx-build-pdf.log\nhttps://ecbb8cfc7e1592890056-5f9169eacee10b8db2558d8a4c802669.ssl.cf5.rackcdn.com/791410/1/check/openstack-tox-docs/669eed7/sphinx-build-pdf.log","accounts_in_message":[],"_revision_number":1},{"id":"3af3afa85eca4301d708326304980790368d17a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 16:03:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/449c8aed36734c6f973d4460ebf6d635 : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c484c7401f424ec4a751fca9e20d5139 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fb0b1ef90b14d52b89056994712db95 : SUCCESS in 11m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a600596722940dc96a701f3d9877a04 : SUCCESS in 10m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aac69af3f2f84f15bd90ebecdcca4846 : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a210b7cc5ab44698b01c95d0368cf415 : FAILURE in 8m 22s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/9c7adca61b8e408a830182bc2d761b89 : SUCCESS in 22m 01s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/b17c0b0745a04c4ea41c53a3228585ec : SUCCESS in 1h 21m 29s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/5c3383f847fd48cbb8061c786e6346d1 : FAILURE in 32m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49f4ad0881d5953494f42598ca3542706ebed1fb","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-17 03:36:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7cb94a9cf7114dc27c60aa1dc5e28d380ec60c70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 04:43:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a1a669fee934bc9a9a89a028076781b : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3107586a128f4cceb4b20ac6d0e6e22d : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2365ecf905354b3c8e3ebae8eac3477a : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17a70b6dfe7b4b90b503e3e70c334828 : SUCCESS in 7m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d19a5cac9d9642018beafda2a2fb0eff : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8509f7da01e24c3491f43527fe3f2e8f : FAILURE in 5m 40s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/62fe33d4a908418fabdefc3bd185a021 : SUCCESS in 18m 46s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/d59fb7c6f9924a15a62f997ce2bd37fd : SUCCESS in 1h 06m 21s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/314966e03c694200a5fb57027a2c1e06 : FAILURE in 26m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"80c3c4e9a99702c7462e37369cad4dc620a159c7","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-05-17 07:15:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c17808c0393c7aed8c326e33baf785cbf3411153","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-17 09:45:54.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck\ntox-docs pdf build issue?","accounts_in_message":[],"_revision_number":1},{"id":"2c0cab4c348de98ba7949d5ef549d9d47dc4ade1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 10:51:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0bc8b894db14c5cb282d32e0aed3262 : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0188b50e9bea40f4bba50c7b0a9d3507 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ed287fe782b48f98418cfd172029296 : SUCCESS in 8m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49271de71c774737943e99878987389c : SUCCESS in 7m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/041fddd287fd4004bc63011d9bdd9a2e : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c73ad3e467343858eabd117a9dd636c : FAILURE in 6m 09s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/6b2102facc974f518206b7c937c895b0 : SUCCESS in 14m 22s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/429c1d49339f409490b1bb75b69773c9 : SUCCESS in 1h 03m 26s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/0381d53d7e1543c6b5601e566fb129fc : FAILURE in 25m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6ee5366ead2aa7bb36eccb3bd8eac5804862ae5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-19 07:57:18.000000000","message":"Patch Set 1:\n\nThis should fix the doc build issue: https://review.opendev.org/c/openstack/networking-sfc/+/792078","accounts_in_message":[],"_revision_number":1},{"id":"f2a3271f069390cb3cd26d56669fb2ca08aca364","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-05-19 08:01:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"03ea6afa3f22aa2d70c843228ccbd95ef97c2a23","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-05-19 08:01:25.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e \n\u003e This should fix the doc build issue: https://review.opendev.org/c/openstack/networking-sfc/+/792078\n\nThanks!!","accounts_in_message":[],"_revision_number":2},{"id":"b482c19a450e52311e31e39d1042b2d4a7541f58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 09:35:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5693c657ffe34e71a17caf88590c2d5b : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a03d72ecbdc944ce9fe9403ee770c193 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d20479f882c4b8b9f2fdecfab0e35a1 : SUCCESS in 10m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa4d522a370e49bdb4ab65754d154805 : SUCCESS in 8m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c960a7993aa2488aa3e8c4d9ceb1643c : SUCCESS in 10m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b60651de50a14b5885f7714017953544 : SUCCESS in 6m 21s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/4670d99884eb484fa0cd89a6dd281a0e : SUCCESS in 16m 08s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/05932201cb0b4240abb7b2f3ab1c6518 : SUCCESS in 1h 30m 04s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/d2dd7b995bc94eef8a77dcec172e2958 : FAILURE in 14m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"939d1bb87a20680a1ebf821e1c2020a26df50121","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-05-20 09:22:47.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b871f87a1ce0a8c3bb965690da4635f7054067f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 09:23:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a72b5ba8e4ccbf45065f87a43baf07e02f0ea1e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 10:42:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3009742ce1c460c921cae36f1623741 : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1e49e051c5b4e9f8fee3e4fd6304569 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e88e87b08dd422d8d52d1cd0b5ced15 : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff1ba17051184231a418758961c8aca3 : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c76b3cb8448b4ea2b5e539047a45c885 : SUCCESS in 5m 59s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/514f06c574d345e8aff5b3ba8f542a7d : SUCCESS in 14m 33s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/3e1489d713774b79b54354aa39b07268 : SUCCESS in 1h 11m 21s","accounts_in_message":[],"_revision_number":2},{"id":"ba615eccc9be4513af30c234e734e14e8cab36e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 10:42:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"defc51cc39adf5737e1d8f639b8acb40006acc4e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 10:44:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44f58a084244405f8f34729f30bdab32 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e01048cc521c2c16f36cea23eef8331c6bae0904","revisions":{"da20e5a7516d4252ed971ae3d342efdf3409976b":{"kind":"REWORK","_number":1,"created":"2021-05-14 05:52:15.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/59/791359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/59/791359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/59/791359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/59/791359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/59/791359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/59/791359/1"}}},"commit":{"parents":[{"commit":"c8befc7c60177027a4184e848788fe50eced965b","subject":"Merge \"Remove six\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/c8befc7c60177027a4184e848788fe50eced965b"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-05-14 05:50:48.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-05-14 05:51:59.000000000","tz":330},"subject":"Use assertCountEqual instead of assertItemsEqual","message":"Use assertCountEqual instead of assertItemsEqual\n\nThe assertItemsEqual method has been removed in Python 3.3 [1] but\nit was kept alive by unittest2, imported by testtools. For better\ncompatibility, change to assertCountEqual.\n\n[1] https://bugs.python.org/issue17866\n\nChange-Id: If5e544792e7b3bb8c62bb9d6bb38c189bf923dff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/da20e5a7516d4252ed971ae3d342efdf3409976b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/da20e5a7516d4252ed971ae3d342efdf3409976b"}]},"branch":"refs/heads/master"},"e01048cc521c2c16f36cea23eef8331c6bae0904":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-05-19 08:01:05.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/59/791359/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/59/791359/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/59/791359/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/59/791359/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/59/791359/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/59/791359/2"}}},"commit":{"parents":[{"commit":"069b36ba8803b35d3f3b454f80918888b2ed0a27","subject":"pdf-docs: use UPPER_CONSTRAINTS_FILE for doc build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/069b36ba8803b35d3f3b454f80918888b2ed0a27"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-05-14 05:50:48.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2021-05-19 08:01:05.000000000","tz":0},"subject":"Use assertCountEqual instead of assertItemsEqual","message":"Use assertCountEqual instead of assertItemsEqual\n\nThe assertItemsEqual method has been removed in Python 3.3 [1] but\nit was kept alive by unittest2, imported by testtools. For better\ncompatibility, change to assertCountEqual.\n\n[1] https://bugs.python.org/issue17866\n\nChange-Id: If5e544792e7b3bb8c62bb9d6bb38c189bf923dff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/e01048cc521c2c16f36cea23eef8331c6bae0904"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/e01048cc521c2c16f36cea23eef8331c6bae0904"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
