)]}'
{"id":"openstack%2Fnetworking-sfc~791681","triplet_id":"openstack%2Fnetworking-sfc~master~I7b913bf5d3877717b6703d08586b7811a9e60925","project":"openstack/networking-sfc","branch":"master","topic":"wallaby","hashtags":[],"change_id":"I7b913bf5d3877717b6703d08586b7811a9e60925","subject":"Update index.rst for stable/wallaby","status":"ABANDONED","created":"2021-05-17 08:12:10.000000000","updated":"2021-05-20 09:33:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0a03e3bf3751f6ba56b99f89c0e3cd007924c46b","_number":791681,"virtual_id_number":791681,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-17 09:29:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":-1,"date":"2021-05-20 09:14:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 09:29:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-20 09:14:25.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"e196b50f26725eca28fd3a8860c1fa50bd9ddf46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-17 08:12:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25343247f43812add7315bbf74ad2330cd744ba9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 09:29:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc0bc1ee3ffc4e07b87de9dd95ed8f68 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64963bfc6a8742128fcfe72c7d51b7d6 : FAILURE in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fdab60434e4490aa8a02196cface963 : SUCCESS in 5m 32s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/68fd24239aa6477f98e448657fb46201 : SUCCESS in 1h 16m 12s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/2f9344ac38c344adb200831b82c9eb55 : FAILURE in 30m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f9d8807ef3ec13fcd9f62389ab5ad1bdc11f301","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-05-20 09:14:25.000000000","message":"Patch Set 1: Code-Review-1\n\nAlready in progress with patch by release bot https://review.opendev.org/c/openstack/networking-sfc/+/782790","accounts_in_message":[],"_revision_number":1},{"id":"0a03e3bf3751f6ba56b99f89c0e3cd007924c46b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-20 09:33:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0768ea178b5c5678cda7f1a380cd8ff16b2faa9e","revisions":{"0768ea178b5c5678cda7f1a380cd8ff16b2faa9e":{"kind":"REWORK","_number":1,"created":"2021-05-17 08:12:10.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/81/791681/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/81/791681/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/81/791681/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/81/791681/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/81/791681/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/81/791681/1"}}},"commit":{"parents":[{"commit":"c8befc7c60177027a4184e848788fe50eced965b","subject":"Merge \"Remove six\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/c8befc7c60177027a4184e848788fe50eced965b"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-05-17 08:11:45.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-05-17 08:11:45.000000000","tz":480},"subject":"Update index.rst for stable/wallaby","message":"Update index.rst for stable/wallaby\n\nChange-Id: I7b913bf5d3877717b6703d08586b7811a9e60925\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/0768ea178b5c5678cda7f1a380cd8ff16b2faa9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/0768ea178b5c5678cda7f1a380cd8ff16b2faa9e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
