)]}'
{"id":"openstack%2Fnetworking-sfc~791926","triplet_id":"openstack%2Fnetworking-sfc~master~I29fd81d384ad3c78d20c6bc532622fbe97b35511","project":"openstack/networking-sfc","branch":"master","topic":"tox","hashtags":[],"change_id":"I29fd81d384ad3c78d20c6bc532622fbe97b35511","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","status":"ABANDONED","created":"2021-05-18 11:27:02.000000000","updated":"2023-02-28 10:56:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b2c68386c77687a007b78ef136f045970d79a3bc","_number":791926,"virtual_id_number":791926,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-21 08:03:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"all":[{"value":1,"date":"2021-05-24 06:34:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}],"CC":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-18 12:45:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-18 14:15:32.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2021-05-24 06:34:26.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"}],"messages":[{"id":"39afcab2dfb758431840c071aabe50db98f7a934","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-18 11:27:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"753290e162294c2518a3639f428d8c8456a42053","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 12:45:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db9864c6e2224c6cb83f9f7264f7f7e6 : SUCCESS in 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/395f9842ccb1405aa93790af9e1dd35e : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e06d33d7cd324d8db780f2a87e80a255 : SUCCESS in 8m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e216fce79a246339e9f900a100e83a1 : SUCCESS in 8m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58ea76e5135446fba907fbd07258e9be : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ff5a77255724be8973e4f43f8b2e4db : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eec0cc477bd34795a612ef86729a05d1 : SUCCESS in 4m 57s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/357829b95ea2487782bd762bbe82613b : RETRY_LIMIT in 5m 13s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/7936ad1bab3b4a3a9b322a4d176cfe8f : SUCCESS in 1h 12m 44s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/22e783d83aa64dccae92c02b3f24a224 : FAILURE in 11m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa482b44ebe93ff56417bfac5df14bfe5d31c10a","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-19 01:49:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e95a3bd8d7ac3be7f27c88325e789828c3866c03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 03:11:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02f991e78301433fbda2579717f8e40f : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d508b7272e0414c9d24ab0193b3af9a : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/438a3fbc03b6465098a7c762b5745c7c : SUCCESS in 8m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42764c3bb325429a8c964caf3962fa9c : SUCCESS in 8m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e575796cc0947148232351230c9ece0 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ef979b1c65b4b1f93ce57a9ae86cb3d : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/895d655cfa59498190adaa52942e5cb9 : SUCCESS in 3m 48s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/003efeca2f234638bdecdf42ec2077f6 : SUCCESS in 20m 16s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/0022f3e9df75434081431e1be2b83c19 : SUCCESS in 1h 17m 51s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/0db7bda439cb4156bb850c39d7a423bf : FAILURE in 13m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"791a1a8f81a09db377c85dcc9ccb48f57f766403","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-21 06:28:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fd166463b2ceefa1c32fa64290f4809eb61c4e54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-21 06:30:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d758d3706615960ffc39ed79f10fb0e14426e787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 08:03:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cb9275928f04a4d830bc5eb49682961 : SUCCESS in 14m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43188c4b04dd42e3ad4d0e67021535e6 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61fa1ac5017b43a2ae2ed2568332171f : SUCCESS in 9m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a450ce9aeb9442f6a306e961ea6cdc76 : SUCCESS in 9m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/337ff859c46f4e28948b80bdee08484d : SUCCESS in 9m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94cbea75a96e49af8bed4baade77c568 : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d5ab7e07ad54e2bbf1ff354036c9ee2 : SUCCESS in 5m 00s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/d5729f6081654e8fa703bc53338e6525 : SUCCESS in 25m 09s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/f10703679f034ff0a5482500a8dd7854 : SUCCESS in 1h 28m 36s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/229dc9dcda7e4da9b0ad8326af6ff352 : FAILURE in 15m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"72b93e8eb001c68a4041aeb07a4f507ff7994e63","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2021-05-24 06:34:26.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b2c68386c77687a007b78ef136f045970d79a3bc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-28 10:56:56.000000000","message":"Abandoned\n\nIf needed, feel free to restore the patch","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ed281559f75fe0debf79b3311f848d8348813f0e","revisions":{"d12c606466f18640bd26f3929b7f6e136e27f36a":{"kind":"REWORK","_number":1,"created":"2021-05-18 11:27:02.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/26/791926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/26/791926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/26/791926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/26/791926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/26/791926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/26/791926/1"}}},"commit":{"parents":[{"commit":"c8befc7c60177027a4184e848788fe50eced965b","subject":"Merge \"Remove six\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/c8befc7c60177027a4184e848788fe50eced965b"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-05-18 11:26:42.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-05-18 11:26:42.000000000","tz":480},"subject":"Fix PDF docs build","message":"Fix PDF docs build\n\nUpdate tox config to include upper-constraints as dependencies. This\nresolves issues with building PDF docs.\n\nChange-Id: I29fd81d384ad3c78d20c6bc532622fbe97b35511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/d12c606466f18640bd26f3929b7f6e136e27f36a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/d12c606466f18640bd26f3929b7f6e136e27f36a"}]},"branch":"refs/heads/master"},"438c65966583a7da22ffc6585e4213773fdabd78":{"kind":"REWORK","_number":2,"created":"2021-05-21 06:28:05.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/26/791926/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/26/791926/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/26/791926/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/26/791926/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/26/791926/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/26/791926/2"}}},"commit":{"parents":[{"commit":"e01048cc521c2c16f36cea23eef8331c6bae0904","subject":"Use assertCountEqual instead of assertItemsEqual","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/e01048cc521c2c16f36cea23eef8331c6bae0904"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-05-18 11:26:42.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-05-21 06:27:49.000000000","tz":480},"subject":"These should be indented two spaces to match the other things in this block","message":"These should be indented two spaces to match the other things in this block\n\nChange-Id: I29fd81d384ad3c78d20c6bc532622fbe97b35511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/438c65966583a7da22ffc6585e4213773fdabd78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/438c65966583a7da22ffc6585e4213773fdabd78"}]},"branch":"refs/heads/master"},"ed281559f75fe0debf79b3311f848d8348813f0e":{"kind":"REWORK","_number":3,"created":"2021-05-21 06:30:01.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/26/791926/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/26/791926/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/26/791926/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/26/791926/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/26/791926/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/26/791926/3"}}},"commit":{"parents":[{"commit":"e01048cc521c2c16f36cea23eef8331c6bae0904","subject":"Use assertCountEqual instead of assertItemsEqual","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/e01048cc521c2c16f36cea23eef8331c6bae0904"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-05-18 11:26:42.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-05-21 06:29:25.000000000","tz":480},"subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","message":"Replace deprecated UPPER_CONSTRAINTS_FILE variable\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\n\n[1] https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\n\nChange-Id: I29fd81d384ad3c78d20c6bc532622fbe97b35511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/ed281559f75fe0debf79b3311f848d8348813f0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/ed281559f75fe0debf79b3311f848d8348813f0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
